)]}'
{"id":"openstack%2Foctavia~961733","triplet_id":"openstack%2Foctavia~master~Ie7d7a9585d99f9abc219f1431363739df18f32d5","project":"openstack/octavia","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-23 06:48:30.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-09-23 06:48:30.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ie7d7a9585d99f9abc219f1431363739df18f32d5","subject":"Remove workaround for keepalived \u003c 2.0.14","status":"NEW","created":"2025-09-18 18:46:06.000000000","updated":"2026-07-31 10:21:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"4830ed634de9264fef3194d10aedcb5d3aa1ce3f","_number":961733,"virtual_id_number":961733,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-31 10:21:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-31 07:14:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-18 21:43:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-19 00:07:56.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2025-09-19 06:59:48.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2025-09-19 07:49:36.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"b1e3ad5a3a7c3c73fb1b31b09b6e2311a71f91b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-18 18:46:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31334f050b65cde155d65f2306e84c2f1a2ca2e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 21:43:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5dbc025644fa444aabff4c95c071f321\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c3e3b9656574eaba34b55fd074ee20a : SUCCESS in 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fe5539ae9ba42a2841e83c8956cc2b5 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ec8f38ee68841c4bf7c4fbc2006c118 : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c2d99228b004d6cacee68f161c10ac4 : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/818f3fd6adf44094921899fc9552a734 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bba8eee6aa694d3884c1ded32c56dc87 : SUCCESS in 13m 26s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/08abdf95691b4100aaf2f882b0a01def : SUCCESS in 5m 43s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/c5bf105237054474801aaf85d404c0dc : SUCCESS in 11m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9e6d6932136a476d9f634e69201fa159 : SUCCESS in 4m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5b054c7e083d4be393566730eef06ee0 : SUCCESS in 6m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b97eee666ec24e55a6f3366f311853c6 : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/635e6d5f9b314e7789ef01c99daf614f : SUCCESS in 1h 28m 03s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2aa2ee7c1824409aa0d86ac2ad30910e : FAILURE in 1h 58m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/56add0a2382c47e2999988d8f95dfc2f : SUCCESS in 1h 02m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/db6eb5bef9884d84b2542f78f17ebf9f : FAILURE in 2h 04m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c5d46d4f908f4a2e9014272ef4d7e99a : SUCCESS in 1h 08m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/38d21204c2514b82844585114fb49c47 : SUCCESS in 38m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f55a108424c04149821e130663ad189e : SUCCESS in 46m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d7ca215b0854187a632e940737f9e2f : SUCCESS in 37m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a9e45802009d4c3b9b8bc54ee096b5be : TIMED_OUT in 2h 32m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3afe3a40a1f04e56aec570f7ca51d428 : FAILURE in 2h 55m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c6d96ed90364478d8148fdcca9342bf7 : FAILURE in 2h 25m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b993a57dd72dca09c1ea8184e2b11bcfddeb9b8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-09-19 00:07:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9998d9f92b069d90756ed8df96469cfa5de7f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 02:55:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa490d840f8e40248c9c2ace7c9c6041\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d8f28f449db439fa05119b46ec47a7e : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7402a7e702084d529f6288c2ee1f57b8 : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c504b618379340719a3d54ebb744f4bf : FAILURE in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57ffaeab0c5d49ce95732a8739aa21ce : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb07a7eeac7447daacd9e1b7f6ef9b77 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b8467a16c5f4f659bc113d5c033aede : SUCCESS in 12m 35s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/05c1e86255764b65ae594e5f9a47dcaa : SUCCESS in 6m 45s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/a1fb16d8b73047f0a42bf12a111b6395 : SUCCESS in 7m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7fdfe47f8abb45819abba16f91c5dd31 : SUCCESS in 3m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cfc6a275d88a47558249aa9d1545611e : SUCCESS in 6m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9557ee0adb494f4cb9af3e7f8d2bea88 : SUCCESS in 6m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cce0dc2c8820477ca15d04a038e6bac2 : SUCCESS in 1h 29m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/49ae6d2000d049ac9a113278a75c1eb8 : FAILURE in 1h 57m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5c5eea3f6e2e4e109a8c6b2f4ccb54e1 : SUCCESS in 1h 00m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3fad3ba3bedf48a6883647325bdfb5b3 : FAILURE in 2h 01m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f0165d1ed73649c49d74a6fc81eb4042 : SUCCESS in 1h 13m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/08e419ee3b724c399135f21871ada5e1 : SUCCESS in 38m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f099f31b842c4d63ad53c9e84180b188 : SUCCESS in 42m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/310fd717ea654c3fb01f84036c034bc8 : SUCCESS in 35m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f020fdd3a59545a9978788c5aafcb49b : TIMED_OUT in 2h 34m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/faa79980d85f4193a3652f061f22da28 : FAILURE in 2h 45m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8e8d418676944fb086397d16f2d18b79 : FAILURE in 2h 29m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd92447ddebdc5346e42815fddd835e99ea06af7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-19 06:59:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30d2a7eb09c4e6f874a6a9a11d73d47439a5b2de","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-19 07:01:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68d47c7c5ce4826c5f1fd7005164f749afd18775","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-19 07:49:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aeba79adbc7fae5f32c879de0a0f8db9c309131c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-19 11:42:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e2432ba80b25b33924369be5bb9ee7e1bf6a1e9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-19 13:11:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ad01305c2c0b0e50a79f667bc4a95bfae9147ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-22 14:40:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d5299f129cdc14e194e88415bbcd0aa7e6b6eca8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-22 14:41:17.000000000","message":"Patch Set 1: -Code-Review\n\nCopied votes on follow-up patch sets have been updated:\n* Copied Code-Review vote has been removed from patch set 2 (was Code-Review-1) since the new Code-Review\u003d0 vote is not copyable (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\").\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cef6cd08d2ffc9aea98214d3060c7442ddbe32dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 17:47:01.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebcbb9abe9054cddb530b2403ea2f34a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5215a75a089548f6b0d291b6d09a9278 : SUCCESS in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3222e267babf4dacae1b2cd5eac2e542 : FAILURE in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ee50a7630e6477fb297329b70fde5e1 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6ca4e7996fc4d2fa816e98203701a7c : SUCCESS in 5m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6297842f33a94e59920fc07550c2f7ae : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53c02561c92b4f629af05405432c6810 : SUCCESS in 13m 20s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a7d9b192ca2c4bb0944d8edd2ad569f2 : SUCCESS in 5m 02s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/43b4c13f54d24b72a1f3160d8206ead1 : SUCCESS in 6m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/373d326a5741442988fe48bf48ba2b1e : SUCCESS in 3m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/00a6bcd98d1e45f2b034b0f28de92532 : SUCCESS in 6m 18s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7f4784147906484f8c80c288211d7b56 : SUCCESS in 8m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a6d513d695ca4621be209e5788b56ede : SUCCESS in 1h 24m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c1b9181d02844602a127582610627fd4 : SUCCESS in 1h 55m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ea3710b6f89e437c8d491ddb45462212 : SUCCESS in 1h 14m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/955e67f26c984b46a1a0067d37c457c6 : SUCCESS in 1h 57m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/08874e89c637417ba6ba57ee2840e6a9 : SUCCESS in 1h 02m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f9eb0f594f104f668ba4ccafb3cb8566 : SUCCESS in 38m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/59cee79bd7d948c09ab67f531f50a884 : SUCCESS in 41m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0352a300726045188e6e17c249cdd184 : SUCCESS in 36m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5bd1037dc00b404684438e1494a85615 : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/45cd5acc15364feb98de609e15b5c343 : TIMED_OUT in 3h 03m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/280da8b487d4472d980d1c4fac34707e : SUCCESS in 2h 29m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ff9e0a6420eeccbcdb64cde4652fb36cc909e94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-23 06:48:22.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"612f5b41ebcb8886dfee348f640c36d32ff2a639","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-23 06:48:30.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"70c785daced8ae0c36595f625d3576c97b79138b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 09:52:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06f910b15bbf495fae036eb32c40aae1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6197d35c1d94bbca5a18bc8ebc5c87d : SUCCESS in 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d3fa53f18d44ee8b667ab2e7be18f7f : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38629cb074ec4342be2cf8f1c968ae4a : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/580d1361edd64cb98d14a4dc5c4f5a8c : SUCCESS in 5m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2cc8d0d8ff69422b892484554e43def7 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f984f817c2334f4badb3d9742c5011aa : SUCCESS in 14m 32s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cae4d4c387904e218d0f54b1f10e1571 : SUCCESS in 7m 12s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/7409ae5a58df48ed8e89bf9d9a1db150 : SUCCESS in 7m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/88c7689490044b44bea6168a1dc057cf : SUCCESS in 5m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7b982491270b4f65ae9678a59d6cc9db : SUCCESS in 8m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ba41a576da4049b796c52ecc5ff21e96 : SUCCESS in 10m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a12d17ae871f45fe89a5ceeb704c7ca0 : SUCCESS in 1h 32m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e5bf6fd274ca4386871c138fade492c6 : SUCCESS in 1h 50m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/58f1e3e8c29243a794fa43c604815cf3 : SUCCESS in 1h 05m 11s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e45e54184c3345db8412c3470852e9cc : SUCCESS in 1h 52m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2bf352705850422cb3909cd5fb10d87e : SUCCESS in 1h 08m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3d60ef1c8fc94c49bb0b4dd927470641 : SUCCESS in 35m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/48fa4c81e29b445baa5982cd1821200b : SUCCESS in 43m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54ed1d5de8344be79159e0fe340e06fa : SUCCESS in 33m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/35940ad80f394e75b628a2be98d3e02a : SUCCESS in 2h 26m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c2500281c97145a09a17347f2870893b : TIMED_OUT in 3h 03m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/06a09b1c5c94491fb157cceff72bb0ea : TIMED_OUT in 2h 33m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"753e86baebe37acbfb7e9bb2c451d2f2d314e284","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-31 07:14:59.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"4830ed634de9264fef3194d10aedcb5d3aa1ce3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 10:21:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba43f2c343df4f3d8eae9bf7b339bac2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9358caddac96497c89f4e1f5b2a4e0b1 : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4030cba594b54fd7ae12570b033538b1 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29e93378e758412292497bf346cde33b : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/69211633e96f448c87dd359cd088ea66 : SUCCESS in 4m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/88e756bf48ad45a8886b9d64d6d78f20 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1427353341fe48dba159268a8fd5d139 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c87fd3692354f9db273340107d32062 : SUCCESS in 8m 47s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/84404880345c40df9be6c0fcbaf55443 : SUCCESS in 4m 26s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/c9bb320cf21d4aedb96649e9a8179b17 : SUCCESS in 5m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/42ec4cdd05614a9480955a43bd8ff02f : SUCCESS in 3m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f9f13af07a0a40eba19d10bd5e9dd8ea : SUCCESS in 4m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5d297dee574f4263b818955ebbf5617a : SUCCESS in 1h 18m 21s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/27b63eb3554a48af945d9a86ced954de : SUCCESS in 1h 25m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8993aca6b6b34ab0b2d6338eb6bdec66 : SUCCESS in 1h 06m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a1757c8036db4b9a8dac53d978dbe282 : SUCCESS in 1h 21m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9a14cc363be344acb963647cdcf39038 : SUCCESS in 1h 08m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/76b5c76e0db74a84b062c2ccd2d8bec9 : SUCCESS in 32m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/81be3dba0c6a409a87ac3c3b2391e3f5 : SUCCESS in 1h 01m 02s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/69550545b46d43bb87f1b6e16176a0f0 : SUCCESS in 40m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3195c009c9d44ca7a847daa00a57de74 : SUCCESS in 29m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/68057a8a59a44b6db8cb705adc06c840 : SUCCESS in 46m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/de91d0b3a3544f3d9c100d4bd77acba4 : TIMED_OUT in 3h 05m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fea25363cdc9432a9ec51a8fb813e5ff : SUCCESS in 1h 22m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/270ec881ac5e472c8eef7e461fc81bae : TIMED_OUT in 2h 36m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3dd0270633f642c58ef902827b28603e : SUCCESS in 2h 29m 37s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ccd477522adb4f64fcb6276e4a5e56aef7825a68","revisions":{"51be777fe890141bbe9d57b96b382c14e2b5eac6":{"kind":"REWORK","_number":1,"created":"2025-09-18 18:46:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/961733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/961733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/961733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/961733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/961733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/961733/1"}}},"commit":{"parents":[{"commit":"f6a639bb44127d9ba76b8c1cc72a4f16d75aeb6f","subject":"Merge \"Fix L7Rules with FILE_TYPE and EQUAL_TO\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f6a639bb44127d9ba76b8c1cc72a4f16d75aeb6f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-18 18:43:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-18 18:45:54.000000000","tz":540},"subject":"Remove workaround for keepalived \u003c 2.0.14","message":"Remove workaround for keepalived \u003c 2.0.14\n\nRemove workaround for old keepalived version, which was added by\n0a9f587015176723768a8a861265fda797c5d52e .\n\nRecent operating system versions provide more recent versions.\n - Ubuntu 24.04       : 2.2.x\n - CentOS Stream 9, 10: 2.2.x\n - Debian 12          : 2.2.x\n\nChange-Id: Ie7d7a9585d99f9abc219f1431363739df18f32d5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51be777fe890141bbe9d57b96b382c14e2b5eac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51be777fe890141bbe9d57b96b382c14e2b5eac6"}]},"branch":"refs/heads/master"},"7bee5810441127ec84f974732b14e91d05987d23":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-09-22 14:40:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/33/961733/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/961733/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/961733/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/961733/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/961733/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/961733/2"}}},"commit":{"parents":[{"commit":"5b9affda4e20625c8a6a4b5870d06f97591fea2c","subject":"Avoid invalid status when reloading keepalived","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b9affda4e20625c8a6a4b5870d06f97591fea2c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-18 18:43:56.000000000","tz":540},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-09-22 14:39:36.000000000","tz":0},"subject":"Remove workaround for keepalived \u003c 2.0.14","message":"Remove workaround for keepalived \u003c 2.0.14\n\nRemove workaround for old keepalived version, which was added by\n0a9f587015176723768a8a861265fda797c5d52e .\n\nRecent operating system versions provide more recent versions.\n - Ubuntu 24.04       : 2.2.x\n - CentOS Stream 9, 10: 2.2.x\n - Debian 12          : 2.2.x\n\nChange-Id: Ie7d7a9585d99f9abc219f1431363739df18f32d5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7bee5810441127ec84f974732b14e91d05987d23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7bee5810441127ec84f974732b14e91d05987d23"}]},"branch":"refs/heads/master"},"a1937f13aaf4bb19796c9a9f29fb56f07909da95":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-23 06:48:22.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/33/961733/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/961733/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/961733/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/961733/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/961733/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/961733/3"}}},"commit":{"parents":[{"commit":"e61914a474709d7d3f17dd0ed52c5894b7fa67f1","subject":"Avoid invalid status when reloading keepalived","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e61914a474709d7d3f17dd0ed52c5894b7fa67f1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-18 18:43:56.000000000","tz":540},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-09-23 06:48:22.000000000","tz":0},"subject":"Remove workaround for keepalived \u003c 2.0.14","message":"Remove workaround for keepalived \u003c 2.0.14\n\nRemove workaround for old keepalived version, which was added by\n0a9f587015176723768a8a861265fda797c5d52e .\n\nRecent operating system versions provide more recent versions.\n - Ubuntu 24.04       : 2.2.x\n - CentOS Stream 9, 10: 2.2.x\n - Debian 12          : 2.2.x\n\nChange-Id: Ie7d7a9585d99f9abc219f1431363739df18f32d5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1937f13aaf4bb19796c9a9f29fb56f07909da95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1937f13aaf4bb19796c9a9f29fb56f07909da95"}]},"branch":"refs/heads/master","description":"Rebase"},"ccd477522adb4f64fcb6276e4a5e56aef7825a68":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-07-31 07:14:59.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/33/961733/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/961733/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/961733/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/961733/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/961733/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/961733/4"}}},"commit":{"parents":[{"commit":"5652208250164a8496605bca6c8e072309a13965","subject":"Avoid invalid status when reloading keepalived","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5652208250164a8496605bca6c8e072309a13965"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-18 18:43:56.000000000","tz":540},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-07-31 07:14:59.000000000","tz":0},"subject":"Remove workaround for keepalived \u003c 2.0.14","message":"Remove workaround for keepalived \u003c 2.0.14\n\nRemove workaround for old keepalived version, which was added by\n0a9f587015176723768a8a861265fda797c5d52e .\n\nRecent operating system versions provide more recent versions.\n - Ubuntu 24.04       : 2.2.x\n - CentOS Stream 9, 10: 2.2.x\n - Debian 12          : 2.2.x\n\nChange-Id: Ie7d7a9585d99f9abc219f1431363739df18f32d5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ccd477522adb4f64fcb6276e4a5e56aef7825a68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ccd477522adb4f64fcb6276e4a5e56aef7825a68"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e61914a474709d7d3f17dd0ed52c5894b7fa67f1","ours":"a1937f13aaf4bb19796c9a9f29fb56f07909da95","theirs":"5652208250164a8496605bca6c8e072309a13965","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
