)]}'
{"id":"openstack%2Foctavia~961987","triplet_id":"openstack%2Foctavia~master~Iad80c95f82a4e53479c5b11da1e64a1f8adc825b","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-10-09 05:41:39.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Iad80c95f82a4e53479c5b11da1e64a1f8adc825b","subject":"Avoid invalid status when reloading keepalived","status":"MERGED","created":"2025-09-22 14:40:43.000000000","updated":"2026-10-09 11:19:52.000000000","submitted":"2026-10-09 11:18:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961987","meta_rev_id":"ea687382471ee84905fe32483652a792713feb34","_number":961987,"virtual_id_number":961987,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-09 11:18:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-10-08 08:26:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-10-09 09:54:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-10-09 09:54:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-10-09 09:54:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-22 17:38:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-10-08 08:26:46.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"00bf1c2ec324306448ea4c0227c0446c81329b0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-22 14:40:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f77b119681d1b884d8574dd785f210a4ae176a91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 17:38:56.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33be12b3eaaa41eb9fc36cd3a202f6a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2015c7bffa04cc3809045168f1c97ad : SUCCESS in 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce44039e749420b9d6d0b97e575b312 : FAILURE in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd855a6e9fa542d497099c6752c7793d : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71ce01be46c4403a84d5f769aa0329f0 : SUCCESS in 6m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de4f202f67fb4b1588268c2850d093d2 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/839a7d74f2bf44fc81bbf1c7f6265aa9 : SUCCESS in 15m 20s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f86565671bbc42c1b5efee5b934aae10 : SUCCESS in 5m 08s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/40eed7565ad6449389d31e15aae3fd8e : SUCCESS in 10m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f3a45524cf19485e905fd0061c270fa8 : SUCCESS in 4m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4b5bcb171a574dc2bf58455d448cec30 : SUCCESS in 6m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/699ad52ac4d14abdbf2ca9e005cf3385 : SUCCESS in 8m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c87bb6484817447aaaabce97c91aa22a : SUCCESS in 1h 30m 28s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/58950b10974846c0a3dc668cf0bf1f90 : SUCCESS in 1h 54m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6afa0bde99644e9687de553845e9f04f : SUCCESS in 1h 07m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e2d6bf23eb9641f7b47136587eceb5e3 : SUCCESS in 1h 59m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/51c53d0207fb479d9986445b38e99fc4 : SUCCESS in 1h 13m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5e931bf7b120478f9b1c5b64326afb7b : SUCCESS in 37m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/533e165ea3e6409eba91552564403778 : SUCCESS in 45m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75f1f4eba38b47959320f065a4a83eff : SUCCESS in 32m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3b79d8f27c934b258d2c5fb20aa9a04a : TIMED_OUT in 2h 32m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/788dadfd26df4fa48f5171203482a0ea : SUCCESS in 2h 56m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/071d88c955504968ab8d8763c7cae760 : SUCCESS in 2h 25m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f3986c6df046b0f878cc7dc1168390c9b8798b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-23 06:48:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9914f76dc51c4e558ec507c48a675e5a1f284e07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 09:53:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c08f68da02164fd5b2c481246d7e03ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a44f7af279c476cb18fdae41fab7486 : SUCCESS in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfa22bd3af314298abaa64fb34692b4f : SUCCESS in 7m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f45f211fb6bc4a0fbfeb7399349b749c : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf0c1caf998d4c248678a6e8acd0259c : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6775d63cb8e14c7692e5ecd362d3a5d0 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b23a6d0c06c64079bf63039515c0a49e : SUCCESS in 13m 28s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/226420279faf44fe979392f5f1419607 : SUCCESS in 5m 02s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/488d3171d769453190216c3f59dc66a8 : SUCCESS in 9m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6057b4aad30c432eb262d6c59f5f6409 : SUCCESS in 2m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/49cd25bdc0b642388563e2735a84e96c : SUCCESS in 10m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/182d828b269f4c82a81b7bae5aaef69b : SUCCESS in 5m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ca8d5361a4b043298a12ed4055108700 : SUCCESS in 1h 24m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/457dd1e0745b4f17b037c8fa459c3614 : SUCCESS in 1h 59m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a89f6514d5ed48b49821035b6529684d : SUCCESS in 1h 04m 04s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/607b6fafc40547a49c5791395beca527 : SUCCESS in 2h 04m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9527a16d589446b789d70de67c10cb66 : SUCCESS in 1h 02m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9fc0eb2805324f238f8c9ffbd12f7891 : SUCCESS in 44m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2f820450a20a472f8924a4ab923a433d : SUCCESS in 51m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18cda68616ea4bc4ac5768e98107fe22 : SUCCESS in 42m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/776a6f4b9fc542ab9859383001f70cf9 : TIMED_OUT in 2h 33m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d628334c341b4599a6df50fdd0df1154 : TIMED_OUT in 3h 03m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/361c85464db84aeca667e102a066fff7 : SUCCESS in 2h 18m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d954cb828c35c13da122ad44527b1bf78ddbf65d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-31 06:45:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a24cdac3eec0cb7fb7b6f8e114c48ef163354609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 09:25:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/445fae5cddf645c8b567121fc42dbbbf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1873e8bc15a043aabbbb66a4670db54f : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/874a72953a954533b74ed991034686ad : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5119aa61392485b9529ec4ec44d6cd6 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67b8334e002d4301b1ac8bbccb2b6ed6 : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/99d3efb9e507497fa7ce72eaecdb3e3f : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aef23e7e22c04507ac9cacb28f5113a1 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db7dd08883d544a88d30cb3364acf08a : SUCCESS in 5m 26s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/f1bc7c23f5cc4959b85860ef734c4918 : SUCCESS in 7m 47s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/4c9d580099d9457d82563b5bb819e15c : SUCCESS in 6m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f37327bfda584b8bb7429edc29c66377 : SUCCESS in 2m 36s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7f1da5d3dad84b2a99cb660b06f644fb : SUCCESS in 5m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5a58a5cca271483f9c1f3bb4cd717b12 : SUCCESS in 40m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/108513e0df414be5a28854c95334049c : SUCCESS in 1h 19m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4b39bcf4afb04ebc8968caf9a5c2dca4 : SUCCESS in 54m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/88cd9e8b848343aa9c81b2bf8eef0091 : SUCCESS in 1h 23m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d172ebd09ca94267b48009dfc5949d39 : SUCCESS in 58m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/67eca5424974418098f70b33a0151823 : SUCCESS in 32m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7ddbac8a1753433a8093b983603d18e0 : SUCCESS in 39m 51s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/3669014c4e43432c9c46211c85d21ec3 : SUCCESS in 39m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c5b9aef392b74b8b9ab394641848a607 : SUCCESS in 30m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ff8f4463869b409eba1a9fcfa68fba4d : SUCCESS in 30m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3ff773d8b3194a4ca6f179a1a00b66e6 : SUCCESS in 2h 04m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a7bd7244d4db48a88958c88812f6333c : SUCCESS in 1h 22m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/aa956aed5acb49cb8a560e9934c16bca : TIMED_OUT in 2h 37m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9b2409d539934d6e99fa96ed65cd9308 : SUCCESS in 2h 27m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"db982e11ff96db1d603b16f6f05e2e7b872d7bf0","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-10-08 08:26:46.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3ab6cfbd71abd04dd8578750a438c4cfce5f5cf0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-10-09 05:41:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a932ccf553558b3c027b724942ac134b304e2089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-09 07:50:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69f91dadc29e4911b291520a2f171a49\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/256482bd094c406d852e1f4ee4051181 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3301c526c8743eb978d791100df5fef : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b9df707e512d43fe91f4f9f5e4b73710 : SUCCESS in 4m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/88db7929a50f45f19aa27f9e1fdc653d : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ab7408536d34ac2afc3ab1ff017f459 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d204cd102a244df95b3709fc23439b1 : SUCCESS in 4m 16s\n- octavia-tox-py314-tips https://zuul.opendev.org/t/openstack/build/79adcc136d2149f0af0d3db325e21502 : SUCCESS in 4m 38s\n- octavia-tox-functional-py314-tips https://zuul.opendev.org/t/openstack/build/15d18b391d17405db393a3ed96194c78 : SUCCESS in 9m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5e175fad6b6e4f929c53d009d7a5ec52 : SUCCESS in 2m 37s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/10db68c376484609b5274fd47177a838 : SUCCESS in 8m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6d1b0cb4c9bf412cb7334c060f521af6 : SUCCESS in 45m 38s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/076a7077d3714c32b32d347529dfbb82 : SUCCESS in 1h 21m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a49717436b094fd0995bad05a6f399c9 : SUCCESS in 58m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/24e0abb4176946c48c8b991a38ad8731 : SUCCESS in 1h 29m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/478dd99dd6a246d19c21775ca1b96c5c : SUCCESS in 1h 06m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b2633e062c4048a194aa26ae8ede0e4c : SUCCESS in 37m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/10ee844c16c5472e86b89081c5d88b4f : SUCCESS in 42m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae39d81412554ae0a890eab4304e9022 : SUCCESS in 33m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8eac2a59271849acb63fca19afa3a531 : SUCCESS in 39m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c7f4ba6ac25a474089ced21e4424cf96 : SUCCESS in 2h 06m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fc5a3a0c9089432296e361babc2e9515 : SUCCESS in 1h 18m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/8756315743b24c839503e26fcb17a6cd : SUCCESS in 1h 12m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b94b12de79024cea89ae92aee96cbffa : SUCCESS in 57m 35s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/pre.yaml for job openstack-tox-cover.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/tox/pre.yaml for job openstack-tox-cover.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/tox/run.yaml for job openstack-tox-cover.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/tox/cover-post.yaml for job openstack-tox-cover.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/tox/post.yaml for job openstack-tox-cover.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"2678d53905184357736e6760093909090ee8bd01","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-10-09 09:54:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4c96f279c575c040908ffed08ae506efe3b2cc05","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-10-09 09:54:47.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"2331b6273d258361d1710e99bbc1e238a5fbe452","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-09 09:55:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3bf07042d442a6dc77f5447fb724cbfc546cd965","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-09 11:18:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b0700fae01c4bcbbf5fa8a65423039e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2db1c35e82446d683746a1f174d11ec : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a83fbd5437549548c0fffbc379e24ae : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/655b60fc3f5647709f8f30dc8f77ecec : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20358811ce3945ce88de40afe2a57764 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a055ee4dff394b2082a2a7857420c8f0 : SUCCESS in 10m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/02c26bdcbab24024ab427b0d1270f6d1 : SUCCESS in 5m 23s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/97cbf525f37e48b984731d98f77dd163 : SUCCESS in 6m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/78089763574e4574962af50716e58b57 : SUCCESS in 1h 14m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5151e88f476444d689d6dfe65f5ec22e : SUCCESS in 1h 21m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/17b66e1832384472b7b9fda6701e9344 : SUCCESS in 55m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cc520a553ea943088ce5539d425514cf : SUCCESS in 37m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7e71424e362a42438cd35006e3c68c19 : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":3},{"id":"a0753389b7d006abcd8afe79ff2057c0b2198ef9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-09 11:18:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ea687382471ee84905fe32483652a792713feb34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-09 11:19:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fce2b660b3741c99f6400c581b06ecb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b23e3ba688504a41af577aaa2df1b1b5 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/898589963aa24d59ac72599fbfb31e59 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5652208250164a8496605bca6c8e072309a13965","revisions":{"5b9affda4e20625c8a6a4b5870d06f97591fea2c":{"kind":"REWORK","_number":1,"created":"2025-09-22 14:40:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/961987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/961987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/961987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/961987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/961987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/961987/1"}}},"commit":{"parents":[{"commit":"f6a639bb44127d9ba76b8c1cc72a4f16d75aeb6f","subject":"Merge \"Fix L7Rules with FILE_TYPE and EQUAL_TO\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f6a639bb44127d9ba76b8c1cc72a4f16d75aeb6f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-09-19 09:37:54.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-09-22 14:39:27.000000000","tz":0},"subject":"Avoid invalid status when reloading keepalived","message":"Avoid invalid status when reloading keepalived\n\namphora-agent builds the status of the members by computing the\ndifference between the configuration file and the status of ipvsadm.\n\n[0] fixed an issue that happened when the status message was built after\nwriting a new config file and before reloading keepalived.\nNew members present in the config file but not in ipvsadm were\nincorrectly reported as DOWN.\n\n[1] removes a workaround that restarted keepalived instead of reloading\nit, and it breaks the behavior fixed in [0]\n\nTo mitigate it, amphora-agent preserves the timestamp each time\nkeepalived is started/reloaded, giving the ability to detect new members\nin the config.\n\n[0] Id9e19375ebca6a720e6a85006f5e8948d3aed760\n[1] https://review.opendev.org/c/openstack/octavia/+/961733\n\nChange-Id: Iad80c95f82a4e53479c5b11da1e64a1f8adc825b\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b9affda4e20625c8a6a4b5870d06f97591fea2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b9affda4e20625c8a6a4b5870d06f97591fea2c"}]},"branch":"refs/heads/master"},"e61914a474709d7d3f17dd0ed52c5894b7fa67f1":{"kind":"REWORK","_number":2,"created":"2025-09-23 06:48:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/961987/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/961987/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/961987/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/961987/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/961987/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/961987/2"}}},"commit":{"parents":[{"commit":"f6a639bb44127d9ba76b8c1cc72a4f16d75aeb6f","subject":"Merge \"Fix L7Rules with FILE_TYPE and EQUAL_TO\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f6a639bb44127d9ba76b8c1cc72a4f16d75aeb6f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-09-19 09:37:54.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-09-23 06:48:02.000000000","tz":120},"subject":"Avoid invalid status when reloading keepalived","message":"Avoid invalid status when reloading keepalived\n\namphora-agent builds the status of the members by computing the\ndifference between the configuration file and the status of ipvsadm.\n\n[0] fixed an issue that happened when the status message was built after\nwriting a new config file and before reloading keepalived.\nNew members present in the config file but not in ipvsadm were\nincorrectly reported as DOWN.\n\n[1] removes a workaround that restarted keepalived instead of reloading\nit, and it breaks the behavior fixed in [0]\n\nTo mitigate it, amphora-agent preserves the timestamp each time\nkeepalived is started/reloaded, giving the ability to detect new members\nin the config.\n\n[0] Id9e19375ebca6a720e6a85006f5e8948d3aed760\n[1] https://review.opendev.org/c/openstack/octavia/+/961733\n\nChange-Id: Iad80c95f82a4e53479c5b11da1e64a1f8adc825b\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e61914a474709d7d3f17dd0ed52c5894b7fa67f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e61914a474709d7d3f17dd0ed52c5894b7fa67f1"}]},"branch":"refs/heads/master"},"5652208250164a8496605bca6c8e072309a13965":{"kind":"REWORK","_number":3,"created":"2026-07-31 06:45:53.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/961987/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/961987/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/961987/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/961987/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/961987/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/961987/3"}}},"commit":{"parents":[{"commit":"2b6b3bf55571ce7900b0afa06c978ef1d429eb7c","subject":"Merge \"Fix ZooKeeper session churn in ZookeeperTaskFlowDriver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2b6b3bf55571ce7900b0afa06c978ef1d429eb7c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-09-19 09:37:54.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-07-31 06:43:07.000000000","tz":0},"subject":"Avoid invalid status when reloading keepalived","message":"Avoid invalid status when reloading keepalived\n\namphora-agent builds the status of the members by computing the\ndifference between the configuration file and the status of ipvsadm.\n\n[0] fixed an issue that happened when the status message was built after\nwriting a new config file and before reloading keepalived.\nNew members present in the config file but not in ipvsadm were\nincorrectly reported as DOWN.\n\n[1] removes a workaround that restarted keepalived instead of reloading\nit, and it breaks the behavior fixed in [0]\n\nTo mitigate it, amphora-agent preserves the timestamp each time\nkeepalived is started/reloaded, giving the ability to detect new members\nin the config.\n\n[0] Id9e19375ebca6a720e6a85006f5e8948d3aed760\n[1] https://review.opendev.org/c/openstack/octavia/+/961733\n\nChange-Id: Iad80c95f82a4e53479c5b11da1e64a1f8adc825b\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nAssisted-By: Claude (Anthropic)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5652208250164a8496605bca6c8e072309a13965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5652208250164a8496605bca6c8e072309a13965"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
