)]}'
{"id":"openstack%2Foctavia~962274","triplet_id":"openstack%2Foctavia~stable%2F2024.1~I71309a0f7d57ccbe9bade5413fbd7fec01eed627","project":"openstack/octavia","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-10-14 23:36:47.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-10-14 19:45:25.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I71309a0f7d57ccbe9bade5413fbd7fec01eed627","subject":"Fix L7Rules with FILE_TYPE and EQUAL_TO","status":"MERGED","created":"2025-09-25 12:16:32.000000000","updated":"2025-10-14 23:37:43.000000000","submitted":"2025-10-14 23:36:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962274","meta_rev_id":"8c72ca66f8c6ad9d3b91abde9b2551f74d096534","_number":962274,"virtual_id_number":962274,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-14 23:36:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-10-14 19:45:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2025-10-14 19:45:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-25 14:36:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-29 18:35:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ffa22068a86ee33aa4f1db8caa590b693323292d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-25 12:16:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"195cc99c181005495fae18c27548b8267df70c21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 14:36:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38d47cea3bff469a8e1d9e709ea1dd40\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0ef311fed674f979fee2bbcb9d6b238 : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d63452e28604256ad25167d6dc84221 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/149a1679fa48438f912ba5e5a8b2257d : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c0c9c5242b443788dbd1427de94cb72 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90db96bb27a04ced9c6bf86ae162dbb8 : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7393a2fb04db4d9683bbeaf9c07844be : SUCCESS in 4m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6e2959a93f914aaebb771f229583d067 : SUCCESS in 4m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bb25ae7c6eea4393a6171129ca1854ad : SUCCESS in 9m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c641df9c7396447094fda88e2d5361c0 : SUCCESS in 11m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2d9edde0a5514af49cb929e150f1be27 : SUCCESS in 7m 33s (non-voting)\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/4df5887f94494c89ba59aa3dd87a4bce : SUCCESS in 58m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/0eca034a5e784c0a829235ef07160ec3 : SUCCESS in 1h 16m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/54e0037dddee4c889df4c9e290f0560e : SUCCESS in 1h 11m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/f29b411cf92144af97aa43fee23ca05a : SUCCESS in 1h 40m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/72778f7e2802499cada41153e121c8fa : SUCCESS in 1h 15m 45s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e70ac86ab4f040af9b64c15f0b26956d : SUCCESS in 38m 34s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/bad4f76aa6494f8b9a7ba8af7b7233fd : SUCCESS in 1h 04m 53s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/45105e5bfc174554aafcfedb3c2d4724 : FAILURE in 51m 23s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e4f75749db9641a0b8b5e32ccf43cea6 : FAILURE in 24m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d6d2cf501c9c4dbe9dc6637ca0937dd7 : SUCCESS in 30m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-1 https://zuul.opendev.org/t/openstack/build/43aab480742944d1a34f8b2755ceb92d : SUCCESS in 2h 17m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-1 https://zuul.opendev.org/t/openstack/build/679ec6122c3440eebf29f3ea3326c424 : RETRY_LIMIT in 5m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-1 https://zuul.opendev.org/t/openstack/build/590394a298f744cd8cddd751fdb49e88 : FAILURE in 36m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3866c75c80504496a8ab338877168913 : FAILURE in 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"826648d5402d937f290aac3dba1ea965e891d6ad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-09-29 18:35:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9716c917925fd9169d701f33e20d7fe644d3d862","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-09 17:49:28.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3bc3028fcd28ca36eaf2de9f8dd5d141dc0d2e26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 17:50:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40995d34f633d10e3016f6096e972e8559ae0bbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 21:18:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/36289a7e6b424ac0a4afa0d3c79daa54\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f22687e6b964f87b9ee1ce8c2f6d244 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72233966a0eb45fe8d1c79fbbed4a0c2 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d48ae325709641e2bfbace13f7af7b11 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46babba6507243f9a7865f603fdc0c9c : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0f25f3ba3ce4073b7c184bfb08e6389 : SUCCESS in 5m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cd399b7c57e24ed59a392120765b0023 : SUCCESS in 5m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/68d65e4c60fe4e7782a91d70387765c2 : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/ac462c5dc0aa4d0ebb2769d44256e0ed : SUCCESS in 1h 42m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/a5635fdefe494f9a9011496ccea7a312 : FAILURE in 1h 46m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/8cb3c0250f78431fb02382a93d01cac7 : SUCCESS in 1h 07m 50s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d54e4b59de8c40e0be4355287f526bc6 : SUCCESS in 29m 26s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/13180d2573894181a41c03390dee4405 : SUCCESS in 28m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b494c0535b1d29e9d2857e0e50f50993df4d305a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-14 19:45:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_29244\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":2},{"id":"b1751ed0aed7e0b091d081cd52ab91087a2bd002","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-14 19:45:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e3faf75012f54ac1c8a39abd4a0c2075cbc42e1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 21:55:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f71c82c59af94f098473cb90131e3173\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4566901eeb734503ba8b556bdc386eff : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abf7d030a1b84a1b9d66b8ef548e095e : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/941a20dd707d4ae0925cbac704389929 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d2bf275a98f47e488ed67f572076dde : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4159ed30a40453a8108461acf94e3af : SUCCESS in 11m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1567f27ebf4f4480bc8a35cfae137182 : SUCCESS in 3m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/918a0bb2b907465cbdf8e4ab404c2eab : SUCCESS in 4m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8dc5a77cf34f417d997d349af7239c39 : SUCCESS in 4m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a8e21baf6fbc44a98889a78e19113e11 : SUCCESS in 7m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/61ee217a2d134a76a92b09ef0a286d80 : SUCCESS in 4m 01s (non-voting)\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/fa748a24a85f41dea04200a630acb962 : SUCCESS in 1h 08m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3d2da5b46a3d4524ac9735f2233366d6 : SUCCESS in 1h 38m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/86ae48364686425ab78609800ffba9f5 : SUCCESS in 1h 00m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/015ebfe27afd45239cd1de199e9daa97 : SUCCESS in 1h 42m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/04a89e555275475196c40013b81130b7 : SUCCESS in 1h 08m 17s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/753cdae720d94df0a79b4c1a82226b3f : SUCCESS in 35m 17s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/cad3fcc9f6b64d2e9fcf16ca24de5a77 : SUCCESS in 37m 06s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/fd2d7742c31048b8a28644ba7a1df276 : FAILURE in 34m 08s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-1 https://zuul.opendev.org/t/openstack/build/2fde723b5e474a1cbcd21572b013714d : FAILURE in 45m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/bf7200b89ce3401fa48743ed9dd75459 : SUCCESS in 32m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-1 https://zuul.opendev.org/t/openstack/build/cc24b8fac1ea404196df6f457da6561c : SUCCESS in 2h 08m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-1 https://zuul.opendev.org/t/openstack/build/c58b1b76cfb24e5b9da9931a45fbef03 : RETRY_LIMIT in 3m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-1 https://zuul.opendev.org/t/openstack/build/b0e49ffaa20a4801bad6c54de3edb17d : FAILURE in 21m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-1 https://zuul.opendev.org/t/openstack/build/82cd0a8306a443a19212a677eff96d69 : FAILURE in 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c36ba82d6f2bcab616d43e46f56b8a38bbdfa235","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 21:56:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"53fa937392bf4c007b95b390c61663ca51620fc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 23:36:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caab807a62a04ad98215360087d64f0c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bde4228ffd55485195d32aa6d5b61bec : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/296a9ce780d843aaad56cfcf83baffed : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/352e811d319b4326bf29fc5a2a76bf55 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62f3d79687a4345918f5a59e5a59dab : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5732548256874347b0d5c44fcdd06626 : SUCCESS in 5m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/21d049679c1943eca3092321accb2813 : SUCCESS in 4m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aec87f2601b94134b1da8e529204d70d : SUCCESS in 8m 56s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/32c03ed08b1d43b6b9fad4b8ad8d67b9 : SUCCESS in 1h 02m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/631f98d5641a46c88f8ecefb89a75db0 : SUCCESS in 1h 39m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/76db4ccdf99e4a809283c72217b45d20 : SUCCESS in 1h 07m 37s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/1585ae0cb1d24ecfb414da9665d2d43d : SUCCESS in 34m 18s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e31103f1143643c9915f480adeff20de : SUCCESS in 38m 09s","accounts_in_message":[],"_revision_number":2},{"id":"68902b5ea9bfd28ed6928f7b4feab64e75e653cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 23:36:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8c72ca66f8c6ad9d3b91abde9b2551f74d096534","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 23:37:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8bab07295e4469dbf4a06dcc810b8bb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fbf50eec9094a27a0d5fa9122466b8c : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ead1ae7fb284b808cfe867a086130e3 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"711f97bc8c71c6238038e422ec0db07dacf5ccc9","revisions":{"8f39651a34e921bbda9285a27e0e1079bddeccba":{"kind":"REWORK","_number":1,"created":"2025-09-25 12:16:32.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/74/962274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/962274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/962274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/962274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/962274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/962274/1"}}},"commit":{"parents":[{"commit":"cc3dd1abe3e23071662494786a5e5bddb3538e89","subject":"Reduce tune.ssl.cachesize for HTTPS terminating listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cc3dd1abe3e23071662494786a5e5bddb3538e89"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-08-19 09:18:55.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-09-25 12:16:24.000000000","tz":120},"subject":"Fix L7Rules with FILE_TYPE and EQUAL_TO","message":"Fix L7Rules with FILE_TYPE and EQUAL_TO\n\nWhen using a L7Rule with FILE_TYPE and EQUAL_TO, Octavia generated an ACL with\n\"path_end -m str \u003cvalue\u003e\"\nwhich doesn\u0027t work in HAProxy (https://github.com/haproxy/haproxy/issues/2567)\nUsing \"path_end \u003cvalue\u003e\" fixes the issue.\n\nCloses-Bug: #2066165\nChange-Id: I71309a0f7d57ccbe9bade5413fbd7fec01eed627\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit c64c5f54430c1fb647183b46e1fdaf25c643cb69)\n(cherry picked from commit b44406ea4b3e4376c40b339d53953ae76c8efd45)\n(cherry picked from commit 1f0be521fc1e80573ea20d6455882d3a9981eaeb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f39651a34e921bbda9285a27e0e1079bddeccba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f39651a34e921bbda9285a27e0e1079bddeccba"}]},"branch":"refs/heads/stable/2024.1"},"711f97bc8c71c6238038e422ec0db07dacf5ccc9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-14 19:45:14.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/74/962274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/962274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/962274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/962274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/962274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/962274/2"}}},"commit":{"parents":[{"commit":"2dcd82a827ae8bc6208f7d2e3ebe9b51c8fc66f5","subject":"Merge \"Fix race condition in cascade delete\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2dcd82a827ae8bc6208f7d2e3ebe9b51c8fc66f5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-08-19 09:18:55.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-14 19:45:14.000000000","tz":0},"subject":"Fix L7Rules with FILE_TYPE and EQUAL_TO","message":"Fix L7Rules with FILE_TYPE and EQUAL_TO\n\nWhen using a L7Rule with FILE_TYPE and EQUAL_TO, Octavia generated an ACL with\n\"path_end -m str \u003cvalue\u003e\"\nwhich doesn\u0027t work in HAProxy (https://github.com/haproxy/haproxy/issues/2567)\nUsing \"path_end \u003cvalue\u003e\" fixes the issue.\n\nCloses-Bug: #2066165\nChange-Id: I71309a0f7d57ccbe9bade5413fbd7fec01eed627\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit c64c5f54430c1fb647183b46e1fdaf25c643cb69)\n(cherry picked from commit b44406ea4b3e4376c40b339d53953ae76c8efd45)\n(cherry picked from commit 1f0be521fc1e80573ea20d6455882d3a9981eaeb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/711f97bc8c71c6238038e422ec0db07dacf5ccc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/711f97bc8c71c6238038e422ec0db07dacf5ccc9"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
