)]}'
{"id":"openstack%2Foctavia~962571","triplet_id":"openstack%2Foctavia~stable%2F2025.1~I920676dd4672b9bdcba8c771b8d7470fa88fe14d","project":"openstack/octavia","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-10-17 07:02:18.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-10-17 03:22:54.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I920676dd4672b9bdcba8c771b8d7470fa88fe14d","subject":"Fixed missing port_id in additional_vips API","status":"MERGED","created":"2025-09-30 09:02:41.000000000","updated":"2025-10-17 11:55:43.000000000","submitted":"2025-10-17 11:54:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962571","meta_rev_id":"9b017f6cd6c9d87dda354530cd2c3a964af40ed4","_number":962571,"virtual_id_number":962571,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-17 11:54:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"all":[{"value":1,"date":"2025-10-17 03:21:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":2,"date":"2025-10-17 03:22:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":1,"date":"2025-10-17 03:22:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-30 12:02:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-09 12:08:29.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"},{"updated":"2025-10-13 16:55:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"be378b411e51c86123e79779838f0120e82301af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-30 09:02:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"891bab28a42f7fb62f29e58eb8cad1b5660e6a6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 12:02:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/301e700a51004b5896e22f4270c2d164\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8ee7d690aa24478a297e8e0ecd9e771 : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0aaa1bf78414900aa1e5d3a4ddf89b0 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac3806105d824d72b274632a18ad82b6 : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/970b673f6d0a46fd80abd642585e0662 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdc13e2690764791991adee97b825ced : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33ab3765257a433a834a65976f36ddc2 : SUCCESS in 7m 07s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/25b9285906ff424f8466f9417f3475f6 : SUCCESS in 5m 31s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/8569291fe24d4545b2baf0fd72d851fc : SUCCESS in 6m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b7ba6fa5b8314fb98ff34b142f2dd160 : SUCCESS in 3m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/beba210438524559b7f5c185334e7328 : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7a63d6a69d9745a3ac2bbb09d257945e : SUCCESS in 7m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f9e83a669c8c446e832ec5f5afd55576 : SUCCESS in 8m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e5418f05b7f0456fb28dd4a2f8c2d2ff : SUCCESS in 1h 08m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/dfb004955f4a42f9be5d66fe0701f321 : SUCCESS in 1h 50m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8740619d42d74b83b16d8f35c4293aa9 : SUCCESS in 1h 08m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/108da171d78345c1b5de1203895e13e8 : SUCCESS in 1h 46m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02a4eb2290bd42c1b4f9a7c83193606f : SUCCESS in 1h 08m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/afbdbdecdf8341f4902a712ca7d27419 : SUCCESS in 1h 51m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8e1ec683c8e24f51b3eda6624b6b55af : SUCCESS in 1h 02m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/88b628a84da0459c99377ed9333c224e : SUCCESS in 35m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d2e0bf1700bd42528672bf01635af923 : FAILURE in 45m 21s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f66366cdad15479cb3b4d7b884ac4116 : FAILURE in 46m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0ba8f98c1844c14a8a0a934a687241f : SUCCESS in 42m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/80d2b8b2944d4520beffc9129ed2d830 : SUCCESS in 2h 31m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/92444b858ce34ed3ba4a1e8ffbbbe2c2 : SUCCESS in 2h 57m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3f7ccea001a6427599301b128e62a528 : SUCCESS in 2h 09m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/67f55f6255444005a36f3469598993dc : FAILURE in 37m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e08ab2189810b634b20e805eaa9c5a00ba2ca077","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-10-09 12:08:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2b5d885dbce8cf3124f69697d2b1a7966d475f09","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-13 16:55:36.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"115806f6ab8c28992471e87ebd558bac55ec7005","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-14 00:44:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d368f20df76d0731b5b2025ca360563173682b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 00:45:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1989fd939a4a7c250e6eec701281da8168d869f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 02:34:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c860ddad28f34fe3b3156f4d61dae0c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c4b347c18f944629f03adb3e9b964e8 : FAILURE in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6ec3228ace54c29bc3e8500865b3738 : SUCCESS in 6m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a9cdd38c81641c6aa47522376eb2d31 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e31d843370f7476faf2408c26e1d8630 : CANCELED in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e690133fb014084b87dc10c8eaf882c : SUCCESS in 3m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b21918b3b2b441b9b4193e5bde7996c2 : CANCELED in 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/432d75bf24714faeaac024bb82aa0858 : CANCELED in 7m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ae5bb92167ca47238720a19286164752 : CANCELED in 9m 25s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/bff4bc66a2834b18ae4449a82b3bafb6 : CANCELED in 10m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/deebae879afc4a4a9abe1829d234c36a : CANCELED in 10m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6376109dfc7a48ecb08220c2674ef55e : CANCELED in 10m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6ec8151e33746179d5f3fe9903196fe : CANCELED in 10m 46s\n\nWarning:\n  Comments left for invalid file octavia/tests/common/sample_data_models.py\n  Comments left for invalid file octavia/tests/functional/amphorae/backend/agent/api_server/test_keepalivedlvs.py\n  Comments left for invalid file octavia/tests/functional/amphorae/backend/agent/api_server/test_server.py\n  Comments left for invalid file octavia/tests/functional/api/drivers/driver_agent/test_driver_agent.py\n  Comments left for invalid file octavia/tests/functional/api/v2/base.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_health_monitor.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7policy.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7rule.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_listener.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_member.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_pool.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/functional/db/base.py\n  Comments left for invalid file octavia/tests/functional/db/test_models.py\n  Comments left for invalid file octavia/tests/functional/db/test_repositories.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_amphora_info.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_haproxy_compatibility.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_keepalived.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_keepalivedlvs.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_loadbalancer.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_plug.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_util.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/test_agent_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/health_daemon/test_health_daemon.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_haproxy_query.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_interface.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_interface_file.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_ip_advertisement.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_keepalivedlvs_query.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_network_namespace.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_network_utils.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_nftable_utils.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/haproxy/test_exceptions.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/haproxy/test_rest_api_driver.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/haproxy/test_rest_api_driver_1_0.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/health/test_heartbeat_udp.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/keepalived/jinja/test_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/keepalived/test_vrrp_rest_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_get.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_listener.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_updater.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/noop_driver/test_agent.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_factory.py\n  Comments left for invalid file octavia/tests/unit/certificates/common/auth/test_barbican_acl.py\n  Comments left for invalid file octavia/tests/unit/certificates/common/test_barbican.py\n  Comments left for invalid file octavia/tests/unit/certificates/common/test_local.py\n  Comments left for invalid file octavia/tests/unit/certificates/generator/local_csr.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_barbican.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_barbican_legacy.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_castellan_mgr.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_local.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_noop.py\n  Comments left for invalid file octavia/tests/unit/cmd/test_health_checker.py\n  Comments left for invalid file octavia/tests/unit/common/jinja/logging/test_logging_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/common/jinja/lvs/test_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/common/jinja/test_user_data_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/common/sample_configs/sample_configs_combined.py\n  Comments left for invalid file octavia/tests/unit/common/test_base_taskflow.py\n  Comments left for invalid file octavia/tests/unit/common/test_config.py\n  Comments left for invalid file octavia/tests/unit/common/test_constants.py\n  Comments left for invalid file octavia/tests/unit/common/test_data_models.py\n  Comments left for invalid file octavia/tests/unit/common/test_decorators.py\n  Comments left for invalid file octavia/tests/unit/common/test_exceptions.py\n  Comments left for invalid file octavia/tests/unit/common/test_keystone.py\n  Comments left for invalid file octavia/tests/unit/common/test_policy.py\n  Comments left for invalid file octavia/tests/unit/common/test_utils.py\n  Comments left for invalid file octavia/tests/unit/common/test_validate.py\n  Comments left for invalid file octavia/tests/unit/common/tls_utils/test_cert_parser.py\n  Comments left for invalid file octavia/tests/unit/compute/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/compute/drivers/test_nova_driver.py\n  Comments left for invalid file octavia/tests/unit/controller/healthmanager/test_health_manager.py\n  Comments left for invalid file octavia/tests/unit/controller/housekeeping/test_house_keeping.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/test_amphora_rate_limit.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/test_task_utils.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_amphora_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_health_monitor_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_l7policy_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_l7rule_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_listener_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_load_balancer_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_member_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_pool_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_amphora_driver_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_cert_task.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_compute_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_database_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_database_tasks_quota.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_lifecycle_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_network_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_notification_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_retry_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_shim_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/test_controller_worker.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/test_taskflow_jobboard_driver.py\n  Comments left for invalid file octavia/tests/unit/image/drivers/test_glance_driver.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_allowed_address_pairs.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_base.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_utils.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/statistics/drivers/test_update_db.py\n  Comments left for invalid file octavia/tests/unit/test_opts.py\n  Comments left for invalid file octavia/tests/unit/test_version.py\n  Comments left for invalid file octavia/tests/unit/volume/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/volume/drivers/test_cinder_driver.py","accounts_in_message":[],"_revision_number":1},{"id":"39442121dabbf3ddf426ad343b6644646e479457","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-17 03:21:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_29244\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":2},{"id":"ffd5e35a234ab660dc185e43db53a43cb82f2634","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-17 03:22:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2df05db36eedd8309a1849360a723c6d79aa0d93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 06:26:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1718163c7b2c41b1a1dd5025140be246\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a7dd52dc779464a97d099b27906884b : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b6a0551e5404e5a922e2b1294dbf745 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/115b5ce5647b472b809ed19e41a5ca29 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cffb006da5784947a26d5f0711ebe943 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f27bff26c2e8475f81c034ac9f5b54b8 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b5941cc2f774c82bf21967d46a09bd5 : SUCCESS in 5m 17s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/475ae4a85eca42a4aaf164c9643a98da : SUCCESS in 5m 54s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/716b819f38834b08b2c8a190723ca86e : SUCCESS in 7m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c7c6e31d7f794f3da8dc6c6c6f6f91bf : SUCCESS in 2m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2aea637d986249548e188662f60f3031 : SUCCESS in 7m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f679630dcade41e19856b2a802d64339 : SUCCESS in 7m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fd62fbb1ce074efeaa2a3c3dd14c11ef : SUCCESS in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/235e91e1672c48c2811b066c36ee30db : SUCCESS in 1h 46m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/13368c0a1960484f99ef2439208b6058 : SUCCESS in 1h 50m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b44deabd4a434c43a5cb18b3fb2bb645 : SUCCESS in 1h 04m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95e0711b1f1f44e1812aebdf6639874d : SUCCESS in 1h 46m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55b8a18c64f640cfb2b1ed076d23a526 : SUCCESS in 1h 07m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fc5095c2119b4ca6b0554ab93d69ea08 : SUCCESS in 1h 50m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d2a046623bab4dab804c6d25ad9e6df1 : SUCCESS in 1h 03m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a8a006e15a6c418ebf71236875dae727 : SUCCESS in 37m 47s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6cc2c121fb634438a1e9f7b49c4576ae : FAILURE in 38m 46s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e0234266a67d41e59c9404be543a8d9c : FAILURE in 42m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98c221d628d047b19ed46ec21ae88693 : SUCCESS in 34m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/43d8a78e5f564f0cb2e927b505a5a0b5 : TIMED_OUT in 2h 33m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fb816007dfcd41d9a67689cd1e9683d4 : TIMED_OUT in 3h 03m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dbea16cb5c8049f39bb0e369bb312d07 : SUCCESS in 2h 26m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/03016e4c141c4ae7b8e68287026adddf : FAILURE in 31m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7468334b262f221d618cb9c98a4f14bbdfe59b50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 06:27:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6633b04c865157052f80b1c201e4e514446b01e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 06:45:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da75394a211f40e994a18eb6ef7c0326\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10445c7e07f24622b0a214e12fe5f3e6 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e34a58113f54bddbf446c85bd4799c5 : FAILURE in 9m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8255275f1a146f985b2550916e7baf5 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f8d072786041c19e969f88467f9311 : CANCELED in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8934bfb2e6945399f6b78328af21009 : SUCCESS in 5m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8c9022a389b34551bf2519e43e58d250 : SUCCESS in 3m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9283b583e05e4f6c900b5fdafba1bf47 : CANCELED in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3defc0eca6544f758a299e6451513c70 : CANCELED in 11m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/25a6ef4f56624516b2bd9bd525bf4b90 : CANCELED in 15m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b60cc891bbc444608fa676fd8ae80e9c : CANCELED in 15m 39s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/749455e065084a76af3f424ef720d1e5 : CANCELED in 15m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c941874f7ecc481d97d06637a10566a0 : CANCELED in 15m 34s","accounts_in_message":[],"_revision_number":2},{"id":"a7369a9910106a8df90dea3de2a2e052e4d8c1af","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-10-17 07:02:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84317b5afb800182ba44502fea8071f97c0cbf28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 10:03:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6e0e38dc9c849239c659e10a718dc20\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e230156655c4ee88ab6be733067953f : SUCCESS in 13m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6677b32066484090a8c49a36a8de1345 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a56e18d58a349f287433da9054452e7 : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55d6df26784a46b2b04c843a4164c3cf : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255039c1783f4aaabfa235c894581a9f : SUCCESS in 13m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91594f215ed34d10aef2f003b8d32c40 : SUCCESS in 6m 39s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1997ca26bef24bee88f83ae6b2f3a60a : SUCCESS in 5m 03s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/6b3add6c9efa40f4b3d4e4c511ab8438 : SUCCESS in 7m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a37a5772f88a48de9d7853b60b0c73db : SUCCESS in 2m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9ce4d245d32a4daa949468ad91884826 : SUCCESS in 7m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dc7ff0c3ebce4cd0ba29568d30f63b8a : SUCCESS in 8m 41s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/435b8c9afd1f4d95bd5386f7a9394e8f : SUCCESS in 8m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d7fdd55122254310a5b5144a1af9c090 : SUCCESS in 44m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2dd5d126565d43edbf8cc7a8b1f05ee0 : SUCCESS in 2h 01m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c6e093f016b74a17b607651d3c6a85f8 : SUCCESS in 1h 14m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c51016fc672f4b6290b5ec6b10230bee : SUCCESS in 1h 59m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1cc2c16d7ec441cb80a829cdcbcf5cb : SUCCESS in 1h 21m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/acaf52e6adb84fe3871f688db0b71c34 : SUCCESS in 1h 59m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/62160d504d784bd0afde4bc1aac0fe1f : SUCCESS in 1h 14m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/162c69c9c71049faadabbf76490a2e54 : SUCCESS in 39m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8b54e5df2577429db29faad3cdac2bcf : FAILURE in 39m 36s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0d71b2f7b1254dadb8e073cefba11f38 : FAILURE in 40m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3aee94f052cc445d8013dd6ee7cb79c7 : SUCCESS in 44m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/76119152976c4c92bc4728432e2c56b1 : TIMED_OUT in 2h 33m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0cf64ca90fa64fe49b30f0ac22b0e5f2 : SUCCESS in 2h 59m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5674b2d130e04e4185cb9a0d30bb8cdc : SUCCESS in 2h 14m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/40b2845ae452431ea1744fd22cf7a31a : FAILURE in 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"249090ace56da2e6a49abfee29c5dc1575e82221","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 10:03:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"530db9190b1992b6b1a0ba90a406bdca6894d778","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 11:54:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a68a4d740f2f40a89c8783bc3bc89e12\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/998117ddd6dc424683cdbfc92ea99daf : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0bb5ce9908142ba82f38e06c1693ddc : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46585c9269744a628e13c2d659cb4544 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95243c85b86b43a39fdde989c2c8cf66 : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/685b98cff7bd4fb8a3778fb00e270268 : SUCCESS in 5m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9d2202fdb3b84de684e948e43a48406d : SUCCESS in 4m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9a2d978788504800b021e5dfe48506ce : SUCCESS in 8m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2cd5e5c2ed0c42799b1439c48ef99ac9 : SUCCESS in 1h 27m 23s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4da97462d67149c5a8f477e54822c876 : SUCCESS in 1h 50m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/18e9837940924758be6d10b9c9026380 : SUCCESS in 1h 09m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca6544f1d5ac472683182acc519ec580 : SUCCESS in 1h 48m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a87e9e3b6b1c4147a23a26c5a6553f13 : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":2},{"id":"cbed20a50c01d3fcd3ee66babbc1bf8245b8ab9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 11:54:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9b017f6cd6c9d87dda354530cd2c3a964af40ed4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 11:55:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58719c85e8554f7588168defa995898d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71a69ff2fb2c482988c50c7aaf03212e : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a962674e565144fe9ecfa9274ea4c5ab : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ea8128dec4b415d2df51d04b20887dfc58da575","revisions":{"9d2769df87670a0418a76ef5d5d064d072ca1d8c":{"kind":"REWORK","_number":1,"created":"2025-09-30 09:02:41.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/962571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/962571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/962571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/962571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/962571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/962571/1"}}},"commit":{"parents":[{"commit":"46251e4d6ccad5a750cb8260a232f0d766e999bd","subject":"Merge \"Reduce tune.ssl.cachesize for HTTPS terminating listeners\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46251e4d6ccad5a750cb8260a232f0d766e999bd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-10 05:56:43.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-09-30 09:02:22.000000000","tz":120},"subject":"Fixed missing port_id in additional_vips API","message":"Fixed missing port_id in additional_vips API\n\nthe additional_vip object includes a port_id field but this field is not\nreturned by the API.\n\nChange-Id: I920676dd4672b9bdcba8c771b8d7470fa88fe14d\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit bfa1abb2913695ab7ae17f21f8dbcf89f5f94854)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d2769df87670a0418a76ef5d5d064d072ca1d8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d2769df87670a0418a76ef5d5d064d072ca1d8c"}]},"branch":"refs/heads/stable/2025.1"},"8ea8128dec4b415d2df51d04b20887dfc58da575":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-17 03:21:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/71/962571/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/962571/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/962571/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/962571/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/962571/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/962571/2"}}},"commit":{"parents":[{"commit":"5a8d7fe13f074b30af88f26ff71d741a76660b77","subject":"Skip 4.0.{0,1} pylint releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a8d7fe13f074b30af88f26ff71d741a76660b77"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-10 05:56:43.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-17 03:21:55.000000000","tz":0},"subject":"Fixed missing port_id in additional_vips API","message":"Fixed missing port_id in additional_vips API\n\nthe additional_vip object includes a port_id field but this field is not\nreturned by the API.\n\nChange-Id: I920676dd4672b9bdcba8c771b8d7470fa88fe14d\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit bfa1abb2913695ab7ae17f21f8dbcf89f5f94854)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ea8128dec4b415d2df51d04b20887dfc58da575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ea8128dec4b415d2df51d04b20887dfc58da575"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
