)]}'
{"id":"openstack%2Foctavia~962572","triplet_id":"openstack%2Foctavia~stable%2F2024.2~I920676dd4672b9bdcba8c771b8d7470fa88fe14d","project":"openstack/octavia","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-10-14 02:34:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I920676dd4672b9bdcba8c771b8d7470fa88fe14d","subject":"Fixed missing port_id in additional_vips API","status":"MERGED","created":"2025-09-30 09:02:53.000000000","updated":"2025-10-14 02:35:37.000000000","submitted":"2025-10-14 02:34:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962572","meta_rev_id":"30b059b6edc16be7015b22d7aebe3350d6ca9ae8","_number":962572,"virtual_id_number":962572,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-14 02:34:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"all":[{"value":1,"date":"2025-10-09 12:08:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":2,"date":"2025-10-13 16:55:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":1,"date":"2025-10-14 00:44:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-30 11:35:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-09 12:08:23.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"},{"updated":"2025-10-13 16:55:47.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"a0c9292b1afbc39a10e1a3f2c98233d52c264233","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-30 09:02:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14528ad4392268f87c1b2b57ea3a0583bb39dada","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 11:35:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e33b2c1f0e54b3ab44560e6238480aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/082686263d4d4bd1a1e5811eeaf8ce74 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b53edbd38af45839b8be522959e98eb : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/823ef06eade64678ad9c9ad4c35f7760 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fca70e96e4384432bec252dfcb09ca04 : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08284323ae63491f93fca73242535009 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f326a50dec0c453d951e2d48d1508eb1 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07303259df49484b8378f16a30ee65bb : SUCCESS in 5m 48s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/da4d4e87deaf4a99aee6923285dc43ef : SUCCESS in 5m 01s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/5da645d464ee43fbb8ade28a5fa373fe : SUCCESS in 7m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c35a0259c7b8444daa309e5ccabc37f0 : SUCCESS in 4m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/639df63a0f944dbb9f3a245d72f230ab : SUCCESS in 7m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/db71351f3c3f4b5d86090d7f2d5cac88 : SUCCESS in 7m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b2f7ae9116b7420f9d1640799cb440df : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/240f1bddf09d4d94a880e3337f5dad54 : SUCCESS in 1h 16m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/17ede1cd05d54da3a9624e3f36bdc55c : SUCCESS in 1h 50m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/3c30fdf4144c459f8a36d4287d28aa02 : SUCCESS in 1h 29m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/a614b0bd7faf401d9108cc88bef783c5 : SUCCESS in 1h 57m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/91cd4124ff3b410ea4d425630267b711 : SUCCESS in 1h 12m 21s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7dc6253e5d3b482eb06fd2c58c1d52fb : SUCCESS in 35m 28s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/fb0924516f0a4066989a7549576a9243 : SUCCESS in 39m 36s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-2 https://zuul.opendev.org/t/openstack/build/0d5192db7541486fa6112b0c97ca5a36 : FAILURE in 22m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/036928abbef84abd8d3516289036affd : SUCCESS in 22m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-2 https://zuul.opendev.org/t/openstack/build/3ae3007667774351b29e53f5d8adad55 : SUCCESS in 2h 30m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-2 https://zuul.opendev.org/t/openstack/build/8c68d8e83b0649a9b50880077e2cd973 : FAILURE in 43m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-2 https://zuul.opendev.org/t/openstack/build/3a83f64cbfcf4f55ad71fbda876b46ba : SUCCESS in 1h 55m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-2 https://zuul.opendev.org/t/openstack/build/1c33ac53341e4568be62bfb0536f9565 : FAILURE in 29m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7cb1c8fc560ffa28ce20aabeb2e96f2af6997b90","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-10-09 12:08:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"981e3c580823ebffdcbf9f0a266a5cb501fabb0c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-13 16:55:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c87106a1e36258c77fedd2eb0d5309f0fdc5239","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-14 00:44:48.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a5f0c3a5177d797b4e0cb79a07bdf20c8bf8994","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 00:45:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e2514c80e47a5fd58c5c7a444c88659d4a31150e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 02:34:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3213aa6293d4c2fbe9396f0863a1c4d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d0baaf8938e4ff8997fed67d6fcbcfe : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/449616341ccb413985fab313532cb02b : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7bae485fb3e9411298e6f946915042f6 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5862999516cc493bb52c1138b1ee0d40 : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/278500ccbba74093aecbf0f1cf92c108 : SUCCESS in 5m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7fcda8852eb348d7afe4f59815be8a84 : SUCCESS in 3m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5ff2528e7ad9467fa9c1b21f13fc11b5 : SUCCESS in 9m 50s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/3f4f578f62db4ecfb8721b3e2ddbe7f9 : SUCCESS in 1h 20m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/47cc7a3f7d224a3d9d3ff5274e0db64e : SUCCESS in 1h 48m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/83871192009e4802b00b26dbcc53ffd3 : SUCCESS in 1h 09m 57s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7e7d4091fd7f4549ad551bd34c9441fb : SUCCESS in 36m 55s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7a6459fd2eb54ea9a0d754288b8b4ad4 : SUCCESS in 42m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b0c3796168cfc0dccee219b50e649c2c053e229f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 02:34:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"30b059b6edc16be7015b22d7aebe3350d6ca9ae8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 02:35:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0574048cabf54f3b855488bbc89ce094\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19619fa407ca4e28b89b20b8b8410cce : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a939ac6e4b9c4ca6a1273989a90fba49 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2c29c49c0bfe34b400f60e09993aa6a321ed559","revisions":{"a2c29c49c0bfe34b400f60e09993aa6a321ed559":{"kind":"REWORK","_number":1,"created":"2025-09-30 09:02:53.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/72/962572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/962572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/962572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/962572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/962572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/962572/1"}}},"commit":{"parents":[{"commit":"ccafae01828bef0d6f6635d559366ed48822baf5","subject":"Merge \"Reduce tune.ssl.cachesize for HTTPS terminating listeners\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ccafae01828bef0d6f6635d559366ed48822baf5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-10 05:56:43.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-09-30 09:02:48.000000000","tz":120},"subject":"Fixed missing port_id in additional_vips API","message":"Fixed missing port_id in additional_vips API\n\nthe additional_vip object includes a port_id field but this field is not\nreturned by the API.\n\nChange-Id: I920676dd4672b9bdcba8c771b8d7470fa88fe14d\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit bfa1abb2913695ab7ae17f21f8dbcf89f5f94854)\n(cherry picked from commit 9d2769df87670a0418a76ef5d5d064d072ca1d8c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a2c29c49c0bfe34b400f60e09993aa6a321ed559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a2c29c49c0bfe34b400f60e09993aa6a321ed559"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
