)]}'
{"id":"openstack%2Foctavia~962573","triplet_id":"openstack%2Foctavia~stable%2F2024.1~I920676dd4672b9bdcba8c771b8d7470fa88fe14d","project":"openstack/octavia","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"34451":{"account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"last_update":"2025-10-15 04:41:37.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-10-15 04:41:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I920676dd4672b9bdcba8c771b8d7470fa88fe14d","subject":"Fixed missing port_id in additional_vips API","status":"MERGED","created":"2025-09-30 09:03:04.000000000","updated":"2025-10-15 04:42:37.000000000","submitted":"2025-10-15 04:41:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962573","meta_rev_id":"40e297af47decf3c2333c24e3298c9ea5a1c3962","_number":962573,"virtual_id_number":962573,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-15 04:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"all":[{"value":1,"date":"2025-10-14 14:50:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":2,"date":"2025-10-14 19:44:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":1,"date":"2025-10-14 19:44:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-30 11:32:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-09 12:08:18.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"},{"updated":"2025-10-09 17:49:48.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"57a671b4b79f0bfa5def94cbfc2540fda4acd927","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-30 09:03:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8da60404f3bbcce5f9ac0f921baf85ffc451e323","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 11:32:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cb8f195bb4649b2a94800d016486196\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1366f0d00cd94042a5eec27935400e48 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd1b5de1c65c419c93170dadda4cb4bf : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd48970bd1214b47b3bf6b2bdccfcb9d : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46f3e29a9c954d41ae955b07b80bd3a1 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fc235f3d2344bfa9d0e52a17f2ce95d : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/161e078e205f4964b61ece533add6f5e : SUCCESS in 5m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b006b12a3d7b42759f6d9cae4e96345d : SUCCESS in 4m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3f2ea303ae8c4e939794cd1692fb122b : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9a5a1b1922414be19e7ab21d1baad792 : SUCCESS in 7m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7d1fad5032494887a6f5a7990e04190f : SUCCESS in 6m 25s (non-voting)\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/001cae44795e4010a0bbdee8b667acaa : SUCCESS in 56m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/a91c3f3290cb4e03b7d5e4b3314dddb1 : SUCCESS in 1h 21m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/adb3cc05cfa94d6185699b601472f91a : SUCCESS in 1h 09m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/c19bbb742ac5454c959fc8ea8fdd58ef : SUCCESS in 1h 41m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e8da47a3be0c49e1a4f78054c7c01167 : SUCCESS in 1h 39m 45s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/949590a09a9f4491a4032dd784292067 : SUCCESS in 36m 39s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/91a5d98be7a34d1897a06cf7356f135c : SUCCESS in 23m 54s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f32865d86b1a49999819c73cca0ada09 : FAILURE in 38m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e69c217ce5b34513b6457d512b8ee4ae : FAILURE in 45m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/35e37a67428e4f129a5608583095775e : SUCCESS in 39m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3849a6d9c7b84228afe2ca9aab4bb608 : SUCCESS in 2h 27m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-1 https://zuul.opendev.org/t/openstack/build/a54e19db09274cadaddd9e546b7b9614 : RETRY_LIMIT in 3m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-1 https://zuul.opendev.org/t/openstack/build/dd525664c23246cda4df402b388ecf53 : FAILURE in 36m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d92e4ec63e5c495c839f6da9a9b4aa4a : FAILURE in 19m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5a90b2516caf9dfbfeb2bfe0bf086dbf49f4a89","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-10-09 12:08:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"30189281279fe3240bece1a85303c67bae3c0bd4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-09 17:49:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48b774e8df910d307626152371068d3dd00adb8d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-09 17:50:10.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cdb850b2352638e762a08951ceb1bc157f776ae3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 17:50:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d221efbd41b2f925d7969a7c7e9e946cbce0ca53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 22:44:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6456974a22554fc4a973b5ba264344ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fa0ea0d2c4f4916a187f318c561788c : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38a89e5331724e588183ff03691e3609 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4046ea36377844aca26f0aa8a5e70fca : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4787fb9c1c354da690ae78d8b413d7db : SUCCESS in 11m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bea5f07f5684e879f1c62da8e85fd29 : SUCCESS in 5m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0f99c716ab354bb3b3e501a5c8054b25 : SUCCESS in 5m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b684171a2e2d47c9b4481ce965b7148a : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/4f700189edb346f1a87551e42c0048ff : SUCCESS in 1h 12m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/0f77bd16464c4e138d1f12eeef6885f2 : FAILURE in 1h 43m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/178f78e7fb234807baf1f2547f9dcf5e : SUCCESS in 1h 03m 16s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d7dd8d623ee64e6faa6ac3cbf360b610 : SUCCESS in 37m 39s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/6240515a885d49468285d36102e88138 : SUCCESS in 41m 02s","accounts_in_message":[],"_revision_number":1},{"id":"85ae4ec1897e3423fe741e596bd01c4f93b3c122","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-10-13 15:35:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1fe710684874ffba36b9fc0986861f27f1b75ee","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-14 00:52:52.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"296a7fd8abe6428d5407f18fb221744781445bbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-14 14:50:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_29244\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":2},{"id":"8df3d373269ffec41ef833fb3ade5a1cc4bf62e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 17:19:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/747c5b753e4941f58cb5a134a3b289fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eeb000394ccd452084358d2600d19115 : SUCCESS in 12m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f2c80fcdd494a2287c2f95c9775db43 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/601c380179fc4b60b25162c634647e68 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0225266e133479d8739629be2926550 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b41ee1ebbd24f68a86c17c9c5adfa30 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/868d84f47ce646aca24ed3fde8de6bfc : SUCCESS in 7m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d9178837149d4adb865c85b6d219068f : SUCCESS in 4m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b3a1efcdc43a4d899649c800a8a67baa : SUCCESS in 7m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bc20f129af2a48cf99126d427e180e2a : SUCCESS in 7m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6205d41c5c9a4fb2ac411e4d5f0870e6 : SUCCESS in 7m 37s (non-voting)\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/810a2208c10540eeb80984fb0b2c3d5c : SUCCESS in 1h 01m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/844394ad3c1043e6bf7c5656a61acf91 : SUCCESS in 1h 43m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/ad61c36d2efb4cc9947fc64d944ad5b8 : SUCCESS in 1h 01m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/477b9a1acb624e68a6ffbda9758f2d40 : SUCCESS in 2h 24m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/de8885f152b440fca1e26dcc4ca2c76e : SUCCESS in 1h 11m 55s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d378eca742874ed1951364df6a627719 : SUCCESS in 38m 03s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/52c9923436e141b5a54855cabf730421 : SUCCESS in 42m 09s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/635dd215ffc64bc382e921c842eb1b34 : FAILURE in 43m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-1 https://zuul.opendev.org/t/openstack/build/9e7d15c0c33d48559ed9b73ca70bc27a : FAILURE in 46m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/171a19dd8dea4b019cf03c410beb6b27 : SUCCESS in 33m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-1 https://zuul.opendev.org/t/openstack/build/79182161df744d41acd6f75051b743ff : SUCCESS in 2h 26m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-1 https://zuul.opendev.org/t/openstack/build/9ee9ff03b43944de8032a1cdb52aeab1 : RETRY_LIMIT in 4m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-1 https://zuul.opendev.org/t/openstack/build/305ad04886d9429f98fe75d2ff127e8a : FAILURE in 20m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-1 https://zuul.opendev.org/t/openstack/build/57c45dbaeff342de9fff4d079ba1b3b9 : FAILURE in 37m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ba530ada21a36ca8d6c5d99d10bf3b47342d6aa6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-14 19:44:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1df14f6a1086c8f9ae6a8844d798b565d99848ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 19:44:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"65552aa8e20db58b07db3d3c861678d15473bdb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 20:17:18.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e51aef41cd0d42a78304e7427c861233\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11613528cc8e434e9f94e57f5d4294d2 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7fd4fec3a68434fa9d8245c4f152aef : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c82f6643372449fca66754fb49619658 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0944eda26b334a61a0443492c62d24b8 : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bef51fd67fc24de19445a791650202ec : SUCCESS in 4m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6aad625762f84ab6b6a3ca59c3354565 : SUCCESS in 4m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e871a209eeb9421eb34efe489939ddff : SUCCESS in 4m 37s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/b3034d4d07c84c9a996de0515241190e : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/1458ee6559124ba1815f7c593ab6f1e7 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/cbd1bfebd4e04a10a5f883c0a566b3c4 : CANCELED\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e8c2c5edc28244eb8d8859c8d35538b0 : CANCELED\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/cce1d997220d4c8a867d8e865fd107bd : FAILURE in 31m 53s","accounts_in_message":[],"_revision_number":2},{"id":"e53758a2fc2c6b50a2206a95f3e9bcf8bdf76147","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-15 00:43:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c72f649a5fc150f58f2d4d72fad97b0b09e465cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 03:02:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e09b3aadc62c4d06bf737b807ec0eea1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db55d1f9d89040768293d624e00f07d6 : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/764f106e40c742aab2d35f673ba1cbf1 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d56b139ce6ed4a50adacf578b8a8c84d : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f9c26362d1746108bd6ed1d8518dc85 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ea984e95acc4e02bd0c5812c21fe872 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39892ca9a968414cb56795713979fc76 : SUCCESS in 5m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b7ed521f90ae4f3e8346c2f51ae02c28 : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c8c0701c982c46039ef175d164e89a96 : SUCCESS in 7m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f1109f052feb4acaa4499e2c3f8c54c0 : SUCCESS in 7m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/097ac14862624fc08a39017726a30de3 : SUCCESS in 7m 50s (non-voting)\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/bcf630eaa82546fc9f4005635379ebb2 : SUCCESS in 1h 09m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/5e477f5a346b45e9a74605139d74a6a1 : SUCCESS in 1h 40m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/ec92cc874a364ff38c52195077017c00 : SUCCESS in 55m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d306b4647bf143b48054e0ac3818c05e : SUCCESS in 1h 43m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/52800c82afd3445a9bfa3e0e2f76086b : SUCCESS in 1h 01m 37s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/58fb5fc65ab744388c977405d08e4011 : SUCCESS in 33m 40s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/1818fa52e57f4967a7b711f7c5accae1 : SUCCESS in 36m 29s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/439c1d195fcb48c28c0003824db1c50d : FAILURE in 35m 09s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-1 https://zuul.opendev.org/t/openstack/build/a3259c14a25542bd985854bce69ec22d : FAILURE in 41m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/5352e1ca3580435aaf88cce7b25275c9 : SUCCESS in 30m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-1 https://zuul.opendev.org/t/openstack/build/94891385de6542869cb342d0332d356d : SUCCESS in 2h 17m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-1 https://zuul.opendev.org/t/openstack/build/6df6d00185fe49428120094b55704437 : RETRY_LIMIT in 3m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e52aba2f2f024933b29009958a782e2c : FAILURE in 20m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-1 https://zuul.opendev.org/t/openstack/build/036afc0e2081475290cf8c8137995884 : FAILURE in 32m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"383144fe3efc52fa5a739c71129a1fee92ac8456","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 03:03:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cef4f3d2bb2f5accd5f36d9290e9e3349b686bda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 04:41:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10ce071f73e6468aa1ebe398d2319877\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b916b1398264acfac1f983931fa26be : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbc11ac9ff3349fe954569bded9d69a9 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9d3d4bc7482c4b11a34914e449dc7c17 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a4035905f4b4b1ebe394c0c6c96c2ce : SUCCESS in 11m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8c61d63363c441bbb40e36fa5bae94d : SUCCESS in 5m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7123944f60f6488e8fdd290e1435f926 : SUCCESS in 4m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dbf33037a1b645879ae70d57eb714876 : SUCCESS in 6m 42s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/543fa90028304d1088decffc3d987071 : SUCCESS in 58m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3169a456879842e587a375e4e77f484b : SUCCESS in 1h 37m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/6fb395be81d7494296a8454f49a1d0ca : SUCCESS in 1h 05m 27s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/8297329609bf4dd787a54522ec62d769 : SUCCESS in 41m 00s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/41b8d248da3d4f3e8d18cbcb8fa5360f : SUCCESS in 38m 17s","accounts_in_message":[],"_revision_number":2},{"id":"b2d687572f3190f6975014ab22f70f5fd8b668c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 04:41:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"40e297af47decf3c2333c24e3298c9ea5a1c3962","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 04:42:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e3a92d8908d4514b85a8375ecb0ee5e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/665a3b9372ec4b97aea355845653f388 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb8f562d6efa4fc6a846508ec6ac39ad : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"926e3afd9385ff5ad142067a72ec6801f9c29f4a","revisions":{"5c0825ab6de4a3251452ef1241a4503d49a9e101":{"kind":"REWORK","_number":1,"created":"2025-09-30 09:03:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/73/962573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/962573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/962573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/962573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/962573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/962573/1"}}},"commit":{"parents":[{"commit":"cc3dd1abe3e23071662494786a5e5bddb3538e89","subject":"Reduce tune.ssl.cachesize for HTTPS terminating listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cc3dd1abe3e23071662494786a5e5bddb3538e89"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-10 05:56:43.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-09-30 09:03:00.000000000","tz":120},"subject":"Fixed missing port_id in additional_vips API","message":"Fixed missing port_id in additional_vips API\n\nthe additional_vip object includes a port_id field but this field is not\nreturned by the API.\n\nChange-Id: I920676dd4672b9bdcba8c771b8d7470fa88fe14d\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit bfa1abb2913695ab7ae17f21f8dbcf89f5f94854)\n(cherry picked from commit 9d2769df87670a0418a76ef5d5d064d072ca1d8c)\n(cherry picked from commit a2c29c49c0bfe34b400f60e09993aa6a321ed559)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c0825ab6de4a3251452ef1241a4503d49a9e101"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c0825ab6de4a3251452ef1241a4503d49a9e101"}]},"branch":"refs/heads/stable/2024.1"},"926e3afd9385ff5ad142067a72ec6801f9c29f4a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-14 14:50:48.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/73/962573/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/962573/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/962573/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/962573/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/962573/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/962573/2"}}},"commit":{"parents":[{"commit":"2dcd82a827ae8bc6208f7d2e3ebe9b51c8fc66f5","subject":"Merge \"Fix race condition in cascade delete\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2dcd82a827ae8bc6208f7d2e3ebe9b51c8fc66f5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-10 05:56:43.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-14 14:50:48.000000000","tz":0},"subject":"Fixed missing port_id in additional_vips API","message":"Fixed missing port_id in additional_vips API\n\nthe additional_vip object includes a port_id field but this field is not\nreturned by the API.\n\nChange-Id: I920676dd4672b9bdcba8c771b8d7470fa88fe14d\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit bfa1abb2913695ab7ae17f21f8dbcf89f5f94854)\n(cherry picked from commit 9d2769df87670a0418a76ef5d5d064d072ca1d8c)\n(cherry picked from commit a2c29c49c0bfe34b400f60e09993aa6a321ed559)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926e3afd9385ff5ad142067a72ec6801f9c29f4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/926e3afd9385ff5ad142067a72ec6801f9c29f4a"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
