)]}'
{"id":"openstack%2Foctavia~963002","triplet_id":"openstack%2Foctavia~master~I675b48404e3b81602d792d653db348df0968e99c","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-10-16 12:59:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I675b48404e3b81602d792d653db348df0968e99c","subject":"Fix devstack cleanup for log offload location","status":"MERGED","created":"2025-10-03 17:55:08.000000000","updated":"2025-10-16 13:00:32.000000000","submitted":"2025-10-16 12:59:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963002","meta_rev_id":"77764888ba30630572c28c21e28bec0f088d533d","_number":963002,"virtual_id_number":963002,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-16 12:59:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"all":[{"value":1,"date":"2025-10-09 12:08:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":2,"date":"2025-10-14 13:46:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":1,"date":"2025-10-15 09:30:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-03 20:59:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-09 12:08:58.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"},{"updated":"2025-10-14 13:46:58.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"903eb3877c35c23de29c15839488a5208fc7ed12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-03 17:55:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc4aafd89f2d52c1411e0a49e3c9c52db5868594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 20:59:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26e9116305524bbcae1d955989d747d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6dcad27d80b4e588dcaf29192d8ba4c : SUCCESS in 14m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34a5b82d0806402bbe3dd5f875ea696c : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c50e2373d086475cb45cb219e9ec341b : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09e7c8122f854b2e946a5f170d99810d : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2dc083882a5d4ab4a7bed66e37bc8f5d : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a10acd1bb1f492baea90df875f781b9 : SUCCESS in 13m 28s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7c2ba44851ee4b0795782345e367a044 : SUCCESS in 5m 50s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/ac1d075c4cc8462d95cb021e140203a6 : SUCCESS in 8m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4b34b952f1e54cdfa4c505fe73e060b3 : SUCCESS in 4m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d716428ebc1c4c6ba9eae40ad6a8dcf9 : SUCCESS in 7m 48s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fed02cae5b32463b925969483263a12a : SUCCESS in 7m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ea7e4cde92614e30be8b1a14875254b8 : SUCCESS in 1h 27m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/cee92f0d86ce4967ac0a4f6b5e8b4377 : SUCCESS in 1h 25m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/157f6ad7084d46e297677aa649a1de51 : SUCCESS in 1h 10m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8ffe092c86e241caa9c08e3c26ee40c0 : SUCCESS in 1h 27m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2654a30040bb4abfa697e12f69fc477c : SUCCESS in 1h 11m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a094c908d22d4e5a9b7236363dd6b5aa : SUCCESS in 39m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9c84ae8d7ff44221aa64b2dd3ae144f1 : SUCCESS in 45m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/361867be3aa542be92385310d843df23 : SUCCESS in 34m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9f0be1483a6e443c9fc9f3e0708734e0 : SUCCESS in 2h 33m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8ae7c741ff7f4a5abab0650867d5c277 : TIMED_OUT in 3h 02m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0a1d72cc77054121a1458eb359450605 : TIMED_OUT in 2h 32m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b163bc9022d1a1f0f1d24b98ea40634c6a0ff6e5","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-10-09 12:08:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"99ca2b7112906a2f8b882101d38b370a2ee83b54","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-10-14 13:46:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"692f2f67779d601b36da6ea0915cf7cb6bffa3b4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-10-15 09:30:16.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7b19203d24c6bb45b5c345a57be2a0457e2acdb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 09:30:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd354b49cbad9875da8193b2b686aa9141e5fac0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 09:41:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02ae5f081941486fb1d1c9a2d2752ec3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dfe33f73553459b94d894e11bca3451 : FAILURE in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aad95a45b0b54d0fb77e49d39577d8cb : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e51c481dca914185a19d382ad1eaa02c : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f79f6e52b743859e93ad0cad023120 : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/316d5255097145318c2f7dc0a4f124ed : SUCCESS in 2m 50s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e134e05654f04f08903c69c93c13693f : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/68c07111fd1141e1a0b20956c2726ac0 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6893d4d5b36f4822a492b92b0f383525 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f50a4f351dd746ccacdac657ba9f3059 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/76b67d4ee3934ac48b8e7665bd1502cc : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/73eddfab39d042709bd8ec8ded9f9e70 : CANCELED\n\nWarning:\n  Comments left for invalid file octavia/tests/common/sample_data_models.py\n  Comments left for invalid file octavia/tests/functional/amphorae/backend/agent/api_server/test_keepalivedlvs.py\n  Comments left for invalid file octavia/tests/functional/amphorae/backend/agent/api_server/test_server.py\n  Comments left for invalid file octavia/tests/functional/api/drivers/driver_agent/test_driver_agent.py\n  Comments left for invalid file octavia/tests/functional/api/v2/base.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_health_monitor.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7policy.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7rule.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_listener.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_member.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_pool.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/functional/db/base.py\n  Comments left for invalid file octavia/tests/functional/db/test_models.py\n  Comments left for invalid file octavia/tests/functional/db/test_repositories.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_amphora_info.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_haproxy_compatibility.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_keepalived.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_keepalivedlvs.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_loadbalancer.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_plug.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_util.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/test_agent_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/health_daemon/test_health_daemon.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_haproxy_query.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_interface.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_interface_file.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_ip_advertisement.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_keepalivedlvs_query.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_network_namespace.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_network_utils.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_nftable_utils.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/haproxy/test_exceptions.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/haproxy/test_rest_api_driver.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/haproxy/test_rest_api_driver_1_0.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/health/test_heartbeat_udp.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/keepalived/jinja/test_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/keepalived/test_vrrp_rest_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_get.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_listener.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_updater.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/noop_driver/test_agent.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_factory.py\n  Comments left for invalid file octavia/tests/unit/certificates/common/auth/test_barbican_acl.py\n  Comments left for invalid file octavia/tests/unit/certificates/common/test_barbican.py\n  Comments left for invalid file octavia/tests/unit/certificates/common/test_local.py\n  Comments left for invalid file octavia/tests/unit/certificates/generator/local_csr.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_barbican.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_barbican_legacy.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_castellan_mgr.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_local.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_noop.py\n  Comments left for invalid file octavia/tests/unit/cmd/test_health_checker.py\n  Comments left for invalid file octavia/tests/unit/common/jinja/logging/test_logging_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/common/jinja/lvs/test_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/common/jinja/test_user_data_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/common/sample_configs/sample_configs_combined.py\n  Comments left for invalid file octavia/tests/unit/common/test_base_taskflow.py\n  Comments left for invalid file octavia/tests/unit/common/test_config.py\n  Comments left for invalid file octavia/tests/unit/common/test_constants.py\n  Comments left for invalid file octavia/tests/unit/common/test_data_models.py\n  Comments left for invalid file octavia/tests/unit/common/test_decorators.py\n  Comments left for invalid file octavia/tests/unit/common/test_exceptions.py\n  Comments left for invalid file octavia/tests/unit/common/test_keystone.py\n  Comments left for invalid file octavia/tests/unit/common/test_policy.py\n  Comments left for invalid file octavia/tests/unit/common/test_utils.py\n  Comments left for invalid file octavia/tests/unit/common/test_validate.py\n  Comments left for invalid file octavia/tests/unit/common/tls_utils/test_cert_parser.py\n  Comments left for invalid file octavia/tests/unit/compute/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/compute/drivers/test_nova_driver.py\n  Comments left for invalid file octavia/tests/unit/controller/healthmanager/test_health_manager.py\n  Comments left for invalid file octavia/tests/unit/controller/housekeeping/test_house_keeping.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/test_amphora_rate_limit.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/test_task_utils.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_amphora_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_health_monitor_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_l7policy_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_l7rule_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_listener_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_load_balancer_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_member_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_pool_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_amphora_driver_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_cert_task.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_compute_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_database_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_database_tasks_quota.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_lifecycle_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_network_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_notification_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_retry_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_shim_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/test_controller_worker.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/test_taskflow_jobboard_driver.py\n  Comments left for invalid file octavia/tests/unit/image/drivers/test_glance_driver.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_allowed_address_pairs.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_base.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_utils.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/statistics/drivers/test_update_db.py\n  Comments left for invalid file octavia/tests/unit/test_opts.py\n  Comments left for invalid file octavia/tests/unit/test_version.py\n  Comments left for invalid file octavia/tests/unit/volume/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/volume/drivers/test_cinder_driver.py","accounts_in_message":[],"_revision_number":1},{"id":"b5d017dbbb7d664eb56157f039f6e1305b56fc04","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-10-16 08:04:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1fe5cefd5eed9d4dbaef0bb0562d88b1ef615c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 11:05:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c252bfd7e0724641ade69372cfad353b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5f36ef4275d4f0b860dd9e7b7cfb5ee : SUCCESS in 16m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86b30fcabc134b08980118e87bcd074e : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d5afa39c2fc4375b5224285d06c0c29 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e6806edcb3e49a68fb9f6a6af235a25 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b52a99f78e24b56b350589ea8822970 : SUCCESS in 13m 28s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2a81c628c03a4bfeb17868f9843e8e5f : SUCCESS in 4m 58s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/fc9346876cdb4b82884f288e550b8a28 : SUCCESS in 7m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cd2b529a210c4fe5a61b709069c6e1fa : SUCCESS in 2m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/26cf7f0beb694e769bc30e5dbbe9eafe : SUCCESS in 4m 41s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1887bc3371304eb1b3db4fa96fd9f675 : SUCCESS in 5m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ecc06f79a8d042b5b6e9d27dcecaec21 : SUCCESS in 57m 28s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f0b49f5a7eca417a80c8567e4a24a58a : SUCCESS in 1h 53m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/024e2cdc80b14d2f8e7380072ede8ed0 : SUCCESS in 1h 05m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fe4e3fdb2cf94ff4aa3f872e19879f10 : SUCCESS in 1h 57m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e947749377294c13ad0670cb68019331 : SUCCESS in 1h 11m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39ceee5ffce6485aa61301c18154b313 : SUCCESS in 35m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/01819d86aaca417bbf114246ac573da4 : SUCCESS in 44m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b4c0e345b5f6464cbfa855617d131fa9 : SUCCESS in 34m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ae156e396690486c8dce0e7033718a35 : TIMED_OUT in 2h 33m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a488dabfde27405c83b4c9d11cb087b6 : SUCCESS in 2h 59m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0dfc2adcacf54a0388204400e13fcca6 : TIMED_OUT in 2h 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"abb9a1bf017871b17a5a25074208d2c1d8859603","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 11:06:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f7b72b29789e4fd4d73587b650fb76abdbbbe31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 12:59:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f69e3072ff9c4157b48becf5b4ebfe67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f2db571e1642f9a873583804151368 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54aa753ff13d455b8895dd05f970332c : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/763dd8d706824fb3bc267f30b8a6e119 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a09ea4215ee4403986564f6f4413dfd7 : SUCCESS in 14m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9edd6f642de24366968744ba801f8f5e : SUCCESS in 3m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9fd42845dc904dbeadbfb0ae448e9c5d : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b252bc3684304be1b6f013ecd10a2edd : SUCCESS in 1h 21m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/46ef2231bb7a40b2aa41aaf579075780 : SUCCESS in 1h 52m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b91a9050788c4cf3a92da374a0cdbc86 : SUCCESS in 1h 09m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b07bbd438acc41cb95bf30f5ee349fc4 : SUCCESS in 36m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2f3bd93f8e504721b6ed940f4d711ae7 : SUCCESS in 48m 00s","accounts_in_message":[],"_revision_number":1},{"id":"990b990b767eaf55a6429000ff65db1d2e725aa1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 12:59:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"77764888ba30630572c28c21e28bec0f088d533d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 13:00:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4f2675c594442e1aeff1ef3e3030f0a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fb459fd9f714cf8b3ff7a4034a2cec4 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54e4c3e8835c5a16e762207340fe2b8c36bbadc7","revisions":{"54e4c3e8835c5a16e762207340fe2b8c36bbadc7":{"kind":"REWORK","_number":1,"created":"2025-10-03 17:55:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/02/963002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/963002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/963002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/963002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/963002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/963002/1"}}},"commit":{"parents":[{"commit":"078db9c27784c8af7c8aefa15a711df7a5285b7e","subject":"Merge \"reno: Update master for unmaintained/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/078db9c27784c8af7c8aefa15a711df7a5285b7e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-10-03 17:50:00.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-10-03 17:55:07.000000000","tz":0},"subject":"Fix devstack cleanup for log offload location","message":"Fix devstack cleanup for log offload location\n\nA previous patch[1] moved the log offload file location to resolve\ndirectory access permissions. However, if devstack is re-run the\nplugin.sh will run into errors that the directory and symbolic links\nalready exist. This patch fixes that issue such that restacks do not\nrun into these errors.\n\n[1] https://review.opendev.org/c/openstack/octavia/+/955959\n\nChange-Id: I675b48404e3b81602d792d653db348df0968e99c\nSigned-off-by: Michael Johnson \u003cjohnsomor@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54e4c3e8835c5a16e762207340fe2b8c36bbadc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54e4c3e8835c5a16e762207340fe2b8c36bbadc7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
