)]}'
{"id":"openstack%2Foctavia~963656","triplet_id":"openstack%2Foctavia~master~Ie581710636fec38daebe6df560844f14e138e21c","project":"openstack/octavia","branch":"master","topic":"c10s","attention_set":{},"removed_from_attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-02-17 10:09:44.000000000","reason":"\u003cGERRIT_ACCOUNT_38562\u003e replied on the change","reason_account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-02-17 13:55:30.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Ie581710636fec38daebe6df560844f14e138e21c","subject":"Add support for centos-10-stream \u0026 rocky 10","status":"MERGED","created":"2025-10-10 07:15:06.000000000","updated":"2026-02-17 15:39:59.000000000","submitted":"2026-02-17 15:38:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963656-c10s","meta_rev_id":"40c83c986e9c75ee18f7450d5ea627a46f8f14e5","_number":963656,"virtual_id_number":963656,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-17 15:38:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-02-17 10:09:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-17 13:55:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-17 13:55:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-15 15:18:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-13 09:45:15.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"0e693748401ea495e4c37701243220838107d1c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-10-10 07:15:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53f88451241b0e00589e9a169e7245c14e6fb167","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-10-10 07:15:17.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"d5dbbfc0364f75862bbe6c6f9fc73f2e883a8fdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-10-10 13:42:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5173157e17e4c115a9fe883c356d15b28a5561f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-10-15 14:53:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7bd8986e192d7b09f5e3f90a10c2a09ff786b7f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 15:18:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0a69ae72c8c4b00b50b58d3f3677aab\n\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/92e2c4fa796f437e99c169fd0d74db89 : FAILURE in 22m 42s","accounts_in_message":[],"_revision_number":3},{"id":"a924239c5208b81deb4bd3852557adde38b5fd4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-10-16 08:06:02.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"01f2f53b0662c96a5dd4a978ef4d38fa9b93a9d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 09:59:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21df43c3507c4b678287da0faac7f468\n\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3a75281d4c044d648669500e22586e33 : SUCCESS in 1h 51m 54s","accounts_in_message":[],"_revision_number":4},{"id":"52e5640ff34d75b873bb0f112068a726f7d2a637","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-10-17 07:32:47.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"3effb8fbef0538c3d47e1d38e3960f11ab0dd855","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-10-17 07:33:19.000000000","message":"Topic set to c10s","accounts_in_message":[],"_revision_number":5},{"id":"0adceb5977d4568e8199f079958c60981d389c7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 10:38:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/534c867448a846e991c4466867c91d3a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64ec9106b4b242718730d4d2672cabcf : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff30dfff6b1c43f6acd4d04def3d6fff : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f9044d3bba04ae69a5383551e5b6749 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38ef21be686d485a93a457282e09948d : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2d38a19899a413f95ece3f2746dc562 : SUCCESS in 7m 44s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/876b52a818d344b18dcd6dcaa2eaf858 : SUCCESS in 4m 39s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/b20f8b8f1ce5411cab32fc2148565a7b : SUCCESS in 8m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/844a01ca6da5457f884f51e3610f773b : SUCCESS in 4m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a6e225a892a44f12afdba0d38b348b19 : SUCCESS in 8m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/544df52b0c6b479b91b6e8d00bb1dab8 : SUCCESS in 8m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/16ea59b8f9674c7a96dd624509757696 : SUCCESS in 1h 34m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7eae2f9392b444eda2aec3da58908a98 : SUCCESS in 1h 51m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b7354b5a93664498ba5f823ef4c8cb25 : SUCCESS in 1h 14m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/367ecf7a377f49d78041bd94a4bb759a : SUCCESS in 2h 05m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/18aac8fa0a484474817b145b4af3f6a3 : SUCCESS in 1h 05m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2bb363d1011b4aac8971f1389ce11c46 : SUCCESS in 37m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/315f3ed989d44709a625be4c2c08997d : SUCCESS in 52m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3362adcf1a9b4a59a0bbf51f61bf49b3 : SUCCESS in 34m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/13f5d54e711340549b1e06e0914c7423 : TIMED_OUT in 2h 33m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/beef854368a947d3891bd2842252303d : TIMED_OUT in 3h 03m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d1e446861b414a24a95ba9ccea922df9 : TIMED_OUT in 2h 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2652e29af51321f538b3c888f2227065c5943ea1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-11-04 08:51:18.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"71ca8b45fbba8836a74bc4bcd88579446276b287","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-11-04 08:51:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"edf1bf7e8f5191c88c772747670121a0bd7793c1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-11-07 09:31:58.000000000","message":"Patch Set 5: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e5ce1fb5daf2bb56e0289fbd588cb4613f881c55","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-11-12 08:45:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9cce03ea9e66736b84d8bdc7b8192fce83ba1ac1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-11-12 08:46:04.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"e874b2f9ffd6284714eb883a23ab9a125359b7ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-11-12 08:50:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4ef7d915ef61eca5a185756b78386da6418fa831","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 11:57:01.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12a3591c82904e1e9586cf38d15505d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/006ca8d171604ece9f2a6e8f9631d283 : SUCCESS in 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7968db151ba24f3c8a26b20ed6207038 : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d33d420a044492b8652d8999f6513af : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d044ad7b071548e2a7019f7af4b673ec : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e68cf956a1a45a1bc73c0977974dd50 : SUCCESS in 13m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d310e73406740cc958bc44fcdb61ac7 : SUCCESS in 5m 44s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/031ea22de0d04bd0bdceb4c8e3b02ae5 : SUCCESS in 7m 48s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/c7517bd9b50d40b290af73cdf9ee18bb : SUCCESS in 8m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/63d0f5edb9bf45f6b4c12c1198496739 : SUCCESS in 4m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/78517ee923614071ac89d438c5492260 : SUCCESS in 7m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2fe5788eb5ff41b5b6b1f2e300623c45 : SUCCESS in 7m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f0f776e14c4c498c81beda3862ac9c2f : SUCCESS in 1h 32m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f919d4195c84431383624a4de4425922 : SUCCESS in 1h 57m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e6c56d8807ae4310bcadf741910cfba8 : SUCCESS in 1h 05m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4236f60453044f7f96d8153c2da943a9 : SUCCESS in 2h 00m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8b2860d4d45645619f7a3ee3e95355da : SUCCESS in 1h 20m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/203b52a64b4a45cb91aa5a0f6596d7f4 : SUCCESS in 39m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1ef8e388b5fc46099610ba3a128d1e80 : SUCCESS in 46m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/085e0ced82084ae1976d78b84897ff82 : SUCCESS in 38m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/377fc895e87241508ef6f875a71e672c : TIMED_OUT in 2h 33m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8109e252622a4379aba24003b7a508ca : TIMED_OUT in 3h 04m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/351a3dfb99df4ccb91ef45908df936bb : SUCCESS in 2h 31m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7c52f59e39b95c68273e3040de705d5688a69564","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-06 12:43:27.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"11f622e5b365e0bccaacf1fcc1b23e9f86f451d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 15:37:35.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee615e97df4248e7b61246cf6bb9a6a6\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bb34f5d411f04a5b862970c83dff1b96 : SUCCESS in 1m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43f69f0e397c4fc295d47ffe6acb8d2d : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/782327320e1d4bddb9e1a4b357f0f6aa : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/751f5f69b33640fb81d02c5d3b34550f : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c221da50ac8e4a7db2487e6b700a861e : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/939f994bdb6844e58a51295041af5aee : SUCCESS in 13m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad5127382eaa4653b422d33ad103a198 : SUCCESS in 5m 56s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5b87ff032e924ab2afd559d5c6b1c944 : SUCCESS in 3m 25s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/2c8631404c9e4685958b5f9d3655ce1b : SUCCESS in 4m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ea10dacf450f4443a09f58cc0c1b6528 : SUCCESS in 3m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/edfda25b2ae7479683a8343e866d034d : SUCCESS in 7m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/28aba2cf064c400ab3cef53155c178e5 : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7940c59230734b5d8351d366a424a947 : SUCCESS in 36m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/bf81bf0450474b2ca8cfafb0e904ad93 : SUCCESS in 1h 33m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/da395b69d6754ac9ba962a7983c3224f : SUCCESS in 1h 04m 13s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ab33470c363a42a49da84ccd1450be1a : SUCCESS in 1h 36m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1bc55f67e87a42e3bda051694e49fccf : SUCCESS in 59m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/761396823d824c8883596dd806adbaf6 : SUCCESS in 36m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9f3590878aac4b968149c7505539c586 : SUCCESS in 41m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa0d261ff97444fcb7f1d77bd910eca1 : SUCCESS in 33m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/72d1f5071c0a4aefb6a56a17410dd345 : SUCCESS in 35m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a762d9ee1fb7451c9ea52105caa71f17 : SUCCESS in 2h 46m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/081b8e04498041dc96a1308fb3a636d5 : SUCCESS in 1h 31m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2db4875bfd6e5f94242937d9a8044a71d7c2deca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-08 13:46:42.000000000","message":"Patch Set 9: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"ff0dc8e5c7760f7fe65460ea13335d7360f60a10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-08 14:05:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"39e72737ae3e58323bd98fa23408ad4a2e281ad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 14:06:19.000000000","message":"Patch Set 10: Verified-1\n\n(4 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/octavia on branch master.  The\nproblem was:\n\n  The nodeset \"centos-10-stream\" was not found.\n\nThe problem appears in the the \"publish-openstack-octavia-amphora-image-centos-10-stream\" job stanza:\n\n  job:\n      name: publish-openstack-octavia-amphora-image-centos-10-stream\n      nodeset: centos-10-stream\n      parent: publish-openstack-octavia-amphora-image\n      description: |\n        Publish CentOS 10 Stream based amphora image to tarballs.o.o.\n      vars:\n          amphora_os: centos\n          amphora_os_release: 10-stream\n  ...\n\n  in \"openstack/octavia/zuul.d/jobs.yaml@master\", line 21","accounts_in_message":[],"_revision_number":10},{"id":"83491097d9a807754b041b557321d587a8aa89cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-08 14:10:10.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"8c4bc5bc4615c1a447c2e5c8a11ea48964d44c62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 14:12:12.000000000","message":"Patch Set 11: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/octavia on branch master.  The\nproblem was:\n\n  The nodeset \"centos-10-stream\" was not found.\n\nThe problem appears in the the \"publish-openstack-octavia-amphora-image-centos-10-stream\" job stanza:\n\n  job:\n      name: publish-openstack-octavia-amphora-image-centos-10-stream\n      nodeset: centos-10-stream\n      parent: publish-openstack-octavia-amphora-image\n      description: |\n        Publish CentOS 10 Stream based amphora image to tarballs.o.o.\n      vars:\n          amphora_os: centos\n          amphora_os_release: 10-stream\n  ...\n\n  in \"openstack/octavia/zuul.d/jobs.yaml@master\", line 21","accounts_in_message":[],"_revision_number":11},{"id":"8a4514cdd85749f0082f100a4387002bfc0db1f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-08 14:30:04.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"ed72fecbe3a9d3fe9ee388baf658ad794e9b97d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 17:32:28.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d696f3b6773c4ba4bdb90acb11af1d27\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/12d5770a36774e8d8769f67dbff0b588 : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a95b0a2685f425a885541927c21f185 : SUCCESS in 16m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93569bb900f643388d322dd18fcfb1d8 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/895c3aca1e5d46e5b4f6ce80ab1b33a2 : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65a9ae4c27a348ea9769b0396fb212f6 : SUCCESS in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcfc2f3754224474a9b6f0c94449eaf3 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cff91ba57b014799b9882b5a2ee6961a : SUCCESS in 5m 27s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ad5856591314477696e0e02e99af6cd6 : SUCCESS in 5m 32s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/08d0778a4a6045f3b5e00e64aef950ce : SUCCESS in 7m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6cf32b72b2494e1ab38116691c44eabf : SUCCESS in 4m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dec7b4af7f7d4c33b8b9bfb43e317a38 : SUCCESS in 6m 48s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2dc0dfe9c88c4e8ea99eef9a8e6af096 : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5d1e83475461478e8cd99d50384a2321 : SUCCESS in 1h 29m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7746a12b10ec4359b457fe1bbef633db : SUCCESS in 1h 32m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/42b8e75a40304e32bb3f2be5a3226b9b : SUCCESS in 55m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a3eb52e8abb1478a9a426b1f29102628 : SUCCESS in 1h 38m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/80f2a4d7d5db4d37a215ec2f1b7f6d24 : POST_FAILURE in 1h 03m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e23165c6d9dc431b984b5e80f3fdf151 : SUCCESS in 35m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1da4f83c6be24ef682b6785fb94c6ef2 : SUCCESS in 42m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/34f70578f00d452c83117427ba6c764d : SUCCESS in 33m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f4b0c2cb6d954f8d8f97e723ebbbdfe2 : SUCCESS in 34m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c104467349f74e6092779bdea5643683 : SUCCESS in 2h 34m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8d1bb8e77bc5467ab6e86dc9543c081c : SUCCESS in 1h 31m 43s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"57b57f4ab599c4087c7e59a604361116f3e00c41","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-12 13:12:15.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"5d0b1030f2dc672c655396dcfbc41d72e16eade9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 16:13:04.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98fb675df5ff4e70857e00404dd61d80\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/49d6a65ece384d98bc3b71970ddd6c71 : SUCCESS in 1m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f56200da1dcb4d358c2f6519f44e0076 : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf1b4011e6f64fc99930e3cecbe26e68 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cba3938c4a394642a3bf6db0d07e4290 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38da6d4ca8264a078773a70dbe81d3c8 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b26d12761e0f4a0389db14c070244596 : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6ff1fe8331a4ec6ae09346b23744617 : SUCCESS in 4m 33s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a96b09a37d8a426ca63a437a460fd626 : SUCCESS in 4m 46s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/c44b4bb6aebc4f6d9e2c84f03942f403 : SUCCESS in 5m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c91d9a8839894adbb39ebc3dd21527bf : SUCCESS in 2m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/041203d3d1344c2d96b366367a15e8f9 : SUCCESS in 6m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a429ec4355d74c7f8356182a4d3eb67d : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b0973161bf054cc887012540f2436580 : SUCCESS in 1h 15m 57s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a02e4cb41cf246a69ea625b6d46d54e7 : SUCCESS in 1h 34m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e321ceed90a04678851ace93089d300d : SUCCESS in 1h 01m 34s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0a3c9356fd754658a263871206f8c7ff : SUCCESS in 1h 38m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5589da70e3454aafa8d3dad307533306 : SUCCESS in 1h 11m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/43c61e6d4ff14f7495e271db3604e80e : SUCCESS in 36m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/973b441781ba4355b898cdd30fda6f24 : SUCCESS in 43m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/487c66fdab064e1f86aaa809e558d07c : SUCCESS in 33m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1f5f926bffe84dcbad8dd7be5e5daf00 : SUCCESS in 34m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/db57a732ed8f405181ec0d5dec1bea50 : SUCCESS in 2h 59m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/94bce480287a45b992c972965fd103d7 : SUCCESS in 1h 32m 35s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c47ad0264811226b85caa10730ed912ef0565fa7","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-01-13 09:45:15.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"d5c2c23e44a7ac7c014952c31a224d7983e170a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-16 13:58:01.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"0f1782bae742fe46552198a43497ea924a34be74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 16:55:38.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86350721f76944b296f3e667570cf146\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a7aee76b126471daaa556f5371b458d : SUCCESS in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13dd2bc068d64ceaad0b5114a9a97e3d : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c5889dae44449c4a09a3c51c8e11dcc : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/384457e7033e4ebbb49fb84bca390a1b : SUCCESS in 5m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50cddc05438e487e8075df72cb41b6ce : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f15186febe4423a86611ef7e5e4373f : SUCCESS in 13m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5865c4f3ddb04d08b30850ed4119099f : SUCCESS in 6m 01s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/f382d5404ff848b885741a8e4d80d5cf : SUCCESS in 6m 22s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/9ea097ce20064e21bbde7fd076f97671 : SUCCESS in 9m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/67e16826aed041f8b048e2102708b86d : SUCCESS in 4m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6cb51bf96657461c89e07d640ef9010e : SUCCESS in 4m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/506d34c8911e47008faf3ec4aa1e30e1 : SUCCESS in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e0201240d3d442f984194815bc89a3e2 : SUCCESS in 1h 53m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d3f54acf45bb4be48f266a90e6a755e1 : SUCCESS in 1h 33m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/01131843e92546438307661dcc2932f2 : SUCCESS in 1h 09m 26s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cb366b5a639b4fd0a597470e649753bc : TIMED_OUT in 2h 32m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5ff4ae165f284444a83f2498bb854a37 : SUCCESS in 1h 10m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f451072847764109ac5b32d0cef04b63 : SUCCESS in 46m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9c2fb2bc4617452884ac0747cec8d2aa : SUCCESS in 43m 22s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5d1a32b1d5ea4bd7b0d52efca3e4b768 : SUCCESS in 54m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79342848125e489a80297c4bcffbadb2 : SUCCESS in 37m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2fb211540bb34b439c4ad93b1593ec7d : SUCCESS in 39m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/368a3168cc8f4eee80fb57eccb849ea3 : SUCCESS in 2h 53m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a65c19fb7d10455bb1ee8ff7ea846bed : SUCCESS in 1h 38m 20s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"3be1974dccb2fd6ea2618f0e0a5c5dacacbfe8c8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-16 16:56:31.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"01d88c18bbecbcdec52e3b0ec115fe992e09f13a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 19:29:09.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4a8592a501b43578b586d1af0a8ca93\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d766b0c55dd54fc7896de3caf6420457 : SUCCESS in 2m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/800fa9b4f0c24ef594a68f6008fbdec2 : SUCCESS in 11m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d991b695ba654bc2984a3ff0f8dfadec : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0e216ade80444b9af684520c8511439 : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ffe794e009d04a86a612d620de5cfbe1 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0331e75b73154ce18097f9f598843609 : SUCCESS in 13m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dd25a014375436abecc3bea2e99e1fe : SUCCESS in 3m 15s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/e21a6a6c0b5d401ab9f5587fa44c2b44 : SUCCESS in 5m 19s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/274c8feb1d444dd8b1cc6d5fe4b38fd8 : SUCCESS in 8m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8a4d1736fa0e4b318971be155b167417 : SUCCESS in 4m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9269e41a01c640fab9602f5bd705c11c : SUCCESS in 4m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d02f70e499544b8c9ce6ae7839525c00 : SUCCESS in 9m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/46c6934aed46488cb0006e4e7870849f : SUCCESS in 35m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/33cc1b20f76140f8b6915c821f369fe0 : SUCCESS in 1h 33m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0d50a01718984743aad310bf07b6603a : SUCCESS in 1h 05m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4c9df828491544ce96c07d146aa85308 : SUCCESS in 1h 38m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b69a7f5afeb24fea9066337361cc6d4f : SUCCESS in 58m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d0fa4bb70c22484a8c51f3c6cf1b6b9e : SUCCESS in 37m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/519b1cb803624a83af9090d055c28f3e : SUCCESS in 43m 02s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8d0c4b4d72c445fb9a9da2efd29ddeb4 : SUCCESS in 43m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11135db31365446ebd72bebef7940e40 : SUCCESS in 32m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/54d939c5f09e47e7940fec36acfa6e1d : SUCCESS in 34m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c07980b2aede446da9ac2649e5afa600 : SUCCESS in 2h 26m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7bf090f34a52420f8231fb680cf01d55 : SUCCESS in 1h 32m 21s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"79b040c50daaaaf6aa8dfeaa62a5923100b66870","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-02-17 10:09:44.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"3d733db666fe56bae523af2788ff30df213cf8a1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-17 13:55:30.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"daff6886e0a7ad69f1778ac1f26be9276a57d113","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 13:56:15.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"c56d8bafdfbbd055e59c42876818e7098cae7e53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 15:38:52.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/632369f2f8ae4a56b6a7a51e055f1a7e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d60fb5c4430749e8a0c4c0e794324c7e : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/025bd40b999a41f2b42ae1ac5a0b2af3 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33e0c23713fc4ab8b09c4ad9deef61ab : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6190183927b4dbf922085f339ab7f7a : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d034998ed594ec5887fcb5a85185b99 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f3598eb6ccc431da5b16a7d3d55b150 : SUCCESS in 3m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c4741df8c85f4dc18e72d7695ecc19a2 : SUCCESS in 3m 34s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3acbc54f96d14766991146770670ab1f : SUCCESS in 7m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ae4ff9a7b74d42f6a53fd640bf830ff9 : SUCCESS in 43m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/32dee21a20ac47c5ad091461c09bcceb : SUCCESS in 1h 37m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ec0c71d9811c4c4c99e346267faad73f : SUCCESS in 1h 05m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/29a98c55b68a4b80954bb6006f87caef : SUCCESS in 42m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/85496fbc8ca3484d883769ada3359ca1 : SUCCESS in 40m 55s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/aeac93a866484b6ab8ec4e090aecbbbc : SUCCESS in 47m 12s","accounts_in_message":[],"_revision_number":13},{"id":"4502fd856228e932e78e7c8fc367f7d1c1ab4979","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 15:38:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"40c83c986e9c75ee18f7450d5ea627a46f8f14e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 15:39:59.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed451a75a0b941ffb00e117d4d7fef93\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5513f2f367d34660aefcee454ecb2d83 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0df87364cb14339a053639ba167ef4e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"dc5cdf83a784e36e159f096866ac0cc99e0bb0e8","revisions":{"3c93c8f36c4857f827d3efea39096e10c036ef82":{"kind":"REWORK","_number":1,"created":"2025-10-10 07:15:06.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/963656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/963656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/963656/1"}}},"commit":{"parents":[{"commit":"40e13e517b3000a0e4b305fdfebe0c12d339f2bd","subject":"Merge \"Remove unused amphora flow and status constants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40e13e517b3000a0e4b305fdfebe0c12d339f2bd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-10 07:14:17.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-10 07:14:54.000000000","tz":120},"subject":"WIP Add support for centos-10-stream","message":"WIP Add support for centos-10-stream\n\nChange-Id: Ie581710636fec38daebe6df560844f14e138e21c\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c93c8f36c4857f827d3efea39096e10c036ef82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c93c8f36c4857f827d3efea39096e10c036ef82"}]},"branch":"refs/heads/master"},"4a355f9af1222525dd635c253c1cfaa4f828f29c":{"kind":"REWORK","_number":2,"created":"2025-10-10 13:42:07.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/963656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/963656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/963656/2"}}},"commit":{"parents":[{"commit":"40e13e517b3000a0e4b305fdfebe0c12d339f2bd","subject":"Merge \"Remove unused amphora flow and status constants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40e13e517b3000a0e4b305fdfebe0c12d339f2bd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-10 07:14:17.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-10 13:41:57.000000000","tz":120},"subject":"WIP Add support for centos-10-stream","message":"WIP Add support for centos-10-stream\n\nChange-Id: Ie581710636fec38daebe6df560844f14e138e21c\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a355f9af1222525dd635c253c1cfaa4f828f29c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a355f9af1222525dd635c253c1cfaa4f828f29c"}]},"branch":"refs/heads/master"},"36abdd9fd790ebc59a0b632332deb06ff6dca61b":{"kind":"REWORK","_number":3,"created":"2025-10-15 14:53:44.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/963656/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/963656/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/963656/3"}}},"commit":{"parents":[{"commit":"54e4c3e8835c5a16e762207340fe2b8c36bbadc7","subject":"Fix devstack cleanup for log offload location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54e4c3e8835c5a16e762207340fe2b8c36bbadc7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-14 13:47:41.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-15 14:53:01.000000000","tz":-240},"subject":"WIP Add support for centos-10-stream","message":"WIP Add support for centos-10-stream\n\nTODO: handle dhcp-client replacement\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/963939\n\nChange-Id: Ie581710636fec38daebe6df560844f14e138e21c\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/36abdd9fd790ebc59a0b632332deb06ff6dca61b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/36abdd9fd790ebc59a0b632332deb06ff6dca61b"}]},"branch":"refs/heads/master"},"d521197f75658ae1f82f50f2f04675d9ebfbc845":{"kind":"REWORK","_number":4,"created":"2025-10-16 08:06:02.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/963656/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/963656/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/963656/4"}}},"commit":{"parents":[{"commit":"54e4c3e8835c5a16e762207340fe2b8c36bbadc7","subject":"Fix devstack cleanup for log offload location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54e4c3e8835c5a16e762207340fe2b8c36bbadc7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-14 13:47:41.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-16 07:58:25.000000000","tz":-240},"subject":"WIP Add support for centos-10-stream","message":"WIP Add support for centos-10-stream\n\nTODO: handle dhcp-client replacement\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/963939\n\nChange-Id: Ie581710636fec38daebe6df560844f14e138e21c\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d521197f75658ae1f82f50f2f04675d9ebfbc845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d521197f75658ae1f82f50f2f04675d9ebfbc845"}]},"branch":"refs/heads/master"},"56ba60729b7c053b23b211c1befce49d34ce6058":{"kind":"REWORK","_number":5,"created":"2025-10-17 07:32:47.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/963656/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/963656/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/963656/5"}}},"commit":{"parents":[{"commit":"54e4c3e8835c5a16e762207340fe2b8c36bbadc7","subject":"Fix devstack cleanup for log offload location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54e4c3e8835c5a16e762207340fe2b8c36bbadc7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-14 13:47:41.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-17 07:31:50.000000000","tz":-240},"subject":"Add support for centos-10-stream","message":"Add support for centos-10-stream\n\nNote: c10s doesn\u0027t include a dhcp-client, NetworkManager handles dhcp\nconnections but it doesn\u0027t work in the non-default network namespace.\n\nHowever, the octavia worker always sends the assigned fixed_ips to the\namphora, so it means that amphora-agent always configures static IP\naddresses in the amphora-haproxy namespace.\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/963939\n\nChange-Id: Ie581710636fec38daebe6df560844f14e138e21c\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56ba60729b7c053b23b211c1befce49d34ce6058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56ba60729b7c053b23b211c1befce49d34ce6058"}]},"branch":"refs/heads/master"},"78d7d98ddd44d11cc176dff81874e7f465492da8":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-11-12 08:46:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/963656/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/963656/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/963656/6"}}},"commit":{"parents":[{"commit":"54e4c3e8835c5a16e762207340fe2b8c36bbadc7","subject":"Fix devstack cleanup for log offload location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54e4c3e8835c5a16e762207340fe2b8c36bbadc7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-14 13:47:41.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-11-12 08:46:04.000000000","tz":0},"subject":"Add support for centos-10-stream","message":"Add support for centos-10-stream\n\nNote: c10s doesn\u0027t include a dhcp-client, NetworkManager handles dhcp\nconnections but it doesn\u0027t work in the non-default network namespace.\n\nHowever, the octavia worker always sends the assigned fixed_ips to the\namphora, so it means that amphora-agent always configures static IP\naddresses in the amphora-haproxy namespace.\n\nChange-Id: Ie581710636fec38daebe6df560844f14e138e21c\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d7d98ddd44d11cc176dff81874e7f465492da8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d7d98ddd44d11cc176dff81874e7f465492da8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3b4ece41a4d0e70d8645beeb0c8fef061ce09de6":{"kind":"REWORK","_number":7,"created":"2025-11-12 08:50:45.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/963656/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/963656/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/963656/7"}}},"commit":{"parents":[{"commit":"be37addac0a73afe7422e43e4dd0f2e283805567","subject":"Merge \"reno: Update master for unmaintained/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be37addac0a73afe7422e43e4dd0f2e283805567"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-14 13:47:41.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-11-12 08:50:31.000000000","tz":60},"subject":"Add support for centos-10-stream","message":"Add support for centos-10-stream\n\nNote: c10s doesn\u0027t include a dhcp-client, NetworkManager handles dhcp\nconnections but it doesn\u0027t work in the non-default network namespace.\n\nHowever, the octavia worker always sends the assigned fixed_ips to the\namphora, so it means that amphora-agent always configures static IP\naddresses in the amphora-haproxy namespace.\n\nChange-Id: Ie581710636fec38daebe6df560844f14e138e21c\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b4ece41a4d0e70d8645beeb0c8fef061ce09de6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b4ece41a4d0e70d8645beeb0c8fef061ce09de6"}]},"branch":"refs/heads/master"},"f192625c6af89d966065e50b0ea5f942ef7d5af7":{"kind":"REWORK","_number":8,"created":"2026-01-06 12:43:27.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/963656/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/963656/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/963656/8"}}},"commit":{"parents":[{"commit":"f42860a7302b521dd2895a22535dbd1de1f912f0","subject":"Remove url tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f42860a7302b521dd2895a22535dbd1de1f912f0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-14 13:47:41.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-06 12:42:43.000000000","tz":60},"subject":"Add support for centos-10-stream","message":"Add support for centos-10-stream\n\nNote: c10s doesn\u0027t include a dhcp-client, NetworkManager handles dhcp\nconnections but it doesn\u0027t work in the non-default network namespace.\n\nHowever, the octavia worker always sends the assigned fixed_ips to the\namphora, so it means that amphora-agent always configures static IP\naddresses in the amphora-haproxy namespace.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/972260\n\nChange-Id: Ie581710636fec38daebe6df560844f14e138e21c\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f192625c6af89d966065e50b0ea5f942ef7d5af7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f192625c6af89d966065e50b0ea5f942ef7d5af7"}]},"branch":"refs/heads/master"},"297b61dd6d0f7a84f0b4ee1c2e17bd765eb1d4d9":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2026-01-08 13:46:42.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/963656/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/963656/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/963656/9"}}},"commit":{"parents":[{"commit":"f42860a7302b521dd2895a22535dbd1de1f912f0","subject":"Remove url tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f42860a7302b521dd2895a22535dbd1de1f912f0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-14 13:47:41.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-08 13:46:42.000000000","tz":0},"subject":"Add support for centos-10-stream","message":"Add support for centos-10-stream\n\nNote: c10s doesn\u0027t include a dhcp-client, NetworkManager handles dhcp\nconnections but it doesn\u0027t work in the non-default network namespace.\n\nHowever, the octavia worker always sends the assigned fixed_ips to the\namphora, so it means that amphora-agent always configures static IP\naddresses in the amphora-haproxy namespace.\n\nChange-Id: Ie581710636fec38daebe6df560844f14e138e21c\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/297b61dd6d0f7a84f0b4ee1c2e17bd765eb1d4d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/297b61dd6d0f7a84f0b4ee1c2e17bd765eb1d4d9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"95a53b59f1e19bf82be91294cb95231c31a4245b":{"kind":"REWORK","_number":10,"created":"2026-01-08 14:05:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/963656/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/963656/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/963656/10"}}},"commit":{"parents":[{"commit":"f42860a7302b521dd2895a22535dbd1de1f912f0","subject":"Remove url tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f42860a7302b521dd2895a22535dbd1de1f912f0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-14 13:47:41.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-08 14:05:34.000000000","tz":60},"subject":"Add support for centos-10-stream \u0026 rocky 10","message":"Add support for centos-10-stream \u0026 rocky 10\n\nNote: c10s doesn\u0027t include a dhcp-client, NetworkManager handles dhcp\nconnections but it doesn\u0027t work in the non-default network namespace.\n\nHowever, the octavia worker always sends the assigned fixed_ips to the\namphora, so it means that amphora-agent always configures static IP\naddresses in the amphora-haproxy namespace.\n\nChange-Id: Ie581710636fec38daebe6df560844f14e138e21c\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95a53b59f1e19bf82be91294cb95231c31a4245b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95a53b59f1e19bf82be91294cb95231c31a4245b"}]},"branch":"refs/heads/master"},"eca8783959d55eb278bb8ddfa2b5c1392311d431":{"kind":"REWORK","_number":11,"created":"2026-01-08 14:10:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/963656/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/963656/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/963656/11"}}},"commit":{"parents":[{"commit":"f42860a7302b521dd2895a22535dbd1de1f912f0","subject":"Remove url tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f42860a7302b521dd2895a22535dbd1de1f912f0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-14 13:47:41.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-08 14:08:01.000000000","tz":60},"subject":"Add support for centos-10-stream \u0026 rocky 10","message":"Add support for centos-10-stream \u0026 rocky 10\n\nNote: c10s doesn\u0027t include a dhcp-client, NetworkManager handles dhcp\nconnections but it doesn\u0027t work in the non-default network namespace.\n\nHowever, the octavia worker always sends the assigned fixed_ips to the\namphora, so it means that amphora-agent always configures static IP\naddresses in the amphora-haproxy namespace.\n\nChange-Id: Ie581710636fec38daebe6df560844f14e138e21c\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eca8783959d55eb278bb8ddfa2b5c1392311d431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eca8783959d55eb278bb8ddfa2b5c1392311d431"}]},"branch":"refs/heads/master"},"070d6b9494836481053b3026175ee85580595b8a":{"kind":"REWORK","_number":12,"created":"2026-01-08 14:30:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/963656/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/963656/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/963656/12"}}},"commit":{"parents":[{"commit":"f42860a7302b521dd2895a22535dbd1de1f912f0","subject":"Remove url tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f42860a7302b521dd2895a22535dbd1de1f912f0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-14 13:47:41.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-08 14:24:36.000000000","tz":60},"subject":"Add support for centos-10-stream \u0026 rocky 10","message":"Add support for centos-10-stream \u0026 rocky 10\n\nNote: c10s doesn\u0027t include a dhcp-client, NetworkManager handles dhcp\nconnections but it doesn\u0027t work in the non-default network namespace.\n\nHowever, the octavia worker always sends the assigned fixed_ips to the\namphora, so it means that amphora-agent always configures static IP\naddresses in the amphora-haproxy namespace.\n\nChange-Id: Ie581710636fec38daebe6df560844f14e138e21c\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/070d6b9494836481053b3026175ee85580595b8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/070d6b9494836481053b3026175ee85580595b8a"}]},"branch":"refs/heads/master"},"dc5cdf83a784e36e159f096866ac0cc99e0bb0e8":{"kind":"REWORK","_number":13,"created":"2026-02-16 13:58:01.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/963656/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/963656/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/963656/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/963656/13"}}},"commit":{"parents":[{"commit":"f42860a7302b521dd2895a22535dbd1de1f912f0","subject":"Remove url tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f42860a7302b521dd2895a22535dbd1de1f912f0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-14 13:47:41.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-02-16 13:55:00.000000000","tz":60},"subject":"Add support for centos-10-stream \u0026 rocky 10","message":"Add support for centos-10-stream \u0026 rocky 10\n\nNote: c10s doesn\u0027t include a dhcp-client, NetworkManager handles dhcp\nconnections but it doesn\u0027t work in the non-default network namespace.\n\nHowever, the octavia worker always sends the assigned fixed_ips to the\namphora, so it means that amphora-agent always configures static IP\naddresses in the amphora-haproxy namespace.\n\nChange-Id: Ie581710636fec38daebe6df560844f14e138e21c\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dc5cdf83a784e36e159f096866ac0cc99e0bb0e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dc5cdf83a784e36e159f096866ac0cc99e0bb0e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
