)]}'
{"id":"openstack%2Foctavia~964237","triplet_id":"openstack%2Foctavia~stable%2F2025.2~Ieb023b5cdcdbbcea5e11c6fb936bda762615e3ec","project":"openstack/octavia","branch":"stable/2025.2","hashtags":[],"change_id":"Ieb023b5cdcdbbcea5e11c6fb936bda762615e3ec","subject":"Skip 4.0.{0,1} pylint releases","status":"MERGED","created":"2025-10-16 16:54:10.000000000","updated":"2025-10-16 22:03:06.000000000","submitted":"2025-10-16 22:02:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964237","cherry_pick_of_change":964174,"cherry_pick_of_patch_set":1,"meta_rev_id":"d323286ce6e429dfe94ecdcab9aeabada1486f04","_number":964237,"virtual_id_number":964237,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-16 22:02:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-10-16 16:59:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2025-10-16 16:59:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-16 16:54:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2025-10-16 19:58:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c3b698946df5863b84e72e20a9294afde7cde595","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-16 16:54:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"418c7e9c7e952efac988d593f99ad5e8ea02dd90","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-16 16:59:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6fd58893ad352b88d00c023bf7eb057482f2f9ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 19:58:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fe5d52c03b54b68afbb755551a95721\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/82218e30e34e46d6895b698bbd939df1 : SUCCESS in 2m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8cba21ed5144b7fb9b0ad496997c25a : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1e5305ba22b4279a21c6f30ec21b9df : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68a737a79d104593be4f6aae26d113a7 : SUCCESS in 5m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2fdbacd9046c473db146319103ac638c : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b6260a9116a4b8f8e38d920fda3eabf : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf85c8825160488684af374d5cbb510b : SUCCESS in 14m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c6872c243644177b5716a8f53cf4cdd : SUCCESS in 6m 02s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bc98658306be4967813d3ef95a4004e8 : SUCCESS in 5m 36s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/4649fc3277c645bcb7caf6ca0d9633ed : SUCCESS in 7m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/30a584f0a3b743dd9399a1386ad51544 : SUCCESS in 2m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eed990da8c2b4930a95a5f93545437a0 : SUCCESS in 6m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1a923321673d488aa8e996c4637b9b70 : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c3cd9a02c7834d02b967e0bc6dbd44bd : SUCCESS in 1h 23m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8be26dc36c3c40128727972601abe1dc : SUCCESS in 1h 53m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/61bcb4b81d5148c492634e304e834866 : SUCCESS in 1h 00m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e08661d41e904555be6c1acec52693ac : SUCCESS in 2h 02m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/92d63d7b54ef4149a16f360833e19977 : SUCCESS in 1h 16m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/96c7aaff383e4e3dae63ce09df32ebc1 : SUCCESS in 37m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8fc1e2f0a08a422bab4c3afee5f302a1 : SUCCESS in 39m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a0536aee098847409636f06bc231d74f : TIMED_OUT in 2h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cd5619c507664effa1f15b1f540d3ba7 : TIMED_OUT in 3h 03m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/79b5c34b86334a9eb6e51195d28aad83 : SUCCESS in 2h 32m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9581338d892cd6bf907900b2a93ddaf64a23ba83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 19:59:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1492a8e7bc1c857dbc73be1e555146a07b1c9b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 22:02:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01ae002eb2f440d0a58fd949a9167539\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/06aaee90d8d34ea7a64d0e7e83789f1f : SUCCESS in 1m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffce37e91e0b41ce9654a0f615aab3da : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00c19ad111c64f3db9f80179973789a7 : SUCCESS in 6m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6c3098368c042258f442e6a7b1d9777 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0d9671f4c2840ae98dac8e2bfc4e08c : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06462d5f24d14faaae52d5c71285c564 : SUCCESS in 5m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e14a22146be04bae8ccfa8d2f806d052 : SUCCESS in 3m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/50fa9c564421424b8f817bc776946230 : SUCCESS in 8m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/825d36500401406aa3a2b911e8a59811 : SUCCESS in 1h 33m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/58ef9ef5e5e2491890f4cf36b69c1136 : SUCCESS in 2h 01m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4b4bef621b754b1a881c2b3bd04672b4 : SUCCESS in 1h 09m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0acbd7dccb694e29a3a1704d5246a3bf : SUCCESS in 43m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4ffe5387399e4d85b6b45382bb5ad456 : SUCCESS in 42m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ecc15e0abead8a172ee9e1e927cdeda3ad5b4258","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 22:02:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d323286ce6e429dfe94ecdcab9aeabada1486f04","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 22:03:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54eb8ccb6df9458ab4a72c7dea74bd20\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ed87ee1f4674d6692fe2f59be20d7f2 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f702f1d829d4910b81788043fbd5fd0 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"427928651408a10402f7de3a70e5e27bc3dfad79","revisions":{"427928651408a10402f7de3a70e5e27bc3dfad79":{"kind":"REWORK","_number":1,"created":"2025-10-16 16:54:10.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/37/964237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/964237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/964237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/964237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/964237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/964237/1"}}},"commit":{"parents":[{"commit":"fa44598bd8e410222a2c47b08196a2af3012fbfa","subject":"Fix Amphora controller IP list update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa44598bd8e410222a2c47b08196a2af3012fbfa"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-15 14:46:38.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-10-16 16:54:10.000000000","tz":0},"subject":"Skip 4.0.{0,1} pylint releases","message":"Skip 4.0.{0,1} pylint releases\n\n4.0.0 and 4.0.1 are affected by [0]\n\n[0] https://github.com/pylint-dev/pylint/issues/10669\n\nChange-Id: Ieb023b5cdcdbbcea5e11c6fb936bda762615e3ec\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit 6743aa7bf012612ec7d281618bafc8b208e52e0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/427928651408a10402f7de3a70e5e27bc3dfad79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/427928651408a10402f7de3a70e5e27bc3dfad79"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
