)]}'
{"id":"openstack%2Foctavia~964280","triplet_id":"openstack%2Foctavia~master~I2c1fea1febf6a984a64a02887981d6f6247fc570","project":"openstack/octavia","branch":"master","topic":"c10s","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-02-18 07:14:43.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I2c1fea1febf6a984a64a02887981d6f6247fc570","subject":"Add c10s jobs","status":"MERGED","created":"2025-10-17 07:39:51.000000000","updated":"2026-02-18 08:55:34.000000000","submitted":"2026-02-18 08:53:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964280-c10s","meta_rev_id":"826627017c8d4208a5969e5177fa50e9f0e60540","_number":964280,"virtual_id_number":964280,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-18 08:53:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-02-16 13:58:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-18 07:14:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-18 07:14:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-17 10:46:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-13 09:45:48.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"9bec89acc4334a8c3abd919fe4f862468633a88b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-10-17 07:39:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff5299a19482e44754ad4f4668b6fd8aa5e469e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 10:46:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c94f4e9b0ea4418909f10a729d7f983\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f49cf256dafb45c191c47bf9087ef97a : SUCCESS in 12m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e4a74f8dd8a4c91839fb59b47715029 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ff32031c45f46c986290c37fa25eee1 : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a6e0965e8c974e82b0ed13685f7fdded : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3490c0688064c9dac8c85733f8bdeb9 : SUCCESS in 6m 18s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b43c6a9c9d084b0b999bc7aa3120ec36 : SUCCESS in 5m 06s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/8aa17c9112b14e7c9fff788f4e01dbf2 : SUCCESS in 5m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2b22bdfb22964e64b98f313d869db720 : SUCCESS in 4m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/efc9fb09b6de4007bd6299280c830a8d : SUCCESS in 8m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/47925b464ff749f7bb066fe5dc73c358 : SUCCESS in 8m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c349f72625cf44e69a8b2103a8b5b2c4 : SUCCESS in 1h 23m 21s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/72458b12b8e04256b4eeecd6ca5c3633 : SUCCESS in 2h 04m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/89a2e6cb31e947a0a3338c142d24eed1 : SUCCESS in 1h 12m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a68dbd388d0140f798971d051da48745 : SUCCESS in 2h 03m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8b7d3b0016ca442bb9247554df290d40 : SUCCESS in 1h 11m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b28cf2eee668478785661743a78b7c47 : SUCCESS in 38m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/907856df57614eba89620861b421bb9a : SUCCESS in 50m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b23c1cf8c6ee43e1882e26e1ebd4086e : SUCCESS in 36m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fd109f58946441feb4ea29def838fce7 : TIMED_OUT in 2h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/536b642156a74dbb9be3b35e5963e5bf : TIMED_OUT in 3h 03m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9c07aa225db64004832777903f6c25fa : SUCCESS in 2h 28m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/311c0368cafc4915a622f034ab89b2f0 : SUCCESS in 1h 55m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/80618f895beb4c7490177b85e4aba83c : SUCCESS in 1h 09m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4add743df16c1a9e4fd3ddddc681a151fb42cd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-11-12 08:46:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ce91bbbe7ece2b8c318b44a1a00cce17238c5dfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-11-12 08:50:45.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"3847a52442c803d7d121939b6cd4ae758efedd9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 08:50:52.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/963653 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"8a5e9968f5b5f241ede455372cce680e42e91cff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 11:48:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89ab597d55ee40519790bbf7873a18be\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c0e6502b08a48cf803322941dd59c7d : SUCCESS in 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02eb313b274e4a5bb85234b6668363cd : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b984bbe8e6949689eb7228472e38cec : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/576d01708f8446d5a1e23f248e1445b0 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0929f37d354a7b9b6d87c26f8e73e6 : SUCCESS in 12m 51s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1f6db8b1fd2a49f7adc6bc9d06d1a3a6 : SUCCESS in 7m 37s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/c64796f5c1fd436abb0e650b64c7e4ba : SUCCESS in 8m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a56448d2c9f14d3286e0152efb364ba1 : SUCCESS in 3m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e8f236e197004bf9800dd4444402141c : SUCCESS in 6m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5d393697637c4984960f595e0c27bb2d : SUCCESS in 8m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bf7e91562b09414580c42857ca605fdc : SUCCESS in 1h 48m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/832db0f959714c2bb89dd85561b2db8f : SUCCESS in 1h 51m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/883e15d96fa64ce880145e3b848e0c73 : SUCCESS in 1h 12m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6397ee68dbe54609aa4fcdbe857f69f1 : SUCCESS in 1h 59m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f21f6ee7706c4321b846992f62be654d : SUCCESS in 1h 03m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/18d5ad7781ae4a12ba7096a2dae99965 : SUCCESS in 39m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d7e1a20de2e843238469dfb0ee686aea : FAILURE in 23m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53f6a402729a493cab01010b0016751a : SUCCESS in 36m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/56512f8d62ee4d9a8303dc144ba70e8d : TIMED_OUT in 2h 34m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/06a2e7ace6254ed7b20d5deeb3034aee : SUCCESS in 2h 55m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3cec7532feca4d838b31c81b9b32cf43 : TIMED_OUT in 2h 33m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/522c482076cc4c97a63ea70370337640 : SUCCESS in 1h 59m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1f7866bc6ad842baaa11bb02ad7167f0 : SUCCESS in 1h 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3d6e0bed5ff3f6448af704258cfefc1b1f194fc0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-11-12 12:41:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ba7a8ff40d5df1af7c2a3d2bfc83c70fc2f0afe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 15:58:04.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aaf769e590484d2e96bffd01c5d8e356\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70aa3347019c41ed8a8e2b5632ec9072 : SUCCESS in 12m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24b521f4291e4051b20d420c3b868e98 : SUCCESS in 6m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/610441c0014841f7a2e9691fcc111d7f : SUCCESS in 6m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27379decf24c45e3b07f72bcc43c4da3 : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/332b0cfd5dcf47fb91af473a090db1f1 : SUCCESS in 13m 03s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d13aed411f364dc6bf79249c052271e1 : SUCCESS in 4m 52s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/fd4d5380b5064726abc495ec03880288 : SUCCESS in 5m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8d0b2731dac442f5bf5c011508b4f52c : SUCCESS in 4m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bbe911e7e9da425b8ac22ffeb2a10d5e : SUCCESS in 8m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1f6b77ff7a31494895b553ee8878aaa0 : SUCCESS in 8m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/20e5f72e76364e428784a535b2f2cc0c : POST_FAILURE in 1h 40m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3f2570cfcbdf46d0837ed1b385fc49aa : SUCCESS in 2h 03m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/01273c7954fb4d2fb7b318ad6c7f6c33 : SUCCESS in 1h 05m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c3e8dac527264fd2ac8b6db1d5f0bd82 : POST_FAILURE in 2h 07m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f43ec3bdf2a64d95a2f8ac48b5a55edd : SUCCESS in 1h 22m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ada1560ca4d84c2fab978abffed06c14 : SUCCESS in 45m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6dece8ea1b174c06a0d4478734647db8 : SUCCESS in 45m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d7175cc91c14e8c921a40308b13d94f : SUCCESS in 42m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/eb082e76488a40079dabd631877db105 : TIMED_OUT in 2h 33m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e8c204d11e6842ddaf755bbbc59932e6 : TIMED_OUT in 3h 03m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a23ed922b1244ac2ad233c48ac4777e5 : SUCCESS in 2h 30m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/64655dc4e1b84cd19f2de63f961b46ce : POST_FAILURE in 1h 57m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4642cb9441804ed6ad37bcf2b1356054 : SUCCESS in 1h 22m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a114e2c61f7b87985805e5328719827585081b7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-06 12:43:27.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"248e7434ce9553aa104b04ace6639a606f6b5017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 15:17:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9afc14efcca44b1e84f8b8e83c55e027\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44f1614e3855490590ca576fcbb86f2a : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81556e933dd548af8e26db1431ef8bcc : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60456d54143c4390b5fa14dc19f207b5 : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/06fb5a5067e84974b674095c7ceb11fc : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/202cf64dccdb4e3e876af6954aec9687 : SUCCESS in 12m 32s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3acf43e7e4c746188e3b11f72e4d4ac6 : SUCCESS in 2m 52s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/322c1579090a46a2a4b367640838b573 : SUCCESS in 7m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/96b92557c58c42b9922c15636af47dfb : SUCCESS in 4m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a1e0a3a91e724a0b972b06b54ddeefe8 : SUCCESS in 6m 05s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e0a27712b4d448f385a0e5649990584b : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/347ac066d16e47ccb0f5b16f0c4584e7 : SUCCESS in 46m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/afeb2be61fcb4cad99421bade91c6eb2 : SUCCESS in 1h 33m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f35b067537554e21976ec56e08a0a096 : SUCCESS in 1h 03m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7c52679a39d548ad9071a7a19ddcd221 : SUCCESS in 1h 43m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2cc4549dd7b84206a55993ccff07dbbf : SUCCESS in 1h 09m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5452f01be157481c96d960860ae0e1c5 : SUCCESS in 35m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9166444dede645789d1a607d28ae5efe : SUCCESS in 42m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/93b9d0db2cc545a08c02eb0064e5c517 : SUCCESS in 33m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d7fc7f7d0b464af880b977bd26b320bf : SUCCESS in 34m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/81c3bea7c62d43a1b3fb312324a5406b : SUCCESS in 2h 27m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4b30f118a85a4e3781ea06e11e96c990 : SUCCESS in 1h 31m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/0d5a01877fb543558afe2b1c80ffd2bb : SUCCESS in 1h 31m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ea25d681e67340938a6e1dfc0fc21bff : SUCCESS in 1h 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d32dd96fca0dd2a2033e657c8cc03792e1189fa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-08 13:47:00.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8d084d2f37fd97a1c52b4287bca764ae24afddc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 14:05:49.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/963653 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"897f5e73f002f104620b83cbf3149a92048aa4ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-08 14:10:10.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"42a854088b5cd7e3c12327bca821c425d4e4e24a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 14:12:26.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":6},{"id":"64ca91af8c91a92023eb6b3f278890ed22b8a704","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-08 14:30:04.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"11468a6d3264509b7b00b186392dbfc65cae8c65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 14:31:31.000000000","message":"Patch Set 7: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/octavia on branch master.  The\nproblem was:\n\n  The nodeset \"centos-10-stream\" was not found.\n\nThe problem appears in the the \"publish-openstack-octavia-amphora-image-centos-10-stream\" job stanza:\n\n  job:\n      name: publish-openstack-octavia-amphora-image-centos-10-stream\n      nodeset: centos-10-stream\n      parent: publish-openstack-octavia-amphora-image\n      description: |\n        Publish CentOS 10 Stream based amphora image to tarballs.o.o.\n      vars:\n          amphora_os: centos\n          amphora_os_release: 10-stream\n  ...\n\n  in \"openstack/octavia/zuul.d/jobs.yaml@master\", line 21","accounts_in_message":[],"_revision_number":7},{"id":"48a47363670c8bd1215779420bb02150347f01ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-08 14:37:57.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"fa8a6ed12667ac251e0487a362ef05e7740d8d94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 14:40:06.000000000","message":"Patch Set 8: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/octavia on branch master.  The\nproblem was:\n\n  The nodeset \"centos-10-stream-4GB\" was not found.\n\nThe problem appears in the the \"publish-openstack-octavia-amphora-image-centos-10-stream\" job stanza:\n\n  job:\n      name: publish-openstack-octavia-amphora-image-centos-10-stream\n      nodeset: centos-10-stream-4GB\n      parent: publish-openstack-octavia-amphora-image\n      description: |\n        Publish CentOS 10 Stream based amphora image to tarballs.o.o.\n      vars:\n          amphora_os: centos\n          amphora_os_release: 10-stream\n  ...\n\n  in \"openstack/octavia/zuul.d/jobs.yaml@master\", line 21","accounts_in_message":[],"_revision_number":8},{"id":"403441744693cc2af97a6af2e4fe18d22d071656","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-08 17:34:45.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"db71c4f8fa49b9883da209516e12a461eef4aabf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 17:36:30.000000000","message":"Patch Set 9: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/octavia on branch master.  The\nproblem was:\n\n  The nodeset \"centos-10-stream-8GB\" was not found.\n\nThe problem appears in the the \"publish-openstack-octavia-amphora-image-centos-10-stream\" job stanza:\n\n  job:\n      name: publish-openstack-octavia-amphora-image-centos-10-stream\n      nodeset: centos-10-stream-8GB\n      parent: publish-openstack-octavia-amphora-image\n      description: |\n        Publish CentOS 10 Stream based amphora image to tarballs.o.o.\n      vars:\n          amphora_os: centos\n          amphora_os_release: 10-stream\n  ...\n\n  in \"openstack/octavia/zuul.d/jobs.yaml@master\", line 21","accounts_in_message":[],"_revision_number":9},{"id":"e2be450a5dbbd25e788bf473d8f09c591c505aee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-09 07:40:22.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"57cec6841d09559934a217fc46b163178c2fb19a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 08:34:38.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7750641bb84438baf417cd3edb1faa4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1886f5df4af24a9993735c4d224c666e : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbe3779d5c0649179abcad5c7f906eb5 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97dd27ead9b14bac99567215841a6425 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/566f1db18e8341858fe215e136067770 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/818f006c0bb446389461f59aec254b30 : SUCCESS in 12m 34s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/fe5066741c914a7982b771c92b1c81a1 : SUCCESS in 6m 54s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/8178df4a53ba40dc80b1b848c2db3b89 : SUCCESS in 8m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/40ab243aea884526b3b55dd7eb5313d0 : SUCCESS in 3m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/09f8382d74a149399e82a8a0f31ba014 : SUCCESS in 7m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5a100cdf63274183933ec2c8f2e8fd62 : SUCCESS in 6m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f93aa81ce3904f6fbd64f98511a62d8a : RETRY_LIMIT in 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c7cbf3be2fe44c079a1b0a7d604595e6 : RETRY_LIMIT in 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0c639656cf7e4ef6bacfc6cff4097d5c : RETRY_LIMIT in 48s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fa1ecbe9591b4df984e1cdd45853dbd8 : RETRY_LIMIT in 1m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/699420213b7046449853f1bbbca16ac3 : RETRY_LIMIT in 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dd251ecf6b6d42c58b0a4bb281785d56 : RETRY_LIMIT in 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/85b5d22e6b7440a79d4f557fbddcc202 : SUCCESS in 47m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ec464c49fe4422e8f2802136a279f5c : RETRY_LIMIT in 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7df8bb31a6d449a7a069916d09a73cdc : RETRY_LIMIT in 1m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d01a2ec7fc5480598287cfdc89befa3 : RETRY_LIMIT in 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/59268c50d5fb412ebffe0a4055769b9c : RETRY_LIMIT in 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/b46f8eb8793c464b880a865a348f195c : RETRY_LIMIT in 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e5a698a5f7784e23b04f0edd273df0b3 : RETRY_LIMIT in 48s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9eb95a9e7b6854bd167fc97a5598b3d5d23b6fce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-09 14:32:37.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"dc32eb1648eb38ff51c79c4b1c374ba6434a9613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 17:54:37.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecee5b3d2a954b24ae42f90bd437fae8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50434bd3e2af4719a2ed4c3299ada479 : SUCCESS in 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b055a1c17584ffebdbf7ee6e781c0b8 : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcf925c39e434ae6bc76685535f6bcc3 : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/acdddc919cc440d78c0c95257c111c71 : SUCCESS in 9m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d078503de3bc40a4b8de22218fd715de : SUCCESS in 13m 16s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/70595b98d0934a3ea345ed52322ac9c0 : SUCCESS in 6m 22s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/06009a6369fa492c8174f8d3e86b5710 : SUCCESS in 10m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e4ccbd1bcbe640f590631392a2fec245 : SUCCESS in 5m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/af13efd8e7e14eb2a0c42ad2ab3d5bed : SUCCESS in 8m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0b12c8d20a974582a6c0047e45cc55dc : SUCCESS in 10m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5c300e0f3bef454db6f48bcc173285d3 : SUCCESS in 1h 42m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d10a702851f04af2a68ef8ed43b681eb : SUCCESS in 1h 34m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fd9f64655321477dafc3ec500055e824 : SUCCESS in 1h 08m 04s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6db9f77f90914af9a374458deb5d9af0 : SUCCESS in 1h 38m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/01d49ede98914b638e462b1143eaa501 : SUCCESS in 1h 11m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fe5dcd6849a541aaa44f166fccc2ced7 : SUCCESS in 36m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d2f2c408020a4342b680b6755ee6f61c : SUCCESS in 43m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af85faed20bb49908c2ddd27c716b3fd : SUCCESS in 38m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/34de7bbc146149e1ae16438af0ca54dc : SUCCESS in 41m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/48860aa204c04f0cb24c1b4084a78949 : SUCCESS in 2h 36m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4225bc07450f49f7a0be89ae91dc6815 : SUCCESS in 1h 38m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/a9a25b7401d0465da78bb854f2c3570c : SUCCESS in 1h 25m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9c62b4456e624e0c93686f34c530b531 : SUCCESS in 1h 03m 18s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"43a97a9ea3ea13190e9591bb5d4fc7c3dea77dd5","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-01-13 09:45:48.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a91f041d0fcc49d28ea9f696d33db3e02bb4c168","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-16 13:58:25.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"960c59c38718786f3256de69d2d89d226bba71e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 16:39:41.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3cf8722961b463685f4dccf8de5a140\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/191921419cf146f9ae255da9f3888aa6 : SUCCESS in 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e11bea8e59a64a97be9ff047cc10a68f : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5d9aba2e4d5497c91547bf1273d4eb1 : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c2ff9093bba48dbafba833bc01aa021 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da5a251902334708abea7cebcf9c65d2 : SUCCESS in 14m 13s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/0610f9aa4db848488e7012cf30a94e5f : SUCCESS in 4m 50s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/cac9e511dff9480cbb6f1ef96fed3d92 : SUCCESS in 8m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1aa901ecc1f148d28430c46130e3db3a : SUCCESS in 5m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c16f461d51f44c778e66b86fe43f207e : SUCCESS in 8m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b54f6bc9ba0f473e95a31ad4d628be07 : SUCCESS in 10m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2fbe4f3236724bf1abb449a8bef87dae : SUCCESS in 1h 28m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ef0f4fce62e4453aa517a60e3716a3b3 : SUCCESS in 1h 42m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5d4036428af94d20b057ec7a615cad7e : SUCCESS in 1h 01m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/af6beff162244ef7a27c5468d1416065 : SUCCESS in 1h 36m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ee701914b2b246a59b4312825be1ea4e : SUCCESS in 1h 07m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b40199b8ad974e18a4e9f901b8f46e22 : SUCCESS in 34m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c82a59b9e8394455b615a902d04c338f : SUCCESS in 45m 57s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2f9bd1760e5d41d08786eea9570f5a20 : SUCCESS in 48m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b310ebc176a48b0b291b06cbce9cc06 : SUCCESS in 37m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/87ed7789932f456cba6887bba52eba86 : SUCCESS in 34m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a326e60dc6fb4f1da691922465c8495a : SUCCESS in 2h 35m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d4fe0532458849008902ed603d218c8f : SUCCESS in 1h 39m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/ab9f0e145a544afcb01a3787606027c0 : SUCCESS in 1h 27m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e8c87bd12c3341a58ac8da7466451518 : SUCCESS in 1h 08m 51s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ecfe2fc747973a79b0d6f21f21d5ee0eb3dfce34","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-18 07:14:43.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"eefa672ba586d96e595a0435a1f42f5451f57c5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 07:15:25.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"7c2e808e13b6c3a657339b1732dfcf4c75f73e16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 08:53:17.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc6ccd98c06f4008ad213dd1f0a9ae56\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b4322c6300349288371fe4ad2f75246 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04a79f9372c54e17ab9b5f76a7dc3b89 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f0674d983bb463c8d570414415749e5 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56e4f467d174473581c0710bd3c1f290 : SUCCESS in 12m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c8cdb60da37c4afe870717defc459bfe : SUCCESS in 3m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5280545c020d4747aa97323391c75889 : SUCCESS in 3m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7c36828cac444df8a494f72888a5b73f : SUCCESS in 1h 13m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9e1a0e3bb62345c9bd5f42d2169091a1 : SUCCESS in 1h 37m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d9056886a50a46f1b2a4cafe37c49ad9 : SUCCESS in 1h 03m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1e83da69975741b8a606641cb830738c : SUCCESS in 37m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7b5aa6cca4ed49ed8950e8465afbb911 : SUCCESS in 45m 22s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f8ff8c23d01342579bf6251ebc2fe253 : SUCCESS in 46m 36s","accounts_in_message":[],"_revision_number":12},{"id":"bd2e131e109b8cc7023b01df5427bcbf7367bf0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 08:53:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"826627017c8d4208a5969e5177fa50e9f0e60540","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 08:55:34.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7217a747aa734dc69bbad0584d8cbb6b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31473b0b25af4bbeac4dce527637a6d5 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"b59b03c7b65da067f5315b0608ef577d91d53e8c","revisions":{"88a98afb4d0605531cc9ff237e34c3decc90cb48":{"kind":"REWORK","_number":1,"created":"2025-10-17 07:39:51.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/964280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/964280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/964280/1"}}},"commit":{"parents":[{"commit":"56ba60729b7c053b23b211c1befce49d34ce6058","subject":"Add support for centos-10-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56ba60729b7c053b23b211c1befce49d34ce6058"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-17 07:39:25.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-17 07:39:28.000000000","tz":120},"subject":"Add c10s jobs to the check pipeline (non-voting)","message":"Add c10s jobs to the check pipeline (non-voting)\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/963653\n\nChange-Id: I2c1fea1febf6a984a64a02887981d6f6247fc570\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88a98afb4d0605531cc9ff237e34c3decc90cb48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88a98afb4d0605531cc9ff237e34c3decc90cb48"}]},"branch":"refs/heads/master"},"0ea571def941364a9da51b2b18247f13c98e14f8":{"kind":"NO_CHANGE","_number":2,"created":"2025-11-12 08:46:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/964280/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/964280/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/964280/2"}}},"commit":{"parents":[{"commit":"78d7d98ddd44d11cc176dff81874e7f465492da8","subject":"Add support for centos-10-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d7d98ddd44d11cc176dff81874e7f465492da8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-17 07:39:25.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-11-12 08:46:20.000000000","tz":0},"subject":"Add c10s jobs to the check pipeline (non-voting)","message":"Add c10s jobs to the check pipeline (non-voting)\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/963653\n\nChange-Id: I2c1fea1febf6a984a64a02887981d6f6247fc570\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ea571def941364a9da51b2b18247f13c98e14f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ea571def941364a9da51b2b18247f13c98e14f8"}]},"branch":"refs/heads/master","description":"Rebase"},"45ae73ca9beebecb3dc8e6a976bcf0242e7c5ca6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-11-12 08:50:45.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/964280/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/964280/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/964280/3"}}},"commit":{"parents":[{"commit":"3b4ece41a4d0e70d8645beeb0c8fef061ce09de6","subject":"Add support for centos-10-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b4ece41a4d0e70d8645beeb0c8fef061ce09de6"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-17 07:39:25.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-11-12 08:50:31.000000000","tz":60},"subject":"Add c10s jobs to the check pipeline (non-voting)","message":"Add c10s jobs to the check pipeline (non-voting)\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/963653\n\nChange-Id: I2c1fea1febf6a984a64a02887981d6f6247fc570\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45ae73ca9beebecb3dc8e6a976bcf0242e7c5ca6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45ae73ca9beebecb3dc8e6a976bcf0242e7c5ca6"}]},"branch":"refs/heads/master"},"50ff38df3316174c997f7cc55f4ab7a444da6973":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-06 12:43:27.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/964280/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/964280/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/964280/4"}}},"commit":{"parents":[{"commit":"f192625c6af89d966065e50b0ea5f942ef7d5af7","subject":"Add support for centos-10-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f192625c6af89d966065e50b0ea5f942ef7d5af7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-17 07:39:25.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-06 12:42:43.000000000","tz":60},"subject":"Add c10s jobs to the check pipeline (non-voting)","message":"Add c10s jobs to the check pipeline (non-voting)\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/963653\n\nChange-Id: I2c1fea1febf6a984a64a02887981d6f6247fc570\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50ff38df3316174c997f7cc55f4ab7a444da6973"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50ff38df3316174c997f7cc55f4ab7a444da6973"}]},"branch":"refs/heads/master"},"e99ee4134b0555cc00583e2fbf201c0cccde5650":{"kind":"NO_CHANGE","_number":5,"created":"2026-01-08 13:47:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/964280/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/964280/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/964280/5"}}},"commit":{"parents":[{"commit":"297b61dd6d0f7a84f0b4ee1c2e17bd765eb1d4d9","subject":"Add support for centos-10-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/297b61dd6d0f7a84f0b4ee1c2e17bd765eb1d4d9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-17 07:39:25.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-08 13:47:00.000000000","tz":0},"subject":"Add c10s jobs to the check pipeline (non-voting)","message":"Add c10s jobs to the check pipeline (non-voting)\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/963653\n\nChange-Id: I2c1fea1febf6a984a64a02887981d6f6247fc570\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e99ee4134b0555cc00583e2fbf201c0cccde5650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e99ee4134b0555cc00583e2fbf201c0cccde5650"}]},"branch":"refs/heads/master","description":"Rebase"},"1343ffb8335de61ebecaa28655ef906eb15397fd":{"kind":"REWORK","_number":6,"created":"2026-01-08 14:10:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/964280/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/964280/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/964280/6"}}},"commit":{"parents":[{"commit":"eca8783959d55eb278bb8ddfa2b5c1392311d431","subject":"Add support for centos-10-stream \u0026 rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eca8783959d55eb278bb8ddfa2b5c1392311d431"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-17 07:39:25.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-08 14:09:47.000000000","tz":60},"subject":"Add c10s jobs to the check pipeline (non-voting)","message":"Add c10s jobs to the check pipeline (non-voting)\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/963653\n\nChange-Id: I2c1fea1febf6a984a64a02887981d6f6247fc570\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1343ffb8335de61ebecaa28655ef906eb15397fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1343ffb8335de61ebecaa28655ef906eb15397fd"}]},"branch":"refs/heads/master"},"4936de9d752b2c36d10179ee4a304a2e72255a49":{"kind":"REWORK","_number":7,"created":"2026-01-08 14:30:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/964280/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/964280/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/964280/7"}}},"commit":{"parents":[{"commit":"070d6b9494836481053b3026175ee85580595b8a","subject":"Add support for centos-10-stream \u0026 rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/070d6b9494836481053b3026175ee85580595b8a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-17 07:39:25.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-08 14:28:51.000000000","tz":60},"subject":"Add c10s jobs","message":"Add c10s jobs\n\nAdd scenario tests to the check pipeline (non-voting)\nUpdate centos amphora image build jobs\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/963653\n\nChange-Id: I2c1fea1febf6a984a64a02887981d6f6247fc570\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4936de9d752b2c36d10179ee4a304a2e72255a49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4936de9d752b2c36d10179ee4a304a2e72255a49"}]},"branch":"refs/heads/master"},"accdf4a207c2a89527cb1bfb21b623cf64da472d":{"kind":"REWORK","_number":8,"created":"2026-01-08 14:37:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/964280/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/964280/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/964280/8"}}},"commit":{"parents":[{"commit":"070d6b9494836481053b3026175ee85580595b8a","subject":"Add support for centos-10-stream \u0026 rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/070d6b9494836481053b3026175ee85580595b8a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-17 07:39:25.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-08 14:37:44.000000000","tz":60},"subject":"Add c10s jobs","message":"Add c10s jobs\n\nAdd scenario tests to the check pipeline (non-voting)\nUpdate centos amphora image build jobs\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/963653\n\nChange-Id: I2c1fea1febf6a984a64a02887981d6f6247fc570\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/accdf4a207c2a89527cb1bfb21b623cf64da472d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/accdf4a207c2a89527cb1bfb21b623cf64da472d"}]},"branch":"refs/heads/master"},"09d596417cb3369fe7c4e040d0f3047a1ecbc9da":{"kind":"REWORK","_number":9,"created":"2026-01-08 17:34:45.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/964280/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/964280/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/964280/9"}}},"commit":{"parents":[{"commit":"070d6b9494836481053b3026175ee85580595b8a","subject":"Add support for centos-10-stream \u0026 rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/070d6b9494836481053b3026175ee85580595b8a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-17 07:39:25.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-08 17:34:38.000000000","tz":60},"subject":"Add c10s jobs","message":"Add c10s jobs\n\nAdd scenario tests to the check pipeline (non-voting)\nUpdate centos amphora image build jobs\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/963653\n\nChange-Id: I2c1fea1febf6a984a64a02887981d6f6247fc570\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09d596417cb3369fe7c4e040d0f3047a1ecbc9da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09d596417cb3369fe7c4e040d0f3047a1ecbc9da"}]},"branch":"refs/heads/master"},"73b591c65d2ce927a583265b588abe198f64fd32":{"kind":"REWORK","_number":10,"created":"2026-01-09 07:40:22.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/964280/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/964280/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/964280/10"}}},"commit":{"parents":[{"commit":"070d6b9494836481053b3026175ee85580595b8a","subject":"Add support for centos-10-stream \u0026 rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/070d6b9494836481053b3026175ee85580595b8a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-17 07:39:25.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-09 07:40:13.000000000","tz":60},"subject":"Add c10s jobs","message":"Add c10s jobs\n\nAdd scenario tests to the check pipeline (non-voting)\nUpdate centos amphora image build jobs\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/963653\n\nChange-Id: I2c1fea1febf6a984a64a02887981d6f6247fc570\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73b591c65d2ce927a583265b588abe198f64fd32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73b591c65d2ce927a583265b588abe198f64fd32"}]},"branch":"refs/heads/master"},"7dfba000544831b0c1ce9b2a935396ef3470e01f":{"kind":"REWORK","_number":11,"created":"2026-01-09 14:32:37.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/964280/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/964280/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/964280/11"}}},"commit":{"parents":[{"commit":"070d6b9494836481053b3026175ee85580595b8a","subject":"Add support for centos-10-stream \u0026 rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/070d6b9494836481053b3026175ee85580595b8a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-17 07:39:25.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-09 14:31:53.000000000","tz":60},"subject":"Add c10s jobs","message":"Add c10s jobs\n\nAdd scenario tests to the check pipeline (non-voting)\nUpdate centos amphora image build jobs\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/963653\n\nChange-Id: I2c1fea1febf6a984a64a02887981d6f6247fc570\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7dfba000544831b0c1ce9b2a935396ef3470e01f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7dfba000544831b0c1ce9b2a935396ef3470e01f"}]},"branch":"refs/heads/master"},"b59b03c7b65da067f5315b0608ef577d91d53e8c":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-02-16 13:58:25.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/964280/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/964280/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/964280/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/964280/12"}}},"commit":{"parents":[{"commit":"dc5cdf83a784e36e159f096866ac0cc99e0bb0e8","subject":"Add support for centos-10-stream \u0026 rocky 10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dc5cdf83a784e36e159f096866ac0cc99e0bb0e8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-10-17 07:39:25.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-02-16 13:58:25.000000000","tz":0},"subject":"Add c10s jobs","message":"Add c10s jobs\n\nAdd scenario tests to the check pipeline (non-voting)\nUpdate centos amphora image build jobs\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/963653\n\nChange-Id: I2c1fea1febf6a984a64a02887981d6f6247fc570\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b59b03c7b65da067f5315b0608ef577d91d53e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b59b03c7b65da067f5315b0608ef577d91d53e8c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
