)]}'
{"id":"openstack%2Foctavia~965104","triplet_id":"openstack%2Foctavia~master~I1461df420f6292eeeeac0194884b451bd7e20445","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-11-06 13:22:36.000000000","reason":"Change was abandoned"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-11-06 13:22:36.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1461df420f6292eeeeac0194884b451bd7e20445","subject":"Fix new lint error","status":"ABANDONED","created":"2025-10-28 12:59:02.000000000","updated":"2025-11-06 13:22:36.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f12d589e2595a8570ed9eec8370ec72ed1a2a930","_number":965104,"virtual_id_number":965104,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-28 13:36:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-28 13:02:57.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2025-10-28 13:36:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf0c18c045cc2f16f80bc057aa3f00d5d7db83ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-28 12:59:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5db02f62bf6a78eb548c7cbdcf6642de5137e41","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-10-28 13:02:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"98152f9d364ff0bee62e42de600ec36d039f4ccb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-28 13:04:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cbb6d05d429eba9871ef16de1232d0da3992899d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 13:36:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/24e05ae5d2644ae7b94b24e737234987\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9fc666ac0164eb5a356a57c2784a66a : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24f0d63e84af425bac4360246dc74af7 : FAILURE in 6m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/746dba410caa4b13a72b6c656276b67c : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c07d30cc65444b02853890b7417d3dbc : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/871057c6796f41bb873d75feb359806c : SUCCESS in 6m 27s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a5e9351c53284b08856f36ca04e1e9db : SUCCESS in 4m 39s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/ef7688b674b04d68bde5f5da7502c305 : SUCCESS in 4m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a3bf0c957fe44ed8a85752b3813c49e6 : SUCCESS in 8m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/be2f53d056e84111a48fa81bacf6c396 : SUCCESS in 7m 35s\n\nWarning:\n  Comments left for invalid file octavia/tests/common/sample_data_models.py\n  Comments left for invalid file octavia/tests/functional/amphorae/backend/agent/api_server/test_keepalivedlvs.py\n  Comments left for invalid file octavia/tests/functional/amphorae/backend/agent/api_server/test_server.py\n  Comments left for invalid file octavia/tests/functional/api/drivers/driver_agent/test_driver_agent.py\n  Comments left for invalid file octavia/tests/functional/api/v2/base.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_health_monitor.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7policy.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7rule.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_listener.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_member.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_pool.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/functional/db/test_models.py\n  Comments left for invalid file octavia/tests/functional/db/test_repositories.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_amphora_info.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_haproxy_compatibility.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_keepalived.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_keepalivedlvs.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_loadbalancer.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_plug.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_util.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/test_agent_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/health_daemon/test_health_daemon.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_haproxy_query.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_interface.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_interface_file.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_ip_advertisement.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_keepalivedlvs_query.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_network_namespace.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_network_utils.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_nftable_utils.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/haproxy/test_exceptions.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/haproxy/test_rest_api_driver.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/haproxy/test_rest_api_driver_1_0.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/health/test_heartbeat_udp.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/keepalived/jinja/test_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/keepalived/test_vrrp_rest_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_get.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_listener.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_updater.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/noop_driver/test_agent.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_factory.py\n  Comments left for invalid file octavia/tests/unit/certificates/common/auth/test_barbican_acl.py\n  Comments left for invalid file octavia/tests/unit/certificates/common/test_barbican.py\n  Comments left for invalid file octavia/tests/unit/certificates/common/test_local.py\n  Comments left for invalid file octavia/tests/unit/certificates/generator/local_csr.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_barbican.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_barbican_legacy.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_castellan_mgr.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_local.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_noop.py\n  Comments left for invalid file octavia/tests/unit/cmd/test_health_checker.py\n  Comments left for invalid file octavia/tests/unit/common/jinja/logging/test_logging_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/common/jinja/lvs/test_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/common/jinja/test_user_data_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/common/sample_configs/sample_configs_combined.py\n  Comments left for invalid file octavia/tests/unit/common/test_base_taskflow.py\n  Comments left for invalid file octavia/tests/unit/common/test_config.py\n  Comments left for invalid file octavia/tests/unit/common/test_constants.py\n  Comments left for invalid file octavia/tests/unit/common/test_data_models.py\n  Comments left for invalid file octavia/tests/unit/common/test_decorators.py\n  Comments left for invalid file octavia/tests/unit/common/test_exceptions.py\n  Comments left for invalid file octavia/tests/unit/common/test_keystone.py\n  Comments left for invalid file octavia/tests/unit/common/test_policy.py\n  Comments left for invalid file octavia/tests/unit/common/test_utils.py\n  Comments left for invalid file octavia/tests/unit/common/test_validate.py\n  Comments left for invalid file octavia/tests/unit/common/tls_utils/test_cert_parser.py\n  Comments left for invalid file octavia/tests/unit/compute/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/compute/drivers/test_nova_driver.py\n  Comments left for invalid file octavia/tests/unit/controller/healthmanager/test_health_manager.py\n  Comments left for invalid file octavia/tests/unit/controller/housekeeping/test_house_keeping.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/test_amphora_rate_limit.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/test_task_utils.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_amphora_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_health_monitor_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_l7policy_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_l7rule_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_listener_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_load_balancer_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_member_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_pool_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_amphora_driver_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_cert_task.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_compute_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_database_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_database_tasks_quota.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_lifecycle_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_network_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_notification_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_retry_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_shim_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/test_controller_worker.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/test_taskflow_jobboard_driver.py\n  Comments left for invalid file octavia/tests/unit/image/drivers/test_glance_driver.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_allowed_address_pairs.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_base.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_utils.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/statistics/drivers/test_update_db.py\n  Comments left for invalid file octavia/tests/unit/test_opts.py\n  Comments left for invalid file octavia/tests/unit/test_version.py\n  Comments left for invalid file octavia/tests/unit/volume/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/volume/drivers/test_cinder_driver.py","accounts_in_message":[],"_revision_number":1},{"id":"f12d589e2595a8570ed9eec8370ec72ed1a2a930","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-06 13:22:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dae44ce1b9601d044f5cd2852b2a963ea3f7c55c","revisions":{"dae44ce1b9601d044f5cd2852b2a963ea3f7c55c":{"kind":"REWORK","_number":1,"created":"2025-10-28 12:59:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/965104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/04/965104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/965104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/965104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/965104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/04/965104/1"}}},"commit":{"parents":[{"commit":"ffd4fa757fd3228e4e8c05a8d79b73c2d338bf4c","subject":"Merge \"Fix devstack cleanup for log offload location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffd4fa757fd3228e4e8c05a8d79b73c2d338bf4c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-28 12:57:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-28 12:58:12.000000000","tz":540},"subject":"Fix new lint error","message":"Fix new lint error\n\nFix the following lint error detected in CI.\n\noctavia/tests/functional/db/base.py:85:8: R1705: mUnnecessary \"else\"\nafter \"return\", remove the \"else\" and de-indent the code inside it\n(no-else-return)\n\nChange-Id: I1461df420f6292eeeeac0194884b451bd7e20445\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dae44ce1b9601d044f5cd2852b2a963ea3f7c55c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dae44ce1b9601d044f5cd2852b2a963ea3f7c55c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
