)]}'
{"id":"openstack%2Foctavia~969201","triplet_id":"openstack%2Foctavia~master~If817bca8fd1aa11287e9ec7cf3f7fcdb23f099ad","project":"openstack/octavia","branch":"master","topic":"reduce-lb-get-recursion","hashtags":[],"change_id":"If817bca8fd1aa11287e9ec7cf3f7fcdb23f099ad","subject":"Enhance to_data_model recursion limit","status":"ABANDONED","created":"2025-12-02 09:27:33.000000000","updated":"2025-12-11 10:50:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1a9fdd9ef4766a1655e9d487cbc316eb8ba83e7e","_number":969201,"virtual_id_number":969201,"owner":{"_account_id":38376,"name":"Théo Van Gyzel","display_name":"Théo VG","email":"theo.van-gyzel@ovhcloud.com","username":"theovg"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-12-04 12:25:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-02 12:04:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d88d791e56fe53f55b6b97c1ddc70803dc53f720","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38376,"name":"Théo Van Gyzel","display_name":"Théo VG","email":"theo.van-gyzel@ovhcloud.com","username":"theovg"},"date":"2025-12-02 09:27:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8206a7a689aea20530ec50ea89e5aa7141d4ef49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 12:04:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/526edd72565f46f39ba12286d33aaf6c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8877f1ccc0a24fc39c6f64d37d33f95e : SUCCESS in 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e5e9f1a0f65498eab7def240f18e40a : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d58b87ab101043349628d5130172de6f : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f580968d2954bf1833937097b0d2dd8 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c6294b5ff6404c802a8de71bb5f858 : SUCCESS in 13m 09s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ccb68190d30d4d3884ed2a57da49dbd5 : SUCCESS in 6m 00s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/8c2dcffb19604f7d9717ad0d1bb10351 : SUCCESS in 4m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/485c7563c8a94c77bfc776c67b22f4c1 : SUCCESS in 3m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33bf890c303c43f99089b4f29b1e74dd : SUCCESS in 8m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e7e7cb1467444d86b27f489ea145d34d : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d840abb0f5a4f32ac98f106330d06a7 : SUCCESS in 1h 26m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6a6b8113dcfe4728a0b9036b88317162 : SUCCESS in 1h 38m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6ea56de2183a4237a57b87d389b8e87d : SUCCESS in 1h 02m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/917272291ee24568a7bb0142c70085a0 : SUCCESS in 1h 40m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2c2af783edf34cc9b9c532311253ddc6 : SUCCESS in 1h 12m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3a922e2c045145b8b1125b5c560a0e46 : SUCCESS in 38m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e7bb4cc945f5475dbdbbf2f1d2af3455 : SUCCESS in 42m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a200ff9f4d3645378ea48767c6a442ad : SUCCESS in 35m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/37a24d83e30141fa99b52f8bd37b5da6 : SUCCESS in 33m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e1de7183cd574ae3a51af5422b94371f : SUCCESS in 2h 35m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/55c58a04cecd40efb15dbb06c5290c4c : SUCCESS in 1h 33m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3b7cf6846eb83350776a803cef98f31cec472bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38376,"name":"Théo Van Gyzel","display_name":"Théo VG","email":"theo.van-gyzel@ovhcloud.com","username":"theovg"},"date":"2025-12-04 09:42:15.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"38c0e2c1a6a91ed354003ad98959d77b0c582975","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 12:25:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a8cda0bc4b24506abbba2b69ea2a2b6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d67aa8323e79419995ecc4016f7b13c0 : SUCCESS in 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0bd67906acb4190893049120773c9d6 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/842e367df0e24c6c891404b50a1c507a : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2c363a01b384a508c8945fc2645f9ec : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10c0cced04a1473bb125f54f5dc1768a : SUCCESS in 13m 08s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/55bacdb92d5141cebccc2c77887822b7 : SUCCESS in 4m 48s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/1ce64f5c199f42a0ad2a7cac9e9f4d87 : SUCCESS in 6m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f7de09ad09f2411bb2f960e15b4dcf42 : SUCCESS in 4m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/893c8930dd874829b1b191c2e077efb6 : SUCCESS in 9m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f399cd72ee5646f0a5538730314c0b1b : SUCCESS in 8m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/90390ee1263b4d5ea70a5f704b37cde7 : SUCCESS in 1h 33m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8c6a794f10de4472abcdcbf1da02b85a : SUCCESS in 1h 34m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0c587b0c60f040fbbf63b49acb9a2206 : SUCCESS in 1h 05m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4d0ef6b18b084d0db27dcd1464a598e3 : SUCCESS in 1h 42m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/997303994aa942d1ba4f21cbf504f43f : SUCCESS in 1h 20m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9bb8c2229dbf4d55b38333060372a368 : SUCCESS in 39m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/54747cad72b84862afe03c45eb80bfd5 : SUCCESS in 45m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/231868d22c8846f380c8f7f52a25a4f1 : SUCCESS in 34m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d6b99894e3d24f168b4b451a8f00ef5d : SUCCESS in 37m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/524f2ca8a2dd414a97920d843d1ebc2a : SUCCESS in 2h 41m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2625b806b60d4fe7a5dc5fc7b6912611 : SUCCESS in 1h 35m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1a9fdd9ef4766a1655e9d487cbc316eb8ba83e7e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38376,"name":"Théo Van Gyzel","display_name":"Théo VG","email":"theo.van-gyzel@ovhcloud.com","username":"theovg"},"date":"2025-12-11 10:50:37.000000000","message":"Abandoned\n\nAbandoning in favour of https://review.opendev.org/c/openstack/octavia/+/970404","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b58d96b23367a6914c79867b632d45d0974158ad","revisions":{"de149507c0a254292e5ac3ee78ac60ecf38d54d4":{"kind":"REWORK","_number":1,"created":"2025-12-02 09:27:33.000000000","uploader":{"_account_id":38376,"name":"Théo Van Gyzel","display_name":"Théo VG","email":"theo.van-gyzel@ovhcloud.com","username":"theovg"},"ref":"refs/changes/01/969201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/969201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/969201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/969201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/969201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/969201/1"}}},"commit":{"parents":[{"commit":"be37addac0a73afe7422e43e4dd0f2e283805567","subject":"Merge \"reno: Update master for unmaintained/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be37addac0a73afe7422e43e4dd0f2e283805567"}]}],"author":{"name":"Théo Van Gyzel","email":"theo.van-gyzel@ovhcloud.com","date":"2025-12-02 09:19:23.000000000","tz":60},"committer":{"name":"Théo Van Gyzel","email":"theo.van-gyzel@ovhcloud.com","date":"2025-12-02 09:27:31.000000000","tz":60},"subject":"Enhance to_data_model recursion limit","message":"Enhance to_data_model recursion limit\n\nThis patch offers performance enhancements for simple API gets on very large load balancers by limiting recursion of to_data_model when getting the load balancer from the database (specifically by not processing members and health monitors anymore as they are not returned in the response anyways).\n\nIn order to do so, this patch also unwraps the to_data_model limited_graph feature implemented in a previous patch into a recursion_depth override that allows for finer control over depth instead of a simple on/off toggle.\n\nChange-Id: If817bca8fd1aa11287e9ec7cf3f7fcdb23f099ad\nSigned-off-by: Théo Van Gyzel \u003ctheo.van-gyzel@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de149507c0a254292e5ac3ee78ac60ecf38d54d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de149507c0a254292e5ac3ee78ac60ecf38d54d4"}]},"branch":"refs/heads/master"},"b58d96b23367a6914c79867b632d45d0974158ad":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-12-04 09:42:15.000000000","uploader":{"_account_id":38376,"name":"Théo Van Gyzel","display_name":"Théo VG","email":"theo.van-gyzel@ovhcloud.com","username":"theovg"},"ref":"refs/changes/01/969201/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/969201/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/969201/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/969201/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/969201/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/969201/2"}}},"commit":{"parents":[{"commit":"be37addac0a73afe7422e43e4dd0f2e283805567","subject":"Merge \"reno: Update master for unmaintained/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be37addac0a73afe7422e43e4dd0f2e283805567"}]}],"author":{"name":"Théo Van Gyzel","email":"theo.van-gyzel@ovhcloud.com","date":"2025-12-02 09:19:23.000000000","tz":60},"committer":{"name":"Théo Van Gyzel","email":"theo.van-gyzel@ovhcloud.com","date":"2025-12-04 09:41:28.000000000","tz":60},"subject":"Enhance to_data_model recursion limit","message":"Enhance to_data_model recursion limit\n\nThis patch offers performance enhancements for simple API gets on very large load balancers by limiting recursion of to_data_model when getting the load balancer from the database (specifically by not processing members and health monitors anymore as they are not returned in the response anyways).\n\nIn order to do so, this patch also unwraps the to_data_model limited_graph feature implemented in a previous patch into a recursion_depth override that allows for finer control over depth instead of a simple on/off toggle.\n\nCloses-Bug: 2133856\nChange-Id: If817bca8fd1aa11287e9ec7cf3f7fcdb23f099ad\nSigned-off-by: Théo Van Gyzel \u003ctheo.van-gyzel@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b58d96b23367a6914c79867b632d45d0974158ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b58d96b23367a6914c79867b632d45d0974158ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
