)]}'
{"id":"openstack%2Foctavia~969487","triplet_id":"openstack%2Foctavia~stable%2F2024.2~I34341e1aaad1524e3e0834309c5f6897f176af53","project":"openstack/octavia","branch":"stable/2024.2","topic":"780215-patch4-stable/2024.2","attention_set":{},"removed_from_attention_set":{"38360":{"account":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"last_update":"2026-01-23 23:24:11.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-01-23 21:36:11.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-01-23 07:40:52.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-23 23:24:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I34341e1aaad1524e3e0834309c5f6897f176af53","subject":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API","status":"MERGED","created":"2025-12-03 15:56:55.000000000","updated":"2026-01-23 23:25:11.000000000","submitted":"2026-01-23 23:24:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969487-780215-patch4-stable/2024.2","cherry_pick_of_change":780215,"cherry_pick_of_patch_set":10,"meta_rev_id":"b721777235f5826bee96313a66c7105f463b17b1","_number":969487,"virtual_id_number":969487,"owner":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-23 23:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2026-01-23 07:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2026-01-23 21:36:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2026-01-23 21:36:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-03 15:56:55.000000000","updated_by":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2025-12-03 17:48:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-10 12:57:52.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2025-12-11 18:50:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-12-11 18:50:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"324392040a5db77521577288d67f33b477971ab4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2025-12-03 15:56:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5a79661091ef3d3aa1626bbcdba1a73910a18fbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 17:48:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cda0ee7ee18400b80c37663fdd83102\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f49a4f67429423587c42d1e0262fdc0 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17c2f612c949472ba1a8569e735ede88 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1cf1aef939f4f7281fbdd5b20736ed4 : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20653d63bc244078b4707d2ebda806c9 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/027df63ab92e4fd99f59534ececf91e7 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfc87d33ffa849c08c33b28e5df0e344 : SUCCESS in 11m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/919780c578634d43bf5c66802ca4c94d : SUCCESS in 5m 39s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/695c08077eb44fb78283490152400a49 : SUCCESS in 4m 58s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/0e63fa17287a4f14953d7ca6acc85e49 : SUCCESS in 7m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ae12b662d1bd48acb7e4061f27c26e75 : SUCCESS in 5m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f0cc63cf069e4e419860e59ba44c8fb2 : SUCCESS in 7m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/441d5e71f0634eb6ae1d116d86cd749c : SUCCESS in 6m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8a6aef9ebd1a4f94b718aaadb90d661a : SUCCESS in 4m 44s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/2cfd05d8730449deb59005c0d6911d74 : SUCCESS in 1h 14m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/c7615d7e3bed4f308acea2c2f953851f : SUCCESS in 1h 42m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/424af950dfa545318eb684a4eb4eab4a : SUCCESS in 57m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b18189a1c8a14cbfb7dc3c27c6db5dff : SUCCESS in 1h 45m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/64b79532db164a1c985e18193b96346b : SUCCESS in 1h 07m 49s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/2962b56d3151489cbbbf41b98763c662 : SUCCESS in 32m 36s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/6c7e718520df4eeebb18bf7526781599 : SUCCESS in 37m 51s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f39b8c64cfcb45fbad5e5227350683b0 : FAILURE in 18m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/e18e1054ddb0463cabbf88561463df97 : SUCCESS in 31m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-2 https://zuul.opendev.org/t/openstack/build/e605219ef90c471199c96944e76d09e6 : SUCCESS in 31m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-2 https://zuul.opendev.org/t/openstack/build/6f1c0014d6af4dd58c71d16c63b8da7b : FAILURE in 42m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-2 https://zuul.opendev.org/t/openstack/build/391216deda3e41f19db6ffb78f6211b0 : SUCCESS in 1h 50m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b3eeb28fbaed4117b4e8df9aee4e8ba2 : FAILURE in 17m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b20fb7ca795b815bb14c6b26611bf9d622fbb28","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-10 12:57:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5bbf012b6df42166e3e31293fe82ed8097051a8c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 18:50:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ac852202479f2c9f2974f3ea5604c543ee70f518","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-01-14 22:24:30.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1a1aad1e94d236a507d30986e7cf059e1db4aa42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 00:11:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccaae0bf511c4db2823b044cff87ca25\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/884d475ae24a4694aaf017e6c8d2a433 : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58d782c5e63f466a927dbc78673e50f3 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccaf8b6b05c64cad9d25896308cd7aeb : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f1f9304bac24114932e4eb63e5a630b : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5556a1b44854ec8b1251b40c7da85fa : SUCCESS in 3m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/748bc27205c24b2183197853b448c2f6 : SUCCESS in 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6eab636af1f34f358229c1bcb7fb8367 : SUCCESS in 5m 34s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8e998d135a6d4623821188b4fdd810b6 : SUCCESS in 4m 44s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/6550c838f6d54d3a93010ff4c8a917eb : SUCCESS in 5m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/35ef39531892433982e52e80cb7cf6f5 : SUCCESS in 4m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/882f8b7a468d4c9987ad681a703629c1 : SUCCESS in 7m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/35de216c645e480ca8e42fc0be92e249 : SUCCESS in 6m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3f4c66a95df6436b8e51331c65019d93 : SUCCESS in 3m 44s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/652873e1cadf4ab8823cbdc8489f3927 : SUCCESS in 59m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/11fe0839aef14128b3e1c3e07f5a97c7 : SUCCESS in 1h 43m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4a5b7eaeca5042e3a7c1612b0bae3ff8 : SUCCESS in 1h 00m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/2aca89fe8a7d46d586b7d38d57512bfc : SUCCESS in 1h 45m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ded9d6609082445c8d97ed67d032fcaa : SUCCESS in 56m 10s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/aac6dc927b494cb2b9e66497a938c2b3 : SUCCESS in 32m 47s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/65a69579c3464885a768693dcea420dd : SUCCESS in 38m 07s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-2 https://zuul.opendev.org/t/openstack/build/e08e89ec8b1243d492b960cc3f96464d : FAILURE in 18m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7b61972d91da4ab7b99d35c5567c42c7 : SUCCESS in 28m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-2 https://zuul.opendev.org/t/openstack/build/0e1cd847e777425cb6292dfe65e12f7b : SUCCESS in 31m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-2 https://zuul.opendev.org/t/openstack/build/96cbdca3da364771bc0666878f4dc42d : FAILURE in 40m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4d29e695d9d5411cabc86dcd92c5e53e : SUCCESS in 1h 41m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f0c468834ed34f4fa51e827eb4bec078 : FAILURE in 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"42107b4a6158552062faa7df20043f5bdf9b5182","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-23 07:40:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"668f35e6fc129c9350ef4bc2ec18a4d01c5d024b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2026-01-23 21:36:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5000ed6754cfef7bd00eac1768ec69a5a77df3c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 21:36:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e84c3f9d05192cca5d10bf7d1aa95899b2f22131","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 23:24:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b3503c049c4476a8e4c93accb4d988b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a4ce2108d74a51b30441c934329587 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3a76bb459a24655a440c3996adaf58d : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e89fba0722a4b7dac0bbeab734f5c65 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e4fb25b66404ce8b6bec33560a8ff90 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b24257634f9e4bf3ac4c4965c94154bb : SUCCESS in 6m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1cfb338b0bb4492789e4644841e33f5d : SUCCESS in 4m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4a74a6427d7349dc8069988dea92ef28 : SUCCESS in 8m 57s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/0e8104f3d65743afb92b4e10b738962e : SUCCESS in 59m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/9ea9709d4ded4c32940e24a346b462f8 : SUCCESS in 1h 42m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ac811edd935941e8814dcd17da820dec : SUCCESS in 1h 03m 53s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/06e08ce376ec48eeb5f4836d01022c59 : SUCCESS in 32m 37s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/8df8b6ffce5b4412a9a1cc9dd81bf844 : SUCCESS in 38m 05s","accounts_in_message":[],"_revision_number":2},{"id":"abbed3e92e4dc5229a4a7abd4d922dbe8c2714d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 23:24:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b721777235f5826bee96313a66c7105f463b17b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 23:25:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7810487ae53340a6a33e0b101c387bab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b9e9386536549c1879a58cf877a4269 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6df964944fa143b2bfe3910ba96a9a66 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f3ed2dc18ecdf377cf6572059b3a750be054f958","revisions":{"5464aa1c95c146968bb5aeec4c5616e16d4733ee":{"kind":"REWORK","_number":1,"created":"2025-12-03 15:56:55.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/87/969487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/969487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/969487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/969487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/969487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/969487/1"}}},"commit":{"parents":[{"commit":"577ff3d6712f0235b3fda3ae3f89ce1b5582d2a8","subject":"Merge \"Fix L7Rules with FILE_TYPE and EQUAL_TO\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/577ff3d6712f0235b3fda3ae3f89ce1b5582d2a8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-11 17:50:06.000000000","tz":60},"committer":{"name":"Zachary Mark Raines","email":"zachary.raines@canonical.com","date":"2025-12-03 15:56:55.000000000","tz":0},"subject":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API","message":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\n\nFix an issue that prevents the Octavia API service to be correctly\ninitialized when it fails to load a provider driver.\n\nWhen enabled_provider_drivers setting is not correctly configured (ex:\nif it contains a non existing driver) or when a provider driver fails\nto load, the exception was not caught by the Octavia API service, so the\nservice was not properly configured and the whole Octavia API was\nunreachable.\n\nNow the Octavia API service skips the driver initialization in case of\nerrors, removes the driver(s) from the enabled list, and the other\nprovider drivers are functional.\n\nStory 2008710\nTask 42044\n\nDepends-on: https://review.opendev.org/c/openstack/octavia/+/964892\nChange-Id: I34341e1aaad1524e3e0834309c5f6897f176af53\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n(cherry picked from commit 6fc1abb780a2fc1413d5b206cec4b033a638c3c4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5464aa1c95c146968bb5aeec4c5616e16d4733ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5464aa1c95c146968bb5aeec4c5616e16d4733ee"}]},"branch":"refs/heads/stable/2024.2"},"f3ed2dc18ecdf377cf6572059b3a750be054f958":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-01-14 22:24:30.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/87/969487/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/969487/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/969487/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/969487/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/969487/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/969487/2"}}},"commit":{"parents":[{"commit":"577ff3d6712f0235b3fda3ae3f89ce1b5582d2a8","subject":"Merge \"Fix L7Rules with FILE_TYPE and EQUAL_TO\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/577ff3d6712f0235b3fda3ae3f89ce1b5582d2a8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-11 17:50:06.000000000","tz":60},"committer":{"name":"Zachary Mark Raines","email":"zachary.raines@canonical.com","date":"2026-01-14 22:24:30.000000000","tz":0},"subject":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API","message":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\n\nFix an issue that prevents the Octavia API service to be correctly\ninitialized when it fails to load a provider driver.\n\nWhen enabled_provider_drivers setting is not correctly configured (ex:\nif it contains a non existing driver) or when a provider driver fails\nto load, the exception was not caught by the Octavia API service, so the\nservice was not properly configured and the whole Octavia API was\nunreachable.\n\nNow the Octavia API service skips the driver initialization in case of\nerrors, removes the driver(s) from the enabled list, and the other\nprovider drivers are functional.\n\nStory 2008710\nTask 42044\n\n\nChange-Id: I34341e1aaad1524e3e0834309c5f6897f176af53\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n(cherry picked from commit 6fc1abb780a2fc1413d5b206cec4b033a638c3c4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f3ed2dc18ecdf377cf6572059b3a750be054f958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f3ed2dc18ecdf377cf6572059b3a750be054f958"}]},"branch":"refs/heads/stable/2024.2","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
