)]}'
{"id":"openstack%2Foctavia~971050","triplet_id":"openstack%2Foctavia~master~I8cd8784ab13ed44fee978aaf5acee310fc7cc51c","project":"openstack/octavia","branch":"master","attention_set":{"38678":{"account":{"_account_id":38678,"name":"Mikhail Chernobrov","display_name":"M. Chernobrov","email":"chernobrov.m@selectel.com","username":"chernobrov-me"},"last_update":"2025-12-15 14:44:37.000000000","reason":"\u003cGERRIT_ACCOUNT_6577\u003e replied on the change","reason_account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-12-15 14:44:37.000000000","reason":"\u003cGERRIT_ACCOUNT_6577\u003e replied on the change","reason_account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"}}},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-12-19 13:46:08.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I8cd8784ab13ed44fee978aaf5acee310fc7cc51c","subject":"Listener allowed CIDRs bounds support","status":"NEW","created":"2025-12-15 14:31:26.000000000","updated":"2025-12-19 13:46:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cba63068eb37fce7cc33ec9f2af8ce07c7ceece7","_number":971050,"virtual_id_number":971050,"owner":{"_account_id":38678,"name":"Mikhail Chernobrov","display_name":"M. Chernobrov","email":"chernobrov.m@selectel.com","username":"chernobrov-me"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-12-15 17:04:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"all":[{"value":1,"date":"2025-12-15 14:44:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":1,"date":"2025-12-15 14:44:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},{"value":1,"date":"2025-12-15 14:47:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},{"value":-2,"date":"2025-12-19 13:46:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":1,"date":"2025-12-15 14:44:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-15 14:44:35.000000000","updated_by":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"reviewer":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"state":"REVIEWER"},{"updated":"2025-12-15 14:44:37.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"REVIEWER"},{"updated":"2025-12-15 14:44:37.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-12-15 14:44:37.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2025-12-15 14:47:16.000000000","updated_by":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"reviewer":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"state":"REVIEWER"},{"updated":"2025-12-15 17:04:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1cbb2f6b9c9517fe1b19989e4c083394512b5e45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38678,"name":"Mikhail Chernobrov","display_name":"M. Chernobrov","email":"chernobrov.m@selectel.com","username":"chernobrov-me"},"date":"2025-12-15 14:31:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa8f40b28ce2ec972df41f443379d4d99cdb62fc","author":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"},"date":"2025-12-15 14:44:35.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e7b3cffe709725b160c20f6c38bd84bc9a1a8256","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-12-15 14:44:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0ea60635adff41048fde337ca2bcd0bfb0b169ee","author":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"date":"2025-12-15 14:47:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"df07ebb97ee73e4cb561024ea42f9c376b5383e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 17:04:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c034cf59594d4716b4df2512a87961b0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6aedd0054fdf467db512c88a9b24e076 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/235eb463b7bd429c9a9f69404521862b : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/221ece88fae24e8cbf29289ae4a7071b : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf7cb88a041440068d2986f2eb6adc02 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe73b6fef61b42aa9cdb48020d39be6f : SUCCESS in 6m 05s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0696d5c8b7744f089f1a143d7e34ce2f : SUCCESS in 4m 14s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/cdbedccca3804a39bb937ff3098d1708 : SUCCESS in 5m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3ac639d0dfd943958acd5846c468ca5d : SUCCESS in 2m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8d09f66e53864f5bbfb02fc784c512e1 : SUCCESS in 7m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fbd3bb2769214c4f83a51d97b7db3cc3 : SUCCESS in 3m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/64207a802df64cc28cbeef4307c8e6c9 : SUCCESS in 1h 24m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ef4f34c7be6c462d8806b7fc8a4c26e6 : SUCCESS in 1h 34m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/273b9daf800049e49059a3cf70df7f5f : SUCCESS in 1h 05m 21s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bdb40c82e2724cebb166d716c784373b : SUCCESS in 1h 41m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/592cf3d73a754a6ba8e4472542ca63ee : SUCCESS in 1h 06m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/35373a71aa964a8ba43551efa7d9a695 : SUCCESS in 36m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c701fbaf0aac4acea733dba5e04b13ad : SUCCESS in 41m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ddaa65a0a5b44b8a49b66decd8e3fad : SUCCESS in 33m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6eb10cea595a4892a1beabdf60d3ae67 : SUCCESS in 35m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b810dd73f5bb4f3392707efc5233654d : SUCCESS in 2h 29m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/15340aa21b6a43488f72beee0a38b5c0 : SUCCESS in 1h 32m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cba63068eb37fce7cc33ec9f2af8ce07c7ceece7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-19 13:46:08.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0a12063883d2f2677b2cc1611e954ed10a7f515","revisions":{"d0a12063883d2f2677b2cc1611e954ed10a7f515":{"kind":"REWORK","_number":1,"created":"2025-12-15 14:31:26.000000000","uploader":{"_account_id":38678,"name":"Mikhail Chernobrov","display_name":"M. Chernobrov","email":"chernobrov.m@selectel.com","username":"chernobrov-me"},"ref":"refs/changes/50/971050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/50/971050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/971050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/971050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/971050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/50/971050/1"}}},"commit":{"parents":[{"commit":"f42860a7302b521dd2895a22535dbd1de1f912f0","subject":"Remove url tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f42860a7302b521dd2895a22535dbd1de1f912f0"}]}],"author":{"name":"chernobrov.m","email":"chernobrov.m@selectel.com","date":"2025-12-15 14:11:21.000000000","tz":180},"committer":{"name":"chernobrov.m","email":"chernobrov.m@selectel.com","date":"2025-12-15 14:31:00.000000000","tz":180},"subject":"Listener allowed CIDRs bounds support","message":"Listener allowed CIDRs bounds support\n\nImplementation of the WSME custom user type - ConstraintArray. This type supports setting restrictions: min_items, max_items and unique items. The type is used in POST and PUT models for Octavia load balancer listener to limit the `allowed_cidrs` field, for example, to max 200 items. Unit and functional tests are implemented. The work was manually checked through the Openstack CLI in the test and production environments.\n\nCloses-Bug: 2136128\nSigned-off-by: Mikhail Chernobrov \u003cchernobrov.m@selectel.com\u003e\nChange-Id: I8cd8784ab13ed44fee978aaf5acee310fc7cc51c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0a12063883d2f2677b2cc1611e954ed10a7f515"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0a12063883d2f2677b2cc1611e954ed10a7f515"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":37685,"name":"Konstantin","display_name":"Konstantin Eremin","email":"k0dt@k0dt.ru","username":"k0dt"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
