)]}'
{"id":"openstack%2Foctavia~971323","triplet_id":"openstack%2Foctavia~master~I855a5a147573bb6ba6ba92958e02a71d9ff432b0","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-12-17 17:11:58.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I855a5a147573bb6ba6ba92958e02a71d9ff432b0","subject":"DNM/WIP Testing TestHAproxyVRRPCheckCMD.test_health_check","status":"ABANDONED","created":"2025-12-17 13:58:00.000000000","updated":"2025-12-18 11:57:36.000000000","total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ddff7eab8f634465e5ebdc443fe31ebe49bd885c","_number":971323,"virtual_id_number":971323,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-12-17 17:42:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-17 14:03:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b845e356ff6c8c81ce7adc6c428ceddec68f82f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 13:58:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e6765318ace78f0bc75084d56218c098261d9e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 14:03:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aa93b8f8dc34378ad6666b9aaa07fe0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c74e5bb16f5a4412be15cb1ebb11fba5 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a50a6bb7a66345e296a8ec991ea9b272 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d6883176aab49ca9a6d228bee857548 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e162bf1dc21390e2d7aae771272eaf41ef601180","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 14:04:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3da25a474faa10917724e994df7c789f88b7b632","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 14:14:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75a73574639144558c472554e2f06a0a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf5e7003d3af4c36b4ca1cd1f835eb64 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d37c95d493ea4cef9112eef340f88433 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cfa69ec7aef64eeea7ce6e550153a944 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"826b1f674c6b9b840e87165bb341d5ce459dbc5a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 14:19:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32cfadf24ebe4a48d08ffec2cf20126c0a64b062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 14:31:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35924d486db3468ca689411eee19268c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c410842cf6f43a69682a5a6c5eea6c8 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/412f59694edd4b938dea9526ba116a7c : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6f92394f54a41e99ad90fe1bbfde908 : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f81d091527e381a90c0a1993a4e9bdc21cf732f7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 14:43:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"034a090d39e4869df97ba1d09bedcc57c7e45531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 14:55:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87d88bb2d94149efad9593d7706a6a22\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3f66fdf74b445acb1954c6b558b9a07 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3eabf58d53a4a6bb8da50795aca7ed8 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e938ac291cff43e58a8fd9631cc688f1 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"adfbe854ad5de81303134fd1ee0efd97125552db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 14:55:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3784431ca11158ec640140c80ea94c9d07386162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 15:07:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/832ef455aef443ad9d960bbc26e5b879\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40d8f9de3c54499d8e7a671d4668eb1f : SUCCESS in 6m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce1025b13e134e1283959103b2430fcd : SUCCESS in 7m 00s\n- openstack-tox-py310-2 https://zuul.opendev.org/t/openstack/build/a039e5550f5146f6a48af1b13e43f130 : SUCCESS in 4m 35s\n- openstack-tox-py313-2 https://zuul.opendev.org/t/openstack/build/e6117899208f4e9280b24cc0386bac99 : SUCCESS in 9m 14s\n- openstack-tox-py310-3 https://zuul.opendev.org/t/openstack/build/4da6b916b1ff4d81a398ae102398e77d : SUCCESS in 4m 37s\n- openstack-tox-py313-3 https://zuul.opendev.org/t/openstack/build/19c04459cd36481d93b9bf0bd24929f0 : SUCCESS in 5m 48s\n- openstack-tox-py310-4 https://zuul.opendev.org/t/openstack/build/28d2949bc5734cc68732f5dde2466fd3 : SUCCESS in 2m 50s\n- openstack-tox-py313-4 https://zuul.opendev.org/t/openstack/build/9f5097b25eeb4bc3b1e440f91d10904f : SUCCESS in 4m 08s\n- openstack-tox-py310-5 https://zuul.opendev.org/t/openstack/build/1208f17a0b3141dfa741132712bff457 : FAILURE in 3m 52s\n- openstack-tox-py313-5 https://zuul.opendev.org/t/openstack/build/6237fc8507a146c09769987fbd5c6228 : SUCCESS in 5m 02s\n- openstack-tox-py310-6 https://zuul.opendev.org/t/openstack/build/8db76ea740cd442694586c4160b85a11 : SUCCESS in 3m 33s\n- openstack-tox-py313-6 https://zuul.opendev.org/t/openstack/build/7c78881d58fb4bbf82f8cbdeb23733f2 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":2},{"id":"17a77cbb0476d6267417b169952fc249a8be6232","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 15:07:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0c52159970b6fcb5fab1c8d9d12e2104f6ed4cf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 15:21:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eea17f66010541bf8e6f3b7f75226018\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4acbcc335a84187b94b89abb0b79adf : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f94949aac58d48fbab3e022becdcdb9a : SUCCESS in 5m 35s\n- openstack-tox-py310-2 https://zuul.opendev.org/t/openstack/build/76923b3eef6049ad9da449a22786734b : SUCCESS in 3m 02s\n- openstack-tox-py313-2 https://zuul.opendev.org/t/openstack/build/d1ae8eca0c4a492c90e422a49efc176f : SUCCESS in 7m 51s\n- openstack-tox-py310-3 https://zuul.opendev.org/t/openstack/build/08bc61fa35fa4ad09ef008a7c43fdd3f : SUCCESS in 5m 58s\n- openstack-tox-py313-3 https://zuul.opendev.org/t/openstack/build/47ada49b6ef644d7a9f97aef2895b00e : SUCCESS in 7m 08s\n- openstack-tox-py310-4 https://zuul.opendev.org/t/openstack/build/7a3410b20d2b4962aeed14a222d7c7f7 : SUCCESS in 4m 44s\n- openstack-tox-py313-4 https://zuul.opendev.org/t/openstack/build/d00202f474714f1cab877a7fe46b9bd5 : SUCCESS in 6m 44s\n- openstack-tox-py310-5 https://zuul.opendev.org/t/openstack/build/efd3a6a1a4e54b0f83f04d7cb89856a0 : SUCCESS in 4m 23s\n- openstack-tox-py313-5 https://zuul.opendev.org/t/openstack/build/b78d0821b6154704b8877f9053dce5f2 : SUCCESS in 6m 39s\n- openstack-tox-py310-6 https://zuul.opendev.org/t/openstack/build/e70e2a2874dc4da7ae3a4f3d40d9216b : SUCCESS in 3m 59s\n- openstack-tox-py313-6 https://zuul.opendev.org/t/openstack/build/1a5a4921ad254107b3f1be67b5064bf9 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":3},{"id":"468671631197aeeb70ee9d3ad43563f19bce3b8f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 15:26:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c8983d0dd1c3fbc5e0d5465a3e69aa660a4957a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 15:54:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1de6d49cf974f09b5688930cf7b6514\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/466b4653941a419e840835d7c23dc9fb : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2da82efc4968455fab93cb86678346a6 : SUCCESS in 6m 43s\n- openstack-tox-py310-2 https://zuul.opendev.org/t/openstack/build/0bec2cd973ed41aa959ac13a3a45d68d : SUCCESS in 4m 29s\n- openstack-tox-py313-2 https://zuul.opendev.org/t/openstack/build/6a7aaf2fb7e8478a9d3307d1ca8ffc0c : SUCCESS in 9m 56s\n- openstack-tox-py310-3 https://zuul.opendev.org/t/openstack/build/587bd7d668e04991a583ac372b97597e : SUCCESS in 4m 30s\n- openstack-tox-py313-3 https://zuul.opendev.org/t/openstack/build/3115287bfaed46b29e87fe34642d09ba : SUCCESS in 6m 43s\n- openstack-tox-py310-4 https://zuul.opendev.org/t/openstack/build/6b67a810fe254dd7bc2c79fb03775fac : SUCCESS in 4m 08s\n- openstack-tox-py313-4 https://zuul.opendev.org/t/openstack/build/32e8ee5043f443c3ba7ccdc5c4f82a6c : SUCCESS in 8m 43s\n- openstack-tox-py310-5 https://zuul.opendev.org/t/openstack/build/10574b5504814de3ab7e799c393f3636 : SUCCESS in 5m 19s\n- openstack-tox-py313-5 https://zuul.opendev.org/t/openstack/build/0e15096621f745e2b360fd35f5c3f5bd : SUCCESS in 7m 07s\n- openstack-tox-py310-6 https://zuul.opendev.org/t/openstack/build/4ebad30993954b1dbd20648018002079 : SUCCESS in 4m 23s\n- openstack-tox-py313-6 https://zuul.opendev.org/t/openstack/build/efe60240862640d38ea59ef2e371f71b : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":3},{"id":"1a6d70101377b40eaca30091d751a97586688be4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 15:57:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"477b6c68db25a99ded657acacc9353561ef398e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 16:15:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84b0104381234a2e8b7d06cadee9b13f\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/674fdc11b91a42e2b685cd03a990fe23 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be06aba3f8f34824bf463343d6b78b5f : SUCCESS in 7m 23s\n- openstack-tox-py310-2 https://zuul.opendev.org/t/openstack/build/bc8522aa4ecf43caa2d48d8599c4fd13 : SUCCESS in 5m 57s\n- openstack-tox-py313-2 https://zuul.opendev.org/t/openstack/build/36b9e6b4f8fe4ab7877048d7c2ff069b : SUCCESS in 7m 52s\n- openstack-tox-py310-3 https://zuul.opendev.org/t/openstack/build/be4a360f883944adbb2024b926696113 : SUCCESS in 4m 39s\n- openstack-tox-py313-3 https://zuul.opendev.org/t/openstack/build/ccb9ca03358f4f44b14a96aa0cf06da7 : SUCCESS in 7m 55s\n- openstack-tox-py310-4 https://zuul.opendev.org/t/openstack/build/8c3e08623b874cdaa9ce45b24b41e478 : SUCCESS in 4m 57s\n- openstack-tox-py313-4 https://zuul.opendev.org/t/openstack/build/226e4c2dccda4a3db02eae269f6368e7 : SUCCESS in 7m 44s\n- openstack-tox-py310-5 https://zuul.opendev.org/t/openstack/build/f5d121593b3d4c9f9642439a8f43191a : SUCCESS in 4m 35s\n- openstack-tox-py313-5 https://zuul.opendev.org/t/openstack/build/50b880611d524405b1196cde0c55316b : SUCCESS in 7m 40s\n- openstack-tox-py310-6 https://zuul.opendev.org/t/openstack/build/ac30733a1a634538a1693e54bfe6fc61 : SUCCESS in 4m 31s\n- openstack-tox-py313-6 https://zuul.opendev.org/t/openstack/build/1c2c6e10081b49dab44aa1218848f7b9 : SUCCESS in 9m 17s","accounts_in_message":[],"_revision_number":3},{"id":"c7c750bd04dfba0f0cb5d5304d2a1ca3efffd916","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 16:20:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"601a6afdb57df9067b37ebb10bb34250e56ae4f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 16:38:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3f9c8f596994f5bab9e08d5ad7d0101\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0509072aa21b406e88c3c302a6dd6948 : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ebe7db3e86964702a592fbb13103fb05 : SUCCESS in 7m 54s\n- openstack-tox-py310-2 https://zuul.opendev.org/t/openstack/build/9ce9403f43444f83b4f3acff90950009 : SUCCESS in 3m 39s\n- openstack-tox-py313-2 https://zuul.opendev.org/t/openstack/build/5d2f6addb8764e329cd4df0439b265a9 : SUCCESS in 7m 24s\n- openstack-tox-py310-3 https://zuul.opendev.org/t/openstack/build/0536493ed66a481eb3cbac9d007238bd : SUCCESS in 5m 00s\n- openstack-tox-py313-3 https://zuul.opendev.org/t/openstack/build/573f58e7c2cc4e518c2f3f10bc47a4a8 : SUCCESS in 7m 49s\n- openstack-tox-py310-4 https://zuul.opendev.org/t/openstack/build/db979c218110460582aa69790276f348 : SUCCESS in 4m 26s\n- openstack-tox-py313-4 https://zuul.opendev.org/t/openstack/build/4c068e210b1144a4a509a3bf542ebfb2 : SUCCESS in 8m 22s\n- openstack-tox-py310-5 https://zuul.opendev.org/t/openstack/build/0b44f3b064624dd0834e63bb07f6b2c9 : SUCCESS in 5m 04s\n- openstack-tox-py313-5 https://zuul.opendev.org/t/openstack/build/ae3ba4a4b8054c8a9f38303d7adb6011 : SUCCESS in 7m 22s\n- openstack-tox-py310-6 https://zuul.opendev.org/t/openstack/build/c9d46d2d90774963bed56730dddc5e7b : SUCCESS in 4m 46s\n- openstack-tox-py313-6 https://zuul.opendev.org/t/openstack/build/466bafcf996547a4a571ec264ac8f3de : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":3},{"id":"07a3dcf863c910bae2ba851d509883430d539acc","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 16:38:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a53b84f8854435ad947a4c423d59a4098608190e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 16:52:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b82c017f4c794ffb8c1789ce8fdcf732\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fcf28ec385d64f639b7ffed077fd3d93 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/510adf07baeb427a9f01b5ca77784685 : SUCCESS in 6m 53s\n- openstack-tox-py310-2 https://zuul.opendev.org/t/openstack/build/f0d6402ed68e4cb3bfb68719b82e207f : SUCCESS in 3m 59s\n- openstack-tox-py313-2 https://zuul.opendev.org/t/openstack/build/e0aeb404aed94238a53419d9eb8bcdb2 : SUCCESS in 9m 29s\n- openstack-tox-py310-3 https://zuul.opendev.org/t/openstack/build/9a3ec2a8b7444577b403b27413d71e94 : SUCCESS in 3m 54s\n- openstack-tox-py313-3 https://zuul.opendev.org/t/openstack/build/3f1e6e6a0fad428cb3148c0ec8b6cf88 : SUCCESS in 7m 02s\n- openstack-tox-py310-4 https://zuul.opendev.org/t/openstack/build/e56ae3a6fc4144ae9fc5643921582668 : SUCCESS in 4m 00s\n- openstack-tox-py313-4 https://zuul.opendev.org/t/openstack/build/ce56811f1b4d44a594af71f89eedeac6 : SUCCESS in 7m 58s\n- openstack-tox-py310-5 https://zuul.opendev.org/t/openstack/build/69c8fd36b8bb480f8daad165dd8db40e : SUCCESS in 4m 06s\n- openstack-tox-py313-5 https://zuul.opendev.org/t/openstack/build/02e07f8f7caa49a7a9be6f6b340e72ba : SUCCESS in 9m 55s\n- openstack-tox-py310-6 https://zuul.opendev.org/t/openstack/build/b7788ade9b5a4684b2910f7a883c0495 : SUCCESS in 4m 25s\n- openstack-tox-py313-6 https://zuul.opendev.org/t/openstack/build/a2a5d410ad4d4e7bbef51effa343b2ac : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":3},{"id":"cdbe7a4e1649b8c25b9665648c23cfbfbfc6580f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 16:53:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"99ce7a7f602bfbed255a280eb46dd082a8e7c1ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 17:05:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7653c50ad92045e48ff0bda328856ae9\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be05e47b404a4e7f9585f1f2f299ebdf : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bfbcc9a5af0444b981d647cc22187973 : SUCCESS in 6m 58s\n- openstack-tox-py310-2 https://zuul.opendev.org/t/openstack/build/2e0f04bd3def4e53b5f509dfe21223f3 : SUCCESS in 4m 08s\n- openstack-tox-py313-2 https://zuul.opendev.org/t/openstack/build/48c3bb20ac8b46b983b5bc36c9ef65f1 : SUCCESS in 6m 44s\n- openstack-tox-py310-3 https://zuul.opendev.org/t/openstack/build/0f64e9706786479d869b984ba82bd394 : SUCCESS in 4m 18s\n- openstack-tox-py313-3 https://zuul.opendev.org/t/openstack/build/56d8966406b549518ee686979825e0b3 : SUCCESS in 4m 20s\n- openstack-tox-py310-4 https://zuul.opendev.org/t/openstack/build/0bbb1c5deb324124be18263e63011a08 : SUCCESS in 3m 44s\n- openstack-tox-py313-4 https://zuul.opendev.org/t/openstack/build/1bcae2f3daa546aebb502fbc2a6729e6 : SUCCESS in 7m 31s\n- openstack-tox-py310-5 https://zuul.opendev.org/t/openstack/build/73b80fd9610e4ff19f2ac04c438a0fcd : SUCCESS in 4m 34s\n- openstack-tox-py313-5 https://zuul.opendev.org/t/openstack/build/a20dfdb0c2924a89bf065479e984d76e : SUCCESS in 5m 03s\n- openstack-tox-py310-6 https://zuul.opendev.org/t/openstack/build/8d1d065f647043e2b046c245c121a77b : SUCCESS in 5m 10s\n- openstack-tox-py313-6 https://zuul.opendev.org/t/openstack/build/da70cb77d0b841a5a429d6fa367a8e4d : FAILURE in 4m 47s","accounts_in_message":[],"_revision_number":3},{"id":"5c625fa94103b663c5cde44bfb67f324445d0e72","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 17:11:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"adb3960ccc7a02ffda3c91b9bf319976b36e2f22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 17:28:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"10685be30848262ccaa9de1c31dec6b068a11c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 17:42:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cd5640e4c88481caa661b08cb3b188c\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddff6f3b7e524f3691c1e5007979362e : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e9c1d3cccc36474b98b38111ae927ca9 : SUCCESS in 9m 05s\n- openstack-tox-py310-2 https://zuul.opendev.org/t/openstack/build/ab6b94c0e55e4dbebd17037f5b1dc072 : SUCCESS in 4m 31s\n- openstack-tox-py313-2 https://zuul.opendev.org/t/openstack/build/7b9d5b7cd1d64173b1e1ff37286b5aca : SUCCESS in 7m 58s\n- openstack-tox-py310-3 https://zuul.opendev.org/t/openstack/build/beb7e5548e424563b6c0deb5e243e01a : SUCCESS in 4m 39s\n- openstack-tox-py313-3 https://zuul.opendev.org/t/openstack/build/71b46ab781364a3a87749afef9b00c3b : SUCCESS in 4m 20s\n- openstack-tox-py310-4 https://zuul.opendev.org/t/openstack/build/74af34c43ba1464f8939668b4bd6d044 : SUCCESS in 4m 20s\n- openstack-tox-py313-4 https://zuul.opendev.org/t/openstack/build/a105350c6532494893fab6bbaa541644 : SUCCESS in 7m 15s\n- openstack-tox-py310-5 https://zuul.opendev.org/t/openstack/build/7b0cba54fd1f4e0982caab1558a06845 : SUCCESS in 4m 26s\n- openstack-tox-py313-5 https://zuul.opendev.org/t/openstack/build/da22c760979d4326a812a233d9e08219 : SUCCESS in 8m 16s\n- openstack-tox-py310-6 https://zuul.opendev.org/t/openstack/build/f4bc16e5d3f64a8f84d698f2fa925b41 : SUCCESS in 4m 30s\n- openstack-tox-py313-6 https://zuul.opendev.org/t/openstack/build/0222adecf80c4a44a0e4d663ff947bf9 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":4},{"id":"d732127efc309b9ea38ce2f55a339c67268b9a78","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 17:43:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0aad7b69ce114b156e38ff0bb1b9ec78ff421023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 17:57:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8435a1e600624589a2b952ed49ea9f72\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0325619edaa548fca2728439bb323629 : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95d91c9d9fd84ccc970935ca8840984d : SUCCESS in 7m 35s\n- openstack-tox-py310-2 https://zuul.opendev.org/t/openstack/build/0c92caad93be411d9b3394d52d5a559f : SUCCESS in 4m 01s\n- openstack-tox-py313-2 https://zuul.opendev.org/t/openstack/build/11f73047282545798c43463bbb85c8f2 : SUCCESS in 6m 55s\n- openstack-tox-py310-3 https://zuul.opendev.org/t/openstack/build/3c5d9e0f9ba94fe3ad685956f38e59de : SUCCESS in 4m 24s\n- openstack-tox-py313-3 https://zuul.opendev.org/t/openstack/build/484d6f85e8cb4cf1943e10a26ddc6aed : SUCCESS in 7m 49s\n- openstack-tox-py310-4 https://zuul.opendev.org/t/openstack/build/ea1be9a3c3074c949169a686f33ebbfe : SUCCESS in 2m 45s\n- openstack-tox-py313-4 https://zuul.opendev.org/t/openstack/build/d4da52828a7f408d8a585bf4cbbaafdc : SUCCESS in 7m 51s\n- openstack-tox-py310-5 https://zuul.opendev.org/t/openstack/build/4990a37ad19f40ce9281ac874225ad11 : SUCCESS in 4m 20s\n- openstack-tox-py313-5 https://zuul.opendev.org/t/openstack/build/b1bc582f82ad4a4ca35baacf5811a799 : SUCCESS in 8m 15s\n- openstack-tox-py310-6 https://zuul.opendev.org/t/openstack/build/e63e54f30d194a17863b2b1311eda290 : SUCCESS in 4m 13s\n- openstack-tox-py313-6 https://zuul.opendev.org/t/openstack/build/934d4b8e5d984452b9da2dcf753c750c : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":4},{"id":"68beb6f2e28d4a6ee81a6016c5d52ad600f0eeab","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 18:25:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9dc26a3795687caf13dc141195c141d4d8cb786f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 18:37:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/455b47e790e5444dbd25d5023b74d8e0\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d681b0f06f543d6822a80e5f8e6475f : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a08705b5c8dd48be9b2f398511b04d9a : SUCCESS in 4m 28s\n- openstack-tox-py310-2 https://zuul.opendev.org/t/openstack/build/c469a2a6f0f643198966a272a8136a05 : SUCCESS in 4m 23s\n- openstack-tox-py313-2 https://zuul.opendev.org/t/openstack/build/3b67da79875340ab8406ea035592f163 : SUCCESS in 8m 31s\n- openstack-tox-py310-3 https://zuul.opendev.org/t/openstack/build/1fec86cf69074f1fb36ef775cbf545a8 : SUCCESS in 4m 11s\n- openstack-tox-py313-3 https://zuul.opendev.org/t/openstack/build/0520c4c2e22345eb8e574e1bb1195059 : SUCCESS in 6m 43s\n- openstack-tox-py310-4 https://zuul.opendev.org/t/openstack/build/69b87c4e0f2e4afcbda6c2ffbbeef87c : SUCCESS in 5m 04s\n- openstack-tox-py313-4 https://zuul.opendev.org/t/openstack/build/a57ca8bd573a40b286ee92010191d6e4 : SUCCESS in 8m 43s\n- openstack-tox-py310-5 https://zuul.opendev.org/t/openstack/build/178aed92a97f4f30ae5ccaf520f53a84 : SUCCESS in 4m 24s\n- openstack-tox-py313-5 https://zuul.opendev.org/t/openstack/build/8be7db86ef43400dacbd27c31485be05 : SUCCESS in 7m 13s\n- openstack-tox-py310-6 https://zuul.opendev.org/t/openstack/build/39dae01af86b48a9b46f60549a2b472c : SUCCESS in 4m 12s\n- openstack-tox-py313-6 https://zuul.opendev.org/t/openstack/build/329bc3887f124024b7d75d44615cff9a : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":4},{"id":"125244f47101abb0bb676491cbfb58b51ab868f9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 19:03:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"904f603b40bc3edbbffb9f59f7dbd43510d4dfa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 19:15:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b0353a6ed8e43038a4fd79c52e61ba1\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c352e0178f4c48f1ad29d7ff5ee21135 : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4f99683877c47bbab15f738d98bc35c : SUCCESS in 7m 02s\n- openstack-tox-py310-2 https://zuul.opendev.org/t/openstack/build/eb15d47f5d124fac92a83470fc833938 : SUCCESS in 4m 47s\n- openstack-tox-py313-2 https://zuul.opendev.org/t/openstack/build/224f35ce9a1a4fd68ec76519b3f5b3a3 : SUCCESS in 6m 49s\n- openstack-tox-py310-3 https://zuul.opendev.org/t/openstack/build/cdbed479cb0b4abb85aa312f2132b035 : SUCCESS in 2m 49s\n- openstack-tox-py313-3 https://zuul.opendev.org/t/openstack/build/f14bd1117d4943cf80561540ef17ae66 : SUCCESS in 6m 43s\n- openstack-tox-py310-4 https://zuul.opendev.org/t/openstack/build/f255d603de1649509fa49f93ea0aef53 : SUCCESS in 4m 18s\n- openstack-tox-py313-4 https://zuul.opendev.org/t/openstack/build/429ce7e3222b4614abaa01062cfd4ae4 : SUCCESS in 8m 39s\n- openstack-tox-py310-5 https://zuul.opendev.org/t/openstack/build/90045a8759324b8bb75d337c52e53b83 : SUCCESS in 4m 22s\n- openstack-tox-py313-5 https://zuul.opendev.org/t/openstack/build/db1c4ffd6625434fb5e7a69d98c0f166 : SUCCESS in 7m 36s\n- openstack-tox-py310-6 https://zuul.opendev.org/t/openstack/build/19087037635c4956b15638c57bda4d29 : SUCCESS in 4m 24s\n- openstack-tox-py313-6 https://zuul.opendev.org/t/openstack/build/22c00323d1a0420d8d85b56b0529f783 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":4},{"id":"778ccb347d3ce84386665e84098c7742b5d390a2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 19:22:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f6ff281b61e328a5661aad468519103bbef4e7a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 19:32:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a3c4769edc44b6caabb8362e3571314\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17e2b9a187f243acb4f16cf884e7bc64 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/00f19420fc59483a836e463580c79422 : SUCCESS in 7m 15s\n- openstack-tox-py310-2 https://zuul.opendev.org/t/openstack/build/d66ad5668a334b978e0a75a42f0ab660 : SUCCESS in 3m 49s\n- openstack-tox-py313-2 https://zuul.opendev.org/t/openstack/build/34ad174b3aef46f58718d3a4799dd8f3 : SUCCESS in 7m 34s\n- openstack-tox-py310-3 https://zuul.opendev.org/t/openstack/build/a83943e852c747d78bb89307d66843f6 : SUCCESS in 4m 01s\n- openstack-tox-py313-3 https://zuul.opendev.org/t/openstack/build/a0c7602971f94551aa19904bd6b1130c : SUCCESS in 7m 32s\n- openstack-tox-py310-4 https://zuul.opendev.org/t/openstack/build/916fe78818b04d85a99672d92f92822f : SUCCESS in 4m 06s\n- openstack-tox-py313-4 https://zuul.opendev.org/t/openstack/build/f72f53a941164328be205c0c54ccb7ae : SUCCESS in 6m 41s\n- openstack-tox-py310-5 https://zuul.opendev.org/t/openstack/build/5e997e300e2d4984958fe981d2d4ac72 : SUCCESS in 3m 15s\n- openstack-tox-py313-5 https://zuul.opendev.org/t/openstack/build/42951bcd426a438ba3521a2243979451 : SUCCESS in 7m 10s\n- openstack-tox-py310-6 https://zuul.opendev.org/t/openstack/build/1f76c7703e324552b016ed04a8486621 : SUCCESS in 3m 12s\n- openstack-tox-py313-6 https://zuul.opendev.org/t/openstack/build/a99a87add9ff4422ad2bd1ead23559d4 : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":4},{"id":"3a7a42b7595694dc5070076054d55b302d908cf2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-17 19:54:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"144f5cba8bb86bf2eab28fc9bca10d37fc0b8576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 20:06:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3d81b19567c4cdfa5df24f6211e4a39\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e366422d5fae46b488ffbc7a47fe7637 : SUCCESS in 5m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d8513a8fcc4476887656d49d9522059 : SUCCESS in 9m 16s\n- openstack-tox-py310-2 https://zuul.opendev.org/t/openstack/build/4b08101c7254401b83bdaae0806b3e96 : SUCCESS in 3m 31s\n- openstack-tox-py313-2 https://zuul.opendev.org/t/openstack/build/9e08f5c95b9540b0beefd918c262f379 : SUCCESS in 9m 12s\n- openstack-tox-py310-3 https://zuul.opendev.org/t/openstack/build/105c953dcda743c69c000e9348e701e0 : SUCCESS in 3m 03s\n- openstack-tox-py313-3 https://zuul.opendev.org/t/openstack/build/ddf1e3c80c96470daf46b5ba9eb7f593 : SUCCESS in 8m 58s\n- openstack-tox-py310-4 https://zuul.opendev.org/t/openstack/build/34bdc82b9ed7436ca21592c3db385a9e : SUCCESS in 2m 58s\n- openstack-tox-py313-4 https://zuul.opendev.org/t/openstack/build/ca5bdcfbb72047439ba0af3c25e9e919 : SUCCESS in 7m 16s\n- openstack-tox-py310-5 https://zuul.opendev.org/t/openstack/build/67e9097675ed4b56bcfc703e3ab76d87 : SUCCESS in 4m 22s\n- openstack-tox-py313-5 https://zuul.opendev.org/t/openstack/build/b2b0922e00fb449c853d2a0403afea84 : SUCCESS in 4m 11s\n- openstack-tox-py310-6 https://zuul.opendev.org/t/openstack/build/5a87dab200b442e39cbb99bb55fdac27 : SUCCESS in 4m 17s\n- openstack-tox-py313-6 https://zuul.opendev.org/t/openstack/build/f2b5ad7014544a3a901245eca41db11d : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":4},{"id":"4bc4efdfd5d13bddaa76b910ffdb65f99795eaeb","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-18 06:21:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7246985be54a766a5ea1b231af07bafa3f26e855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 06:38:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54c749d5eb3047c3986b2488129e69bf\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4caaa4fd1f014b688d751b133e78aece : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53babb357bd145b2b71950197074789b : SUCCESS in 5m 08s\n- openstack-tox-py310-2 https://zuul.opendev.org/t/openstack/build/6e5e8731f39345f4897f47b2563e5b9f : SUCCESS in 4m 49s\n- openstack-tox-py313-2 https://zuul.opendev.org/t/openstack/build/b8ab8fbf4c284d8fba1f982be9b489a8 : SUCCESS in 6m 49s\n- openstack-tox-py310-3 https://zuul.opendev.org/t/openstack/build/89d4a8c88ddf4e429059322ee8019a5f : SUCCESS in 7m 52s\n- openstack-tox-py313-3 https://zuul.opendev.org/t/openstack/build/5652cc5e546f4a93936f9e2782350ea1 : SUCCESS in 6m 44s\n- openstack-tox-py310-4 https://zuul.opendev.org/t/openstack/build/95a840d6e0194d49a11a010d70b412ca : SUCCESS in 4m 23s\n- openstack-tox-py313-4 https://zuul.opendev.org/t/openstack/build/b0a6356aacc34b7ebfd1e60f056db7cb : SUCCESS in 7m 07s\n- openstack-tox-py310-5 https://zuul.opendev.org/t/openstack/build/e29b2fee85484fd29ac707a1dfdb2785 : SUCCESS in 4m 31s\n- openstack-tox-py313-5 https://zuul.opendev.org/t/openstack/build/785120832be04e08b262f86f8839f3b4 : SUCCESS in 7m 07s\n- openstack-tox-py310-6 https://zuul.opendev.org/t/openstack/build/84851e48eab24466808a304ff6872532 : SUCCESS in 5m 00s\n- openstack-tox-py313-6 https://zuul.opendev.org/t/openstack/build/4e84ae5c3ab8453baca66fb1a02cb049 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":4},{"id":"10b4d18b8f108e1135dc0eac461f1ba6ed875c00","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-18 06:57:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"11ec76add5bb3a5a4d0a881ea4907910aec6fb09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 07:15:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a81b9b1ae13478fbe184c0f079aee63\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5c39c03980f4351b8e1a619e0170135 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a87625ce420845cea78e4d9a6392fbeb : SUCCESS in 6m 45s\n- openstack-tox-py310-2 https://zuul.opendev.org/t/openstack/build/19bc4df0ee8d419eab9cee8312c8b320 : SUCCESS in 6m 16s\n- openstack-tox-py313-2 https://zuul.opendev.org/t/openstack/build/260aa8b0768a4c869c1485b6c0b9ee44 : SUCCESS in 6m 46s\n- openstack-tox-py310-3 https://zuul.opendev.org/t/openstack/build/873858d7a6ac4406b1360c340ecf8e99 : SUCCESS in 4m 20s\n- openstack-tox-py313-3 https://zuul.opendev.org/t/openstack/build/3095bbdacbdf458cb8eb1e2e2908a754 : SUCCESS in 4m 16s\n- openstack-tox-py310-4 https://zuul.opendev.org/t/openstack/build/9a5d293c6f114a1498f1ea325159342a : SUCCESS in 3m 58s\n- openstack-tox-py313-4 https://zuul.opendev.org/t/openstack/build/227f1b4f85ee4e1cbc727c137f61a65b : SUCCESS in 7m 13s\n- openstack-tox-py310-5 https://zuul.opendev.org/t/openstack/build/dfaf6b87e3f141879e6b79fabfe2b7d9 : SUCCESS in 2m 58s\n- openstack-tox-py313-5 https://zuul.opendev.org/t/openstack/build/c39690dfaef0421a9576243e436171f2 : SUCCESS in 7m 36s\n- openstack-tox-py310-6 https://zuul.opendev.org/t/openstack/build/e2b79caec4f04c60b79ce5d059a02512 : SUCCESS in 5m 29s\n- openstack-tox-py313-6 https://zuul.opendev.org/t/openstack/build/698362e65413410fbd5736c89ba64fb3 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":4},{"id":"ddff7eab8f634465e5ebdc443fe31ebe49bd885c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-18 11:57:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a371920a1944c9e2dbc249b4646cc7d2c19be790","revisions":{"69034b88a1bf534e1055b863a795bb8b58a52b98":{"kind":"REWORK","_number":1,"created":"2025-12-17 13:58:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/23/971323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/971323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/971323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/971323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/971323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/971323/1"}}},"commit":{"parents":[{"commit":"f42860a7302b521dd2895a22535dbd1de1f912f0","subject":"Remove url tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f42860a7302b521dd2895a22535dbd1de1f912f0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-12-17 13:56:36.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-12-17 13:57:23.000000000","tz":60},"subject":"DNM/WIP Testing TestHAproxyVRRPCheckCMD.test_health_check","message":"DNM/WIP Testing TestHAproxyVRRPCheckCMD.test_health_check\n\nChange-Id: I855a5a147573bb6ba6ba92958e02a71d9ff432b0\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69034b88a1bf534e1055b863a795bb8b58a52b98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69034b88a1bf534e1055b863a795bb8b58a52b98"}]},"branch":"refs/heads/master"},"67d98961ca25d2a7cf29d93c9e751172bff71380":{"kind":"REWORK","_number":2,"created":"2025-12-17 14:55:49.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/23/971323/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/971323/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/971323/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/971323/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/971323/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/971323/2"}}},"commit":{"parents":[{"commit":"f42860a7302b521dd2895a22535dbd1de1f912f0","subject":"Remove url tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f42860a7302b521dd2895a22535dbd1de1f912f0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-12-17 13:56:36.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-12-17 14:55:02.000000000","tz":60},"subject":"DNM/WIP Testing TestHAproxyVRRPCheckCMD.test_health_check","message":"DNM/WIP Testing TestHAproxyVRRPCheckCMD.test_health_check\n\nChange-Id: I855a5a147573bb6ba6ba92958e02a71d9ff432b0\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/67d98961ca25d2a7cf29d93c9e751172bff71380"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/67d98961ca25d2a7cf29d93c9e751172bff71380"}]},"branch":"refs/heads/master"},"45ffe249c4dacc35a22a33be82d0655da455b966":{"kind":"REWORK","_number":3,"created":"2025-12-17 15:07:45.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/23/971323/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/971323/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/971323/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/971323/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/971323/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/971323/3"}}},"commit":{"parents":[{"commit":"f42860a7302b521dd2895a22535dbd1de1f912f0","subject":"Remove url tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f42860a7302b521dd2895a22535dbd1de1f912f0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-12-17 13:56:36.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-12-17 15:06:56.000000000","tz":60},"subject":"DNM/WIP Testing TestHAproxyVRRPCheckCMD.test_health_check","message":"DNM/WIP Testing TestHAproxyVRRPCheckCMD.test_health_check\n\nChange-Id: I855a5a147573bb6ba6ba92958e02a71d9ff432b0\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45ffe249c4dacc35a22a33be82d0655da455b966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45ffe249c4dacc35a22a33be82d0655da455b966"}]},"branch":"refs/heads/master"},"a371920a1944c9e2dbc249b4646cc7d2c19be790":{"kind":"REWORK","_number":4,"created":"2025-12-17 17:28:13.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/23/971323/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/971323/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/971323/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/971323/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/971323/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/971323/4"}}},"commit":{"parents":[{"commit":"f42860a7302b521dd2895a22535dbd1de1f912f0","subject":"Remove url tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f42860a7302b521dd2895a22535dbd1de1f912f0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-12-17 13:56:36.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-12-17 17:28:01.000000000","tz":60},"subject":"DNM/WIP Testing TestHAproxyVRRPCheckCMD.test_health_check","message":"DNM/WIP Testing TestHAproxyVRRPCheckCMD.test_health_check\n\nChange-Id: I855a5a147573bb6ba6ba92958e02a71d9ff432b0\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a371920a1944c9e2dbc249b4646cc7d2c19be790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a371920a1944c9e2dbc249b4646cc7d2c19be790"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
