)]}'
{"id":"openstack%2Foctavia~971408","triplet_id":"openstack%2Foctavia~master~Ib80e7b295827bd8521dcf4960cf8e825ab12314b","project":"openstack/octavia","branch":"master","topic":"bug_2133752_VRRP_and_HAProxy_peer_port_open","attention_set":{},"removed_from_attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-08-24 09:33:15.000000000","reason":"Change was submitted"},"38360":{"account":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"last_update":"2026-08-24 09:33:15.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-08-24 09:33:15.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-08-13 11:58:45.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Ib80e7b295827bd8521dcf4960cf8e825ab12314b","subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","status":"MERGED","created":"2025-12-18 12:03:44.000000000","updated":"2026-08-24 09:35:12.000000000","submitted":"2026-08-24 09:33:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":60,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971408-bug_2133752_VRRP_and_HAProxy_peer_port_open","meta_rev_id":"af00d23c932f1d129c8b16d9d7a18cfa53eae070","_number":971408,"virtual_id_number":971408,"owner":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-24 09:33:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-08-13 12:31:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-13 11:58:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-24 08:06:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"date":"2026-08-13 13:11:01.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2026-08-13 11:58:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-18 12:26:04.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2025-12-18 14:43:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-14 22:19:40.000000000","updated_by":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"reviewer":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"state":"CC"},{"updated":"2026-01-28 09:27:26.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-02-11 22:50:06.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2026-07-29 21:20:36.000000000","updated_by":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"reviewer":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"state":"REVIEWER"}],"messages":[{"id":"8647fbbd53a47d2df0ae8d5383cd660870bde429","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2025-12-18 12:03:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8df05810c4b3d8ac0d8426b378f7ca5092874787","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2025-12-18 12:04:51.000000000","message":"Topic set to WIP Fix VRRP and HAProxy peer ports open to 0.0.0.0/0 | https://review.opendev.org/c/openstack/octavia/+/971408","accounts_in_message":[],"_revision_number":1},{"id":"e7c201272a724f7cb5421fffc47a5a162dc2855c","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2025-12-18 12:05:43.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"bda859fcecad88c959d0da62abcc6c464ad279be","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2025-12-18 12:06:03.000000000","message":"Topic WIP Fix VRRP and HAProxy peer ports open to 0.0.0.0/0 | https://review.opendev.org/c/openstack/octavia/+/971408 removed","accounts_in_message":[],"_revision_number":1},{"id":"c28a92aaf7cc07a7da3ca5584cabbe517c627c29","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-12-18 12:26:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c552373334394316fe6bf4912e857edc9fc19e9d","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2025-12-18 13:00:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b011691a2ef4eb33c42da07de5fe9b6828f7ee9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 14:43:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94e836dab13646f08cd2fb745d441dc2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85984af97a7145af85fb72014aa2a161 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9973e05ce9944eedb2288400d1981ed3 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1843d016ce74d8ebe97e67d5a8540b6 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c6e1ef8ac17458a9edea8727eff927e : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/747e8703e4504f57af708be9b9e650a9 : SUCCESS in 6m 04s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/da760d6664424e4c861f72e531150145 : SUCCESS in 3m 56s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/17b622ffccd941c4915f15f40d65e7d7 : SUCCESS in 7m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4bdfca757ebd48a5b1af54a03c6f6d98 : SUCCESS in 2m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6bfc719b29664e58a38457b73d10a070 : SUCCESS in 5m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/559df7006a6440d9a80223dad076d61c : SUCCESS in 8m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/89ef045830c04e4ca49ace188a47cc4e : SUCCESS in 1h 04m 58s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/788b2ecc10e04a8a8d32435c321f62d0 : SUCCESS in 1h 33m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f267e3908b094106862ea334dbbe51a3 : SUCCESS in 1h 09m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2cc126c2869d4e06a63e109141cb984b : SUCCESS in 1h 40m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c111476b203b400fb7443ca6827c5f5e : SUCCESS in 1h 07m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fd83e7342d604921a3e9c84aae6a5c92 : SUCCESS in 37m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/48879104a2b24c0c88d86d821b1c6099 : SUCCESS in 43m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2dd24b59c7d34827b49a33431324c912 : SUCCESS in 32m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6ede6ba8b64b46f299c8183105f4469f : SUCCESS in 35m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c5c548fae8fc4655ac6e41a2adfbe16b : SUCCESS in 2h 37m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b2db4525c4544c718d9e49378408c9b8 : SUCCESS in 1h 34m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9da539536b7fa757d426eb7e6d5f47724d298adc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-01-12 09:17:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c03aa9a7c92648136c46979896548b63358311be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 12:11:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac6352a7184240f0acfedeca3a033efb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d6305f81e7b74584ba6031dc4e314de4 : SUCCESS in 2m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c0fd8ef94fd413d89ad6bcb5078d0ec : SUCCESS in 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed6177afa00f4096951ebf062a3265d6 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d62af715b694415bd8f416e3b8c4e5d : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f09e34b5edec4074a28bc2a12598259b : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e2cd8ab859f44958e669e25fd0b4e0b : SUCCESS in 7m 01s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/eb088923b084407f859ff97a58b79ac0 : SUCCESS in 4m 55s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/56f7db1c6bfa4327b69b7d6fbd294bbf : SUCCESS in 7m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b693fed8f4424f5181f91e0a7914350a : SUCCESS in 3m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c44a1654eb8640ad942bc8e0fb57a1bc : SUCCESS in 7m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dd500f550ce0434fadfea68c5ef7a001 : SUCCESS in 8m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d4fa755b34fa498a97c7f27ac3933506 : SUCCESS in 1h 44m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/93cd4b6842324be08db2f2d8619cba18 : SUCCESS in 1h 34m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f549a452034645ee863bd6ba9d0665ce : SUCCESS in 1h 00m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f7c6b46b2df44bf286452945a1ac8d3d : SUCCESS in 1h 36m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f79eff1e1dc143628e169290a7f9b7d9 : SUCCESS in 1h 05m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/521d38ec223a48b0b2893f0d0889b3d8 : SUCCESS in 34m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9af5062d04504d7db9fb7494cbcc4585 : SUCCESS in 41m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3aeb40a04a554a56829c45ec3754dcd5 : SUCCESS in 32m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4fb164456db04939b3196ac01367e812 : SUCCESS in 36m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bd6ebfa8e2cb47ada039bc4b68095dd0 : SUCCESS in 2h 52m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5f9610fb302f4adc8869895ade1b6efe : SUCCESS in 1h 32m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f01154b8da79c766aacb45df9dfd3feb803c4065","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-01-13 12:02:53.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"803878357658354bf9ad1608dc1cbda8c5bb12d0","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-01-13 12:04:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e4855f9cad7f29aeaed71c967815c4ac0e0e3019","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"real_author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2026-01-14 21:58:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_38562\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}],"_revision_number":3},{"id":"f539838e0df8e7bafea970b83babaa5c37532456","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-01-14 22:19:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b139a9d4c6e72399cef66acab5a2c0d63509f2e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 00:39:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9f3f5d9ded64cbdb15a7a24d3b78d35\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7855f8edf03e45e7909fdf1aefc9e7f3 : SUCCESS in 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f37748536424dd5a63f2a3c80ee5f91 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7310fa4453f4f87ba6044306178d1f2 : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15edcdf8b67346c2ae9fb0d28b87c646 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/076071fb29fa427c9f97ab1e884260d1 : SUCCESS in 5m 55s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d1dfc6d9528543fda3eb4969846dc75a : SUCCESS in 2m 54s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/f23566ca5ffb4ad28e827e7eacc61a1b : SUCCESS in 7m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/aa1048f91a5743038ec58538017012ce : SUCCESS in 3m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0e556a7bdbf54067a065194609776b64 : SUCCESS in 7m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f7890aaf16c249b596ba2c8a2b5dc2a7 : SUCCESS in 9m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/44adab97aedb452aad560d58f6a0fb5f : SUCCESS in 1h 24m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2e0f7c3238b6436eb2de458a740eeb51 : SUCCESS in 1h 34m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4bc639bbcdb84c98833ed8271735ed37 : SUCCESS in 1h 02m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/dbe401229f864666889dacc08e0a0889 : SUCCESS in 1h 38m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ad42b3838cff4e2fb1e18c0ff632951b : SUCCESS in 1h 08m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5052d850702d48269fde97d763205bc5 : SUCCESS in 35m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fdc2f7f4da954c06b405a25f2295006f : SUCCESS in 43m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f41468571804b0e9a0cbe50387f1a9c : SUCCESS in 33m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/83cf81ce5bbd4dd287a9bed5e41515c8 : SUCCESS in 35m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/76de532a2d734607b17b46f5acc07a1a : SUCCESS in 2h 38m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2de9cda277264717b30ed2e1220b5dc8 : SUCCESS in 1h 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c29b6398eec08eb3483c04a5d06a75731ae7e6d6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-28 09:27:26.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d2f90fca79231952b7763257b3eed5cc25355b71","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2026-02-11 22:50:06.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c64d65770a8cb4fd4e70aa559230321a09861ad2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-04 11:42:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"bb96aa852f47752c32b07aa13f05407463472fcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-04 11:57:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f0ee1705ad11b390b8adca0395a6d2651a1c7a5d","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-04 13:30:25.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3e7937af4d4fbe53c0bbd5a0a8ff3f4a11d62eac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 15:04:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c5861f5fa5f4ab1b9fb1112881f38ca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aebd683d089f40f99a950c108afefd36 : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a30c6e936c9c48aebf0e3e208a184871 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00c20f6e2e5a4311a301f2f28d0b9f40 : SUCCESS in 6m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff2fb0909b4744a792e55c701d885258 : SUCCESS in 4m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0f2655faf48f411695147e5fcbe30fac : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aedda2601fc84bec8975c9a1c79d77c1 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58c8f5abb7cf4a139bd73b99eceb4f9c : SUCCESS in 5m 19s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/941f234f8b404896aca63d702af9c36c : SUCCESS in 5m 07s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/e991e2dcd7a84da7b2f4c6ff7ece4300 : SUCCESS in 8m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/39d9fc109f96427c9060a540f9b502b7 : SUCCESS in 2m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/93b2fec286394033ac40f875b1a7cfc5 : SUCCESS in 6m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ddd035ce7593492d92e3cb8e9c2e6621 : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/748d24dfaab74e2091d9362b59099b11 : SUCCESS in 1h 20m 38s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8dc1421118cb4d55bdd09465a429460b : SUCCESS in 1h 39m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e2f73ea8d5d540ffb7bce91ab549653f : SUCCESS in 1h 19m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/218e7cfefef54e7fa89f40282536777a : SUCCESS in 1h 42m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b93b9f6914664a7a8c2e59ee2d0357f0 : SUCCESS in 1h 16m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/56fa70de69084b43b3e0b813bc2bf955 : SUCCESS in 47m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/775cc00fb06248ab9a206e234c1d8347 : SUCCESS in 45m 01s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/6c5582847c3e495aae2e1b91e573dd33 : SUCCESS in 43m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ea1e93996a5426482c8e46f5e1390b8 : FAILURE in 57m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a9cfbdbfb32f499eac7e139c4541ea4a : FAILURE in 53m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a5e2a77584864fe5aef2cda404cfaf4e : TIMED_OUT in 3h 05m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b447d25739434b949847d1c403eab79f : SUCCESS in 1h 41m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/cbe01d3cddc74f658cff6a4cb025fdb0 : SUCCESS in 1h 29m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2d64a6e166d14401a7281638c238102d : SUCCESS in 1h 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"295749d027de82f7d1bbd5f1f17a384799abacfb","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-03-07 00:52:43.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"18c3a5d6736113d5686c9f9ae9c8e76e62b21b91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-09 11:18:23.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"e6382468c1abe0440f4c8d636293be8226178850","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-09 11:18:34.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2a1d9b88caa75819fc79298505fc03a5838f3048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-09 11:38:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"30a11ff8937750789c1bf1f488df1788d28db879","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-09 11:38:34.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7a571458e235507b4925eb951d0b122e6f473ba0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:56:33.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ddd6c8729c2485c8146ad9a15275f90\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58ef68fd0e9c4b8081c23eacad615cd4 : SUCCESS in 15m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/626f6de8144245a39b9f67bc99a4575f : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cceb0d9e69a14f99839429ec7839ba84 : SUCCESS in 5m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ecd297c65c34f77b4a02dee80de43b3 : SUCCESS in 6m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/43f4bc8b74b941298226757d51230870 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ac7c33b3ea941ce8d5bdb90e50b0784 : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3baec0f925c34d2499165e3be5db445f : SUCCESS in 4m 59s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/ba2f608e92fb4b85a4d9eb4556758dfa : SUCCESS in 3m 45s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/36063a15876444ad8d6bf6480715dba6 : SUCCESS in 4m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4711c161dd3440db888a6b654907a2c7 : SUCCESS in 4m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ed24835d227482c86d6407b6db57c00 : SUCCESS in 9m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/823b5075183946ddacc2570555026138 : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e1aaa6d3e8914122ba247df2f5b9e4b0 : SUCCESS in 1h 12m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f14af9a9f2b74c49ace38743ef0fded8 : SUCCESS in 1h 38m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/954b3d2228a842be88e8adf1e23e3a30 : SUCCESS in 1h 07m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e47b98b55dec4dcbbca8ffd9739fd6a2 : SUCCESS in 1h 38m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/72e3aea0754b407ea6312df555b4ad6a : SUCCESS in 1h 13m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c4ea34e8cbda492fb6f404e4b45b09c1 : SUCCESS in 34m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/237a68840d954cf7b66fa763250e93b3 : SUCCESS in 44m 10s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a8edea72a72541ea98fdc0fbf6933e76 : SUCCESS in 42m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ead448435fde405ca6e0fb9899687c49 : FAILURE in 39m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5099e062bc4446ea9a6a3436c44d4ea9 : FAILURE in 43m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6b5bf5e9afc742d08db3cc93a183307d : SUCCESS in 2h 55m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/86638b8c42774a468ab8d3d335fc45de : SUCCESS in 1h 41m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/07dadd7b938f4242b751b0e299c9a71a : SUCCESS in 1h 28m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ed49a1d4964e4e2f8d1c4404b9270ab1 : SUCCESS in 1h 06m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"123f25d30218a5a4deb1e1a77bc30f971849e510","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-12 09:29:09.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"bd4b4fa0aa7fcdb6639668593ae829be029a85e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 12:29:59.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea88385d13554dff98432eae02b8cfab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3734bbf5856042e2a7f17f3d05dda32c : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/976b7f94e97d47c694e7da7cfe007897 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c354e67afb5c4b3489a994808cc84248 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83f0deade8d848d8b8888ac4ce0a4cb7 : SUCCESS in 5m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7d363034baa7429d89269312737879fe : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/786d6d3adf4d4f3088775b0af99a9731 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82d28b9692a14b2f8907f77002390b55 : SUCCESS in 5m 21s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/e66a5894257843b1b29d0d89a411a993 : SUCCESS in 4m 38s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/0a43593d258e4caeafa34848ca8feddd : SUCCESS in 6m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6e19e5883ada44ac99c2c5760900bd1b : SUCCESS in 2m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0bbfa0c2204f4344a0f58a3cee7ea5f0 : SUCCESS in 8m 25s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6c9abfeafdd140fdb7fb3f410bf6649b : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a9621e75ca464092b3f9639c41d16740 : SUCCESS in 1h 21m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b6b98efbb9904a08b5e9bd9303055427 : SUCCESS in 1h 35m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/94ab152e8ad54b0f919d328ab1d9d755 : SUCCESS in 1h 17m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/251d07c3e9a44fe2a490319e8d3d25d5 : SUCCESS in 1h 47m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f3c7831f737e42bd9fb3b9f76916864b : FAILURE in 2h 10m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1bf7d145fdd84852a90cb1300177284e : SUCCESS in 39m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/34ddb64488c04f8f8296afc946f8c0b4 : SUCCESS in 43m 46s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c2fbab14646d494ea0d1fe72e26bb5e9 : SUCCESS in 43m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e925bbd7889c45f4bbb51207a3ef7e54 : FAILURE in 59m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a62514e76f264aa1bec5f99390f8567b : FAILURE in 50m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7be65d5566824f4096a83a590d424450 : SUCCESS in 2h 58m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1f2df58632b24c2dbb5403890713c9ba : SUCCESS in 1h 37m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/47e2e9f7197a46cdbcf243e4939346b6 : SUCCESS in 1h 35m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e931dea99114476fbff4f1367ff76ae0 : FAILURE in 1h 42m 39s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ca957fc318629b920dc138cde8611eebe8a54e2b","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-19 11:11:28.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"52ab375c3f4bc402387a82b67f37712951e8465a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 14:13:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ad1a4a730214a71aaa4432eed7491f5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a900ed48e96c44b48df5500e68b711c9 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d98f03de284c4214904da844f1093b86 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7e6e3d0a5274062928685f63b4599a8 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56eb998173ca451e85c8a5a698085018 : SUCCESS in 4m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b4be2cb585cf47a0a7a18a1a3b40e6a6 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86689fff3da3437e84a521669c1629d6 : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe066740d2914710b0bc3e73cf82a77f : SUCCESS in 3m 13s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/3c776cedf4324ea6a5727bb69f7ffa79 : SUCCESS in 2m 58s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/dbd90cdd173441f68c620ac65a4645ab : SUCCESS in 4m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b4615318e3784e8d9fffaa114a9000c4 : SUCCESS in 4m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/50f3ee575d7847babbbf88cfa4ea018e : SUCCESS in 4m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/187ce49d4505446eacaedbdc240c3a29 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da1b191e48bc45a885f4f9663551ffa4 : SUCCESS in 35m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4057188f28e84a628bca8b2b0528da56 : SUCCESS in 1h 44m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/102afee9f6564c79b32600048a624de3 : SUCCESS in 1h 16m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/20ac2bac376a40b6a0d9679637145093 : SUCCESS in 1h 51m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ba1de37f59da49ee80fe8deed5ad269b : SUCCESS in 1h 18m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b8bcf77deb754d7a8f00d364edb0ce1f : SUCCESS in 42m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3cdd95ecc7d343a39554434b57a71bf0 : SUCCESS in 44m 38s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/bdc058f979ce469dae13170468e344f4 : SUCCESS in 43m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d1f8acb3e6d4592be229b28f96bef59 : FAILURE in 39m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/352b779aa893455f9666dd92cbf7f7ce : FAILURE in 39m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c6563f1537f4484d991b7bc294ff110d : SUCCESS in 3h 00m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8d9cd922705e4f72a17587c3012ee952 : SUCCESS in 1h 33m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/f4e8075cbe2643f6902da6bc46612940 : SUCCESS in 1h 31m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/57f2c2cf937740049d774424d2bbccef : SUCCESS in 1h 20m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c690e0c29343d15014515f2c0ca65890dc6cecf5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-03-24 08:41:11.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f8d268147e31a4d2adf375031b82a4cb1eb8ad8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-06-17 13:39:56.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"0e728e53af9b2f8c676ba202e4e6a1ddf4a4fd9c","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-06-17 13:41:47.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"98b3201ef8861384420320b03c1d9cd2e3071888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 16:08:20.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33f27d434ec14bc291846019d33ed348\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63529a2699a94447aa10191de073dbd8 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/065b23df2d794642968a08f0b52e4f64 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a8803c75bb84d7498a846164dd80de0 : SUCCESS in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20015e7806d8402f85bae5594d1ca05a : SUCCESS in 3m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3c428e07af754f3c9ef2b2ee2097df59 : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f5deaf232a84158bc345dcf07aabc02 : SUCCESS in 13m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09fbadb6ec194013877c9e4dea53d48b : SUCCESS in 5m 59s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/b086b6f12ca24276b807fc5105c4433b : SUCCESS in 4m 18s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/599050303b6348ed8c25ad9657694530 : SUCCESS in 4m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/84c609c8a213402fbff672c4bff9f53c : SUCCESS in 3m 57s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/adc387773e2d4d9fa206526354058cda : SUCCESS in 5m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6fe331620e2e47e9a694a66b36265bf6 : SUCCESS in 37m 52s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/22a135ef758f4fc3a94604234bc0104e : SUCCESS in 1h 29m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/55b65d91bc7e4d7bb9802a15aad05265 : SUCCESS in 1h 05m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/652c910ec18045eab0574bf3ae080687 : SUCCESS in 1h 38m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ed5b1df7fe7345838ffb17659b9c85ce : SUCCESS in 1h 02m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/640c5c7be80744fe9f907c5f395cb108 : SUCCESS in 37m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e038710f439b4467931921e1b4bc6757 : SUCCESS in 47m 18s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f5515596107c4907910e24e511d32094 : SUCCESS in 46m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6bd44cee5a34e638db4a614e75b961f : SUCCESS in 38m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/12925e2d669340988f2342b920875dda : SUCCESS in 39m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e8ccbd8268c1488f93069a8134367a6a : SUCCESS in 2h 21m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ce9a046425d742bba7456049046e33dc : SUCCESS in 1h 37m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/b7a5dd7105ef4ecf80e9b35657f6d71c : SUCCESS in 1h 27m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/85844e7ea4854725b0e0b1ca7c4887c9 : SUCCESS in 1h 00m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e4cb83603e6def438c85649cdb70d9e8d4d6f3e2","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-06-18 20:44:23.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"6a5c4ad33ebce97b63767dc9aef775a66277943e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-06-25 12:45:58.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"a3158d91c0a909890ff64d3ea08d9261f4ee1ba8","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-06-25 12:47:30.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4f1c806c19e81f81b6582729996461d5014ea0dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 15:16:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e4fee36184b4406936d33dc8cf459f0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66bfe34889754ca3b8b815e6176c6bcd : SUCCESS in 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/670490de76b2456ead659200a52212ae : SUCCESS in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de247fef310c44bbac8f268752b29a74 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be3d46733a0b4042b5a4aa110d4a2162 : SUCCESS in 5m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/edaeb82732f0496e8373ff399814dec1 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2609e9fbcdd43f6ac7542e52bf3f0fc : SUCCESS in 13m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88704040a9464d6483d5e1f07d42e384 : SUCCESS in 7m 13s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/a135258e5c93433f867516ee75178dcb : SUCCESS in 4m 36s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/d426642fa09a474b8d517225faa00d0e : SUCCESS in 4m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1092cc2eb6424c258123ed30786327c3 : SUCCESS in 2m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/181d154e4d0c4336b48144f09707f0af : SUCCESS in 6m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d3b647468d149968ae2ccfeff42fab0 : SUCCESS in 52m 58s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a92c2dfca9134ac7a7a56d7e901d279e : SUCCESS in 1h 28m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ef7b2af043bf47158207b676c2ce6874 : SUCCESS in 1h 05m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/99d6d71538f442afb6fa8d46e5d0c8cb : SUCCESS in 1h 33m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/aea6ed3c44b54be292ae3ca339ab7f3a : SUCCESS in 1h 06m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8165d226955b427d925e5ab9c6344b53 : SUCCESS in 34m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6a172ef6537b44c8b5d76bfe52da4723 : SUCCESS in 51m 25s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/6a7de0fedf0242b8aedc2d35f736d273 : SUCCESS in 45m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bb2fddb8eef14589ae2f7a2e8e6736ea : SUCCESS in 33m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/03ac6da3080145348506ebffd9a77f08 : SUCCESS in 37m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/63f8ddc5ba8246ba854b5d6548a5b3bf : SUCCESS in 2h 28m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/10dd8823283c40c8a163a4845943c018 : SUCCESS in 1h 27m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/55922fb0f9994bd2ad5bec6ae6abd1c3 : SUCCESS in 1h 29m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d62d6d837f4b4d1e800978aabbce14d4 : SUCCESS in 1h 02m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6411a847e83d9f16ada2ceec3509d456d220dac8","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-07-29 21:20:36.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"644e3f914f56d9e314af589543f8bb5d0e9a4c96","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-31 09:37:47.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"995a1c8a8d97c8020d024efa0c961a7869cc7864","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-07-31 09:56:14.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ebf0ccd1e7951ed9bb54949b198f46ea522aa493","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-31 11:42:44.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1236726eb0258687b14f0fd28546c52453f7c38a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-07-31 13:46:03.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"c5c7f3daea3238a8c5ec66d9504ab721652638ba","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-31 13:54:19.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"a237ca660ad2be062877f75f70ffb559f4977220","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-07-31 14:07:10.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"0b2356d575ba77be0316a0a3a03f25b4659c946b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-07-31 14:09:40.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"8c485e4c010f68ab2b2f8ecc0af6c70aa6f9c7e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-07-31 14:27:44.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"a5f95972e044051fd8928f68f36a6379a3b253ad","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-07-31 14:29:52.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"47d4267c3d92efae519be69f751eb9174606e98c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-07-31 16:02:40.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"874a7b0f5ff13866e26bcdc692272c670d16ac9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 18:42:51.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59b904ffb72e4d7bade77a28b5396d33\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05e6cb2264af4724b4a67b7c9c03ada0 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2751e1e5c14d4bc49ece949a2f5fabda : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2d55aa9c2e84b818a0ebed7d71ef425 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0e4c84a0c884594aed70ae56ad4605e : SUCCESS in 5m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/addb6ec4877f4e1ab31c252cab75d435 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deebab8e3e114ed5b2faa5201c327619 : SUCCESS in 13m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ed2bde6431d4b16954592dfe8a50ff0 : SUCCESS in 9m 31s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/66ce18a4ad964d24b7f0ef78fb351771 : SUCCESS in 5m 12s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/2c0199b428a44de4bf1bbe877a14ce7a : SUCCESS in 5m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/df9ad6bfef504c9891f57ba2751c7b61 : SUCCESS in 5m 05s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6dbdc8bf99764ecb844f21bc54b596b7 : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/18598e4b58fd41968422f9f1b16da1f9 : SUCCESS in 1h 00m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/abc47397517c4a58a60ca5ae6dcf0502 : SUCCESS in 1h 24m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7d754e9f116747ae9b555e2f7321fd64 : SUCCESS in 1h 05m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fb3ef3ecdff743d39cfc9b9f9ef369c3 : SUCCESS in 1h 28m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/466477a190ef48dca48bac2478262076 : SUCCESS in 59m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c158480aacbc4c9391e35e1352aafe96 : SUCCESS in 36m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2997f56853404f558247bb1baf51f87e : SUCCESS in 46m 34s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/1afff823a7ef45fa84be42d474a098d4 : SUCCESS in 44m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d775ff8d66e4c9386a26528a8090644 : SUCCESS in 30m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7f448b8898774b8291bf56c3615c8364 : SUCCESS in 30m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b8647f8ba4b845b586eb1b9cf11b589e : SUCCESS in 2h 17m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d20867f8cba548f8a7b05089088103a5 : SUCCESS in 1h 22m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/345e2b6fc2584defbbbf039bc09fdfdb : TIMED_OUT in 2h 37m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bad0a4ba1a1d4805850f0f6c70f2261e : SUCCESS in 2h 31m 50s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"4596137ddd55a414e563f53dfd16b44786acf0a9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-03 08:21:02.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"008d96eea7063956cee10108fd83e16bdca728ef","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-03 15:07:41.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"4c70d514db2699af98a70fa3559c834271ca6d78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-04 08:37:51.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"642227ecc2cae6065826aa4b7a13aee12fa99cdc","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-04 08:38:51.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"4756d4690fe47fa8c8db773a29657449cfa0785a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-04 08:41:09.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"7e9b323fa6be6928042dbfe4d4b4c60887ddcf70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 11:17:43.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ad029834cfa44b79757dea5a9764834\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b19e93b175df423f92a21604ca7ebd52 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4e0c4a427fb4bcdb36d93284e3876b3 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3e760b2888d47f8aca12dda5f86cfc8 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b00ba1666dd24bd1a792479899846d33 : SUCCESS in 4m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e237931546d04c2b84d7377491961e64 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d52d1fb67d04bbd8a4b0da63db8977b : SUCCESS in 18m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/baa69162c294498d8cda5d4bba1f8aa8 : SUCCESS in 4m 48s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/4c92f1e645c84649b044eab5eaa1e158 : SUCCESS in 4m 55s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/3916d6e4ff8945f480ed59030acf81ed : SUCCESS in 5m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/952555ba2f5b44ee8106b6aeae5f9c22 : SUCCESS in 3m 35s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/02c2d618edaa491f93197f8456ea240c : SUCCESS in 5m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2ff004544ca04a1aae0be58de796dfdf : SUCCESS in 1h 02m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ef678c295ba943a7b36a8d597165b1f3 : SUCCESS in 1h 25m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/021f9d93c7214c49a4db05b23719df75 : SUCCESS in 1h 00m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/56162d3ff56c4fbbbe58e6ef1bc2dfcf : SUCCESS in 1h 34m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8bb2ce3ab3a048a9a6ffe9e95ae12663 : SUCCESS in 1h 12m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/68097bca5b5444de905cebe36bdbbb53 : SUCCESS in 34m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b144e3f878b34389ac5bdeff1b5fb7f9 : SUCCESS in 46m 09s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e8d3b837c2344b88b319d3a5c13508f0 : SUCCESS in 40m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c855c28507ff4705be2282acb739819e : SUCCESS in 30m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/39d079913a1f4ba1b02451ed40a0f7f5 : SUCCESS in 33m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b5d27fd7040a4a7da7804ea2cf1bdc28 : SUCCESS in 2h 04m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3ba6534ee00a4faa9a6d731f6c1449e5 : SUCCESS in 1h 18m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/ef477fb0105a4c098f303000a520a002 : TIMED_OUT in 2h 37m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/20e1d44a1ba64232b9a637e5f6a0b91e : SUCCESS in 2h 35m 18s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"1c8db9e8d2e23ca9d560a1b5ad07989dd969985c","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-04 13:24:14.000000000","message":"Patch Set 16:\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"422c38ce68d65cdc24c2809a8abd75d6c38ea14b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-05 16:12:46.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"2e191455a1b20ff204846637b54a18dc3e22bebf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-05 16:19:22.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"bcacb7ea34db7f637b9f7584e380afbf6f4cac31","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-05 16:20:06.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"e58dda9932d6b73c6f4e187d804dc610c00e7b67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 19:33:32.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78db236474ab497ba6cafd41ef9498f8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45ebf71943134e61aa2dc29033d9ed87 : POST_FAILURE in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5534ef9f594d4290981c7774f9b48e90 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22ddd0ea44594c08b4b25bfbd170cbc7 : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a6647e9dbb2843d0b6aaf52d11fe53b6 : SUCCESS in 7m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/40b33497d2bb405696eb46ba6d4e7164 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/391462d1018f4f379e428808c80a58ea : SUCCESS in 14m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/296fda07476c4c47a2f014e3f4d1167e : SUCCESS in 6m 53s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/807b1ce1181c463dbdb26761c94b7e4c : SUCCESS in 5m 29s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/7429ec70be274b43942565c91c73f168 : SUCCESS in 5m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0ec2c89556594e0188d43cdb74b749df : SUCCESS in 2m 36s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e333c0ebe53048f690fe7d4debd60b76 : SUCCESS in 5m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0c58956be363496e9022c04b23d7b2e1 : SUCCESS in 44m 55s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c38052c618004f95b8faa2ca3f973224 : SUCCESS in 1h 21m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cbc06266a1e84d10bdee4eedbb22e928 : SUCCESS in 1h 08m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c50445b56c594377af8f75cbab6ebd07 : SUCCESS in 1h 28m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c9ca4aa44a5349b1acf1903d30526b6a : SUCCESS in 1h 05m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/07b2557e00474fa780382039ed1ca74c : SUCCESS in 34m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a5ecb69802d148b6a2aef4f9521fc553 : SUCCESS in 46m 00s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/3d91dcd84e9448a99c7f02fed3967283 : SUCCESS in 57m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a0aee2240a3a436ab3629fa410b4867b : SUCCESS in 32m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ba4e8969b1f84d1bbf9c16f6b5b9e4e2 : SUCCESS in 49m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dc8eaec7980443128b3e464cfdf97ae3 : TIMED_OUT in 3h 05m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/93a98c6b2466422d98aa71067d6f33a2 : SUCCESS in 1h 22m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/ee41059462094d12b6fa9c2f3b631231 : TIMED_OUT in 2h 40m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3e78fe7e6c864224a86b9a52958648fc : TIMED_OUT in 2h 37m 28s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"6db9ac9118a65a436d4067911a28c1ba3bb41c81","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-06 07:56:30.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"8dce6c1a935ec3822fcf842d69a50bdf69f59aaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 10:50:59.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/655e32a085044ba9a0bef6de2c3c45db\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc6f4badd02a47da9d40f0fd5312a0cd : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17982ba4bada420c8d5f9761400a594e : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b13794c5c8574fd1946e1d2ed25d73cb : SUCCESS in 6m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9cb0ab26935640809490760fb664c7bf : SUCCESS in 6m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e04f648458b24b33b68eb3dc0fa02cbd : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a036feb9458242c7a7df7037a04cd25f : SUCCESS in 12m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f664ec5bc4d34e04a296ae05ad770b62 : SUCCESS in 5m 38s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/52ed6e5ecfee4724b33128a858035700 : SUCCESS in 6m 22s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/c4ce4ddf08074f449f34be2d1305f1b4 : SUCCESS in 6m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f97795fa526a4cc8b385ad13bbbaafe2 : SUCCESS in 5m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/165b6449060b45a4b67889d861774bf0 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/16a2106e63cc4fbab4458ef9fc67f571 : FAILURE in 28m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c3b24b42c27d49f4b99305bce5c1fea3 : SUCCESS in 1h 22m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/09ff900871d046869b73ea36397d0bd7 : FAILURE in 19m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/26e048ac56d54443864cbce0b22a6944 : SUCCESS in 1h 28m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/894cb05e05b54c56966fe3fb6a256516 : FAILURE in 31m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ee7980f1a2b48e3a30501a2a4584281 : SUCCESS in 46m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a97036ddfe804e1ca91feb239d7ed310 : FAILURE in 31m 26s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c97207324ae244a1912f3538d8ed1072 : SUCCESS in 40m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b33af955b0c4603a54653f6aa0dd298 : FAILURE in 31m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/91ec096989c14ed5bab01fe340ef348c : FAILURE in 19m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/149ffa30daa44f0baa57de2e9f104b43 : SUCCESS in 2h 52m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dd6225ead5df4db4b527a63c0573dd9b : SUCCESS in 1h 19m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/a12b24495b534b60b60c6897f6e303d2 : TIMED_OUT in 2h 36m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/507a93d79a2b4225afda3121b87e1076 : TIMED_OUT in 2h 36m 49s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"ed42dfd8e40da31533f4a9b4be5b6359a7c04316","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-07 10:17:03.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a230ce2812c2cb4f04b77bdafbabce0dabcb3d1b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-07 11:03:26.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"c9d62163cda686563c126c49c3f2e7b46702f6d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 12:56:29.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8a41b34b83a4cf7907c8a11ec2da953\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1f241c368c0456ba643ea86a7db32e5 : FAILURE in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a05a6dccc2543ebaec3139bd5862ff0 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cdaa5a02a0ba4b7d8c00751d68153621 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04e5e8c6e3734847bbd8d2dfa46adf5c : SUCCESS in 6m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6a75c2deac0047cb832d9a94b6a32345 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c67a5e3c3b946298a108f82a28718df : SUCCESS in 10m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e940c356de74234bbe1b75cb68e6204 : SUCCESS in 8m 44s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/78a85abb017642ef8fd698eb319a2087 : SUCCESS in 5m 09s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/2cab7ca3724b44eaba06dc4d2d5f6ebd : SUCCESS in 6m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1a753423e8d04df68bc9c5bd1e1249d5 : SUCCESS in 3m 50s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ef7e614594444c7aa947bfca684f3de2 : SUCCESS in 6m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8e66663212f84d37a18e749f3933788a : SUCCESS in 1h 03m 58s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b14d6c532ece4cd7973b140925b9e450 : SUCCESS in 1h 21m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ffca63a77ccf4741807232d6e2d7dc13 : SUCCESS in 56m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/dcf69cfb258d4a69a1c199dde31fe274 : SUCCESS in 1h 25m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f9cc9b2ea329485ca5d42296010da370 : SUCCESS in 59m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c518a6432bf6430b82313a80ffae814b : SUCCESS in 33m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6588513cd06d44d19aafbf7e3d33b7dc : SUCCESS in 47m 28s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8c59c2b77eca4e26a810ec6d58db2986 : SUCCESS in 41m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ae9ee74454941ef8de0c4483d2a4bbf : SUCCESS in 37m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/839ccf8758b84019967094d8e8fb510f : SUCCESS in 33m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c41a32cacaf044229b75e44858bb6fd1 : SUCCESS in 2h 28m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/19906afc367b4d8c9a91a763bb8e8035 : SUCCESS in 1h 24m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/5fded5f184cf4c5ca6078cb475470162 : TIMED_OUT in 2h 36m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9277ef0e675847a08b4638fc29a469f2 : SUCCESS in 2h 21m 33s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"888ee264dc5e640538a9badd90144072289d52ff","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-10 10:06:47.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"dd3da11f20ddd092799a5b65fb2e0554738a6180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 12:48:26.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a114bbbf3aff4cfe8e745d394256b296\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/280c46c845d249b9864698cf5892b737 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5afccf17014448c888df99e82ca6597e : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58176502ec5848d7a98c703f05322408 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6bb00fca583b47928b00f739d47a9a65 : SUCCESS in 4m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6d43fc1a664b46a5b73bb0ce6fe7e3fb : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cab62d45ede741ca8c6d2d4636fb9bd6 : SUCCESS in 13m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a88e29c936449a68f25df1554463d76 : SUCCESS in 5m 08s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/c946679662104b1cba049aee1baad9f0 : SUCCESS in 6m 17s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/198455bf7cbf4b7787a1c7c3abe54200 : SUCCESS in 5m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/68f0955cd52a407fb5c733294eba883e : SUCCESS in 4m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a35dd4bd8d484542a5931e023785dd7e : SUCCESS in 5m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1070064f9d8843e7868adf112d99596a : SUCCESS in 28m 21s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b8ca051cf9054e28a5b747915589965a : SUCCESS in 1h 20m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/18e601382c8a4c58889d1a3250de0f2e : FAILURE in 19m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8821da716dc24dc3a67bec755de1102d : SUCCESS in 1h 25m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2fd250b3284444509b9cf5eed20d3cf7 : SUCCESS in 1h 00m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/928b5a38377e4ef6a8860e33f99c9eb8 : SUCCESS in 35m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cba0b7f0e1ee49a5b866d733046b2532 : SUCCESS in 44m 00s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c4b8eb71dddb42cbbca4a90816823c4e : SUCCESS in 40m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/295999c137d948f8815ed29cfb520cfb : SUCCESS in 28m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8d2092c6fff4463eac61281228686f37 : SUCCESS in 35m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/31d7f870726c46648a7148b74fd18fda : SUCCESS in 2h 24m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4c5f60dc6efa478ea53dd15de85bab83 : SUCCESS in 1h 21m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/df57c807156f4518ac748ac22f7f443e : TIMED_OUT in 2h 36m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9a4bfa62c96c402f9b5258bfaabb7ca7 : SUCCESS in 2h 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"1e7c6c1952bce8c7e8a55340e7022a8242246242","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-11 07:52:53.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"b180fcec88528ec9eb6005aedfeabd76fd85eed5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 10:21:07.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/167040619bd341c68aae4c421609ee2b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f428d8192443427f9345bbd438ce10f1 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8872c48a7601404595e1d94553d99b93 : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e34f770caa2844d6b5e5faefea328d8d : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60d25e1bcb72497f9d58685aadeb044c : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/55c2893b01834de0a43f1e1bd804c46c : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fabbc77478414e9c9547bf0a070f2d08 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e813b7fa46d4504b28dc05e3d4f2b07 : SUCCESS in 9m 13s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/164ea431bb0b43dbbe522f2a945a5d07 : SUCCESS in 3m 09s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/8c3e3ca4f25042658089495149d01bd6 : SUCCESS in 6m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f66592fd5e8c4f1583e1c6f15701df97 : SUCCESS in 3m 19s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/aa02cff453d74be4b158490f836d2212 : SUCCESS in 4m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/146da952856c49c9ad0d41f5748d5d3f : SUCCESS in 58m 55s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7b491a74ab234c5f9bbcae84284e2a4a : SUCCESS in 1h 21m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/22e91a9cff5548d0a820fec856dc4d28 : SUCCESS in 1h 01m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/21e32ff8402641f1a106633dd59a16be : SUCCESS in 1h 27m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5be9f14cc55d4df4ad65e89ffa402866 : SUCCESS in 1h 07m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e3eba91da0184043ad9e961a4328c756 : SUCCESS in 35m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2ab0bb5a123141ffb39827b04aea057a : SUCCESS in 48m 20s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/767aae8b0c984d8dba529d11a779ffe8 : SUCCESS in 41m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd1d6f979c3e4161b357499ebb23a1dd : SUCCESS in 31m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/63313da2a8204120ab7e09c9e46da676 : SUCCESS in 31m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e0ffd2cbc0e4411f80a69b8567a91477 : SUCCESS in 2h 16m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a98695e99df04883a6e5586fba74f405 : SUCCESS in 1h 19m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/1db4ec1c0c864ffeb38771e8f35deaac : SUCCESS in 2h 25m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3f8e2b59c1c14d5aa6960184ebcb36d4 : SUCCESS in 2h 01m 59s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"18a8518ed73baecec30efa69c5a932cbb6a77891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-13 09:16:17.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"8be8f1738f3fce1c84f0a50e9983914a757de811","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-13 09:40:02.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"8e96fa1bc14ab8d2025735474b6f1de01b6c99f7","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-13 10:02:39.000000000","message":"Patch Set 19: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"7d8c720f01a466ec86d6df8103a7c7c476210c27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-13 10:20:36.000000000","message":"Uploaded patch set 20: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":20},{"id":"39ca8be54b4ed905f6c74900b741ba23c078e2a7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-13 11:58:45.000000000","message":"Patch Set 20: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"3ab9b6c2716f9de84b440650e8a684584ad64d86","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-13 12:31:17.000000000","message":"Patch Set 20: -Backport-Candidate Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"14b972fc7e659e76f2c903614bd2bf8aa2f53e4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 12:44:20.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92c659b788834548a440c99594041337\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90fafbbdb01c40a2a01db382514db866 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d94e05fc77b4b6aaf156f5a2bee222d : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77c123f096ef422b9121a0874e7a5943 : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b71be0fe61004335930d6190aaa361bf : SUCCESS in 4m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b1bd763b2b3244089ba6d10477cbdb85 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fe8ae19a707410ea61df233d8ae5b5e : SUCCESS in 13m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c934c23c4c547e6b53ed26b8e67c90f : SUCCESS in 7m 42s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/df669a92f4ec4461a8aeafa9bac7bbeb : FAILURE in 3m 21s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/6eb8d4a464cd4baca52c741ea5a6300d : FAILURE in 2m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6282176c543043059319362b1202910b : SUCCESS in 2m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/39a595d2ce914c9e8e24ff0f4c46e554 : SUCCESS in 5m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/843b884bd19f46408696429e820aa3c9 : SUCCESS in 1h 05m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1c7bb96d4e4d48a38088aa560a09e821 : FAILURE in 9m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c720a22596a64114b69fc6146478191c : FAILURE in 9m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ca5c0da3cee84de5bb929aad8f1548bd : FAILURE in 9m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2f209383455e4eb2a7f6e93b2cb73b89 : FAILURE in 9m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e13766fc6324f6583e6e306c91dfdad : FAILURE in 9m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b6ff378408a04f398e7e23a85ee4110b : FAILURE in 9m 40s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/3ec2d192c9da4447a1f410df677819cb : FAILURE in 33m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd40445069cb4851ab810ac6ff8344b8 : FAILURE in 8m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2768422a169145e38a8e159b24504a80 : FAILURE in 9m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f6ec64293cad4d0a8559c239efc64347 : FAILURE in 11m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7bd055c8da0447a38ea187981e48c15e : FAILURE in 8m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/212c2b28d0af40838b1a2820da2b27c1 : FAILURE in 47m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2f0dd353a303482584981edc3ac927bd : FAILURE in 2h 21m 28s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"8e30edc947c5ae375d1638a21a220364f9d0f2da","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-13 13:11:01.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"119f67d6789ade667d81e6d0ef8bff4682ddcf34","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-14 10:07:06.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"cb7daa48addef5a88c829e04aa9ff175d484fd65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 12:22:35.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/843e04e55cc54ef98b6f7a9eafcc99ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65a87e7a867c4c44bc91f7f329953f58 : SUCCESS in 11m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/696b55f98083459089440af96f5bfa85 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/776e0487c23247c58db0096e473886a0 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9233782ceb7540ef9f6308e68c0fd109 : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/29ee767c954d47b38ab02642e00508bc : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1ae42bb3f174dd7a70e20676cda9faf : SUCCESS in 13m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57008abb290f48e59dcd4a1807ad6ce3 : SUCCESS in 5m 04s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/957de9bf17a14f49a364ed6aa7831289 : SUCCESS in 3m 16s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/888a9fdfa2664788ba4f4ed8f0c847a9 : SUCCESS in 4m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bf7c0951689f4377a70e8213cbcf99ea : SUCCESS in 2m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2490b8acd8f0401f9d07db136f014e5d : SUCCESS in 3m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f0f574d8b32d43a5af2ca9f0c3d42fbe : SUCCESS in 30m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e365b82b577849e0813dcdedd192735f : SUCCESS in 1h 18m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1541d65d707f45809f4899f6072d28d0 : SUCCESS in 1h 01m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4f02b4d22fa04b3faa157a0ff8ff28dc : SUCCESS in 1h 25m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/724642b9f7414ff0883047fb4d079580 : SUCCESS in 1h 06m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/37c501470c3b40a38def9c97ddc81962 : SUCCESS in 41m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/da9b82c4560b425ba6d105dbcc8c3694 : SUCCESS in 45m 21s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5656e72ef95d4f1bb413b122addaefe3 : SUCCESS in 41m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3da18f415b634157b5f87b22cdcf6375 : SUCCESS in 29m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3da71ed0858d4302bffed1bca5d9ad99 : SUCCESS in 39m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fbdbac02e1ef4311a25aaaffaad30cfd : SUCCESS in 2h 14m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7c25786c80a94b8581e3e2a613cb608c : SUCCESS in 1h 27m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/62980b8fb4564d1bbbbf382420d0671f : SUCCESS in 1h 13m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f68b1c34339b4e06ae0cce69c00441b4 : SUCCESS in 54m 31s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"4cc57e1df6e83965bcb580de04b65d314104cb3d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-24 08:06:34.000000000","message":"Patch Set 20: Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"ca0aeaa96fd02313aa95499e567d14d1be46cb02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 08:07:03.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"d04021196900aa86b2cdbb3063aaedcbbf926611","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 09:33:14.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fe423ae91734ef89b0f0a7071770256\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5739d55cf1664461a4a406998d4ecb95 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d259e64775c45738f7585552b22520d : SUCCESS in 8m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/49d49f065e604948b64ec813676ef1b5 : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6286f940c86c40fd9a30d97e6217b059 : SUCCESS in 14m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d714f2a0bf294aa2931dbb75fea2b6aa : SUCCESS in 10m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ef9e28eb514949c4bd2617ce015aad1b : SUCCESS in 4m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f6a828fd6ec14a4193c3ef6027086e6e : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/29f64ef7e2c74dbeb64c588869d5016f : SUCCESS in 58m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1e1fbad101cf4edb9954ba43ecfae212 : SUCCESS in 1h 25m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/802b3222380c4a16a34a668351aec005 : SUCCESS in 59m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/26c1e8898f694a97acd948184c3d55d3 : SUCCESS in 39m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9f64191ac0b944b7885d9f5e869ea570 : SUCCESS in 45m 32s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/97a6ade1414d4727931b153585723e78 : SUCCESS in 46m 36s","accounts_in_message":[],"_revision_number":20},{"id":"cd08c9612f3b4a6697f52337d5192cf2a012266c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 09:33:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":20},{"id":"af00d23c932f1d129c8b16d9d7a18cfa53eae070","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 09:35:12.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a25639d9c9ba4d5d91b7031f52876f04\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/177ecb935d934c4cbae0a1e5fed49bab : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f09b3de3092c4886ac2c2221b36cc2d1 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"6ff1a3fd58ccfd0b085ac14c27691f2635c64413","revisions":{"405b264292e796d909a3ae64c44aaa355b8abd10":{"kind":"REWORK","_number":1,"created":"2025-12-18 12:03:44.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/1"}}},"commit":{"parents":[{"commit":"f42860a7302b521dd2895a22535dbd1de1f912f0","subject":"Remove url tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f42860a7302b521dd2895a22535dbd1de1f912f0"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2025-12-17 21:38:05.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2025-12-18 11:51:51.000000000","tz":0},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Refactored code so that remote group ID would be added to VIP peer port security rules\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Updated unit tests with some fixes\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/405b264292e796d909a3ae64c44aaa355b8abd10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/405b264292e796d909a3ae64c44aaa355b8abd10"}]},"branch":"refs/heads/master"},"f94235c1c29753c27336786a1bdd42eb61b82e9a":{"kind":"REWORK","_number":2,"created":"2026-01-12 09:17:51.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/2"}}},"commit":{"parents":[{"commit":"bf16995711034b65a761eb85fcd786eb53edc0d4","subject":"Merge \"Unpin pylint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bf16995711034b65a761eb85fcd786eb53edc0d4"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Refactored code so that remote group ID would be added to VIP peer port security rules\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Updated unit tests with some fixes\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f94235c1c29753c27336786a1bdd42eb61b82e9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f94235c1c29753c27336786a1bdd42eb61b82e9a"}]},"branch":"refs/heads/master"},"9da00b96650b1c660406183c8b38a23a6892cd95":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-14 21:58:52.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"real_uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/971408/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/3"}}},"commit":{"parents":[{"commit":"f42860a7302b521dd2895a22535dbd1de1f912f0","subject":"Remove url tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f42860a7302b521dd2895a22535dbd1de1f912f0"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-14 21:58:52.000000000","tz":0},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Refactored code so that remote group ID would be added to VIP peer port security rules\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Updated unit tests with some fixes\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9da00b96650b1c660406183c8b38a23a6892cd95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9da00b96650b1c660406183c8b38a23a6892cd95"}]},"branch":"refs/heads/master","description":"Rebase"},"89d9ee0060d978e2ec87bfa642d935874819bba9":{"kind":"REWORK","_number":4,"created":"2026-03-04 11:42:44.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/4"}}},"commit":{"parents":[{"commit":"a753d57186923cc8824464e644784519e0317010","subject":"Merge \"Add c10s jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a753d57186923cc8824464e644784519e0317010"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-03-04 11:40:13.000000000","tz":0},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Refactored code so that remote group ID would be added to VIP peer port security rules\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use\n  remote groups\n- Updated peer port rule to prevent duplicated rules being added\n- Added unit tests to improve coverage of new code\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89d9ee0060d978e2ec87bfa642d935874819bba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89d9ee0060d978e2ec87bfa642d935874819bba9"}]},"branch":"refs/heads/master"},"7de9c29d6f6b2f29dce1a1b37d57b3715832250b":{"kind":"REWORK","_number":5,"created":"2026-03-04 11:57:14.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/5"}}},"commit":{"parents":[{"commit":"a753d57186923cc8824464e644784519e0317010","subject":"Merge \"Add c10s jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a753d57186923cc8824464e644784519e0317010"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-03-04 11:57:05.000000000","tz":0},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Refactored code so that remote group ID would be added to VIP peer port security rules\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use\n  remote groups\n- Updated peer port rule to prevent duplicated rules being added\n- Added unit tests to improve coverage of new code\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7de9c29d6f6b2f29dce1a1b37d57b3715832250b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7de9c29d6f6b2f29dce1a1b37d57b3715832250b"}]},"branch":"refs/heads/master"},"5ffe84abce188790aaad35b30b4fca3118b1a09c":{"kind":"REWORK","_number":6,"created":"2026-03-09 11:18:23.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/6"}}},"commit":{"parents":[{"commit":"a753d57186923cc8824464e644784519e0317010","subject":"Merge \"Add c10s jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a753d57186923cc8824464e644784519e0317010"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-03-09 11:18:10.000000000","tz":0},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Refactored code so that remote group ID would be added to VIP peer port security rules\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use\n  remote groups\n- Updated peer port rule to prevent duplicated rules being added\n- Added unit tests to improve coverage of new code\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ffe84abce188790aaad35b30b4fca3118b1a09c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ffe84abce188790aaad35b30b4fca3118b1a09c"}]},"branch":"refs/heads/master"},"ccb97025481e54ae43bfaf97a7ef7c8c9e02124c":{"kind":"REWORK","_number":7,"created":"2026-03-09 11:38:01.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/7"}}},"commit":{"parents":[{"commit":"a753d57186923cc8824464e644784519e0317010","subject":"Merge \"Add c10s jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a753d57186923cc8824464e644784519e0317010"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-03-09 11:37:51.000000000","tz":0},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Refactored code so that remote group ID would be added to VIP peer port security rules\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use\n  remote groups\n- Updated peer port rule to prevent duplicated rules being added\n- Added unit tests to improve coverage of new code\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ccb97025481e54ae43bfaf97a7ef7c8c9e02124c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ccb97025481e54ae43bfaf97a7ef7c8c9e02124c"}]},"branch":"refs/heads/master"},"6688e5e6676089fc9ac826890ea70be520a53b50":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-03-12 09:29:09.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/8"}}},"commit":{"parents":[{"commit":"6bad959e12075dd15f80f2b51259430955448db1","subject":"Merge \"Remove failing drivers from enabled_provider_drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bad959e12075dd15f80f2b51259430955448db1"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-03-12 09:29:09.000000000","tz":0},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Refactored code so that remote group ID would be added to VIP peer port security rules\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use\n  remote groups\n- Updated peer port rule to prevent duplicated rules being added\n- Added unit tests to improve coverage of new code\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6688e5e6676089fc9ac826890ea70be520a53b50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6688e5e6676089fc9ac826890ea70be520a53b50"}]},"branch":"refs/heads/master","description":"Rebase"},"743230caa235084e6734eb465a023b08f029789f":{"kind":"REWORK","_number":9,"created":"2026-06-17 13:39:56.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/9"}}},"commit":{"parents":[{"commit":"6bad959e12075dd15f80f2b51259430955448db1","subject":"Merge \"Remove failing drivers from enabled_provider_drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bad959e12075dd15f80f2b51259430955448db1"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-06-17 13:39:46.000000000","tz":60},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Refactored code so that remote group ID would be added to VIP peer port security rules\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use\n  remote groups\n- Updated peer port rule to prevent duplicated rules being added\n- Added unit tests to improve coverage of new code\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/743230caa235084e6734eb465a023b08f029789f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/743230caa235084e6734eb465a023b08f029789f"}]},"branch":"refs/heads/master"},"618131272be6990ca0b4eeebe4ed0770a854359d":{"kind":"REWORK","_number":10,"created":"2026-06-25 12:45:58.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/10"}}},"commit":{"parents":[{"commit":"6bad959e12075dd15f80f2b51259430955448db1","subject":"Merge \"Remove failing drivers from enabled_provider_drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bad959e12075dd15f80f2b51259430955448db1"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-06-25 12:45:24.000000000","tz":60},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Refactored code so that remote group ID would be added to VIP peer port security rules\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use\n  remote groups\n- Updated peer port rule to prevent duplicated rules being added\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/618131272be6990ca0b4eeebe4ed0770a854359d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/618131272be6990ca0b4eeebe4ed0770a854359d"}]},"branch":"refs/heads/master"},"5f0ae13fb03abff28cccd64e13b2ee5e81362b2d":{"kind":"REWORK","_number":11,"created":"2026-07-31 13:46:03.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/11"}}},"commit":{"parents":[{"commit":"78d38afc126c37e2ec65859e8cfc770eb5457075","subject":"Remove setting of oslo_policy[enforce_scope] flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-07-31 13:46:01.000000000","tz":60},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Added a config option to enable setting remote security group for VIP peer ports (default: Enabled)\n- Refactored code so that remote group ID would be added to VIP peer port security rules if enabled\n- If remote groyups are didabled then existing behaviour is preserved (remote_group\u003dNone)\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use remote groups\n- Updated peer port rule to prevent duplicated rules being added\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f0ae13fb03abff28cccd64e13b2ee5e81362b2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f0ae13fb03abff28cccd64e13b2ee5e81362b2d"}]},"branch":"refs/heads/master"},"b3aef903cef2a6f51c45f976c36478e6f0e460ad":{"kind":"REWORK","_number":12,"created":"2026-07-31 14:07:10.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/12"}}},"commit":{"parents":[{"commit":"78d38afc126c37e2ec65859e8cfc770eb5457075","subject":"Remove setting of oslo_policy[enforce_scope] flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-07-31 14:07:00.000000000","tz":60},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Added a config option to enable setting remote security group for VIP peer ports (default: Enabled)\n- Refactored code so that remote group ID would be added to VIP peer port security rules if enabled\n- If remote groyups are didabled then existing behaviour is preserved (remote_group\u003dNone)\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use remote groups\n- Updated peer port rule to prevent duplicated rules being added\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3aef903cef2a6f51c45f976c36478e6f0e460ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3aef903cef2a6f51c45f976c36478e6f0e460ad"}]},"branch":"refs/heads/master"},"7310292a291f4e1bce32ff4ff18d591dcd247e5f":{"kind":"REWORK","_number":13,"created":"2026-07-31 14:09:40.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/13"}}},"commit":{"parents":[{"commit":"78d38afc126c37e2ec65859e8cfc770eb5457075","subject":"Remove setting of oslo_policy[enforce_scope] flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-07-31 14:09:32.000000000","tz":60},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Added a config option to enable setting remote security group for VIP peer ports (default: Enabled)\n- Refactored code so that remote group ID would be added to VIP peer port security rules if enabled\n- If remote groyups are didabled then existing behaviour is preserved (remote_group\u003dNone)\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use remote groups\n- Updated peer port rule to prevent duplicated rules being added\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310292a291f4e1bce32ff4ff18d591dcd247e5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310292a291f4e1bce32ff4ff18d591dcd247e5f"}]},"branch":"refs/heads/master"},"cd5c3231b38d73e1d5f1fdeb6ce4681d003c70ee":{"kind":"REWORK","_number":14,"created":"2026-07-31 14:27:44.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/14"}}},"commit":{"parents":[{"commit":"78d38afc126c37e2ec65859e8cfc770eb5457075","subject":"Remove setting of oslo_policy[enforce_scope] flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-07-31 14:27:37.000000000","tz":60},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Added a config option to enable setting remote security group for VIP peer ports (default: Enabled)\n- Refactored code so that remote group ID would be added to VIP peer port security rules if enabled\n- If remote groyups are didabled then existing behaviour is preserved (remote_group\u003dNone)\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use remote groups\n- Updated peer port rule to prevent duplicated rules being added\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd5c3231b38d73e1d5f1fdeb6ce4681d003c70ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd5c3231b38d73e1d5f1fdeb6ce4681d003c70ee"}]},"branch":"refs/heads/master"},"bd9da4b8d50e4d2ccd615008ddbbfc6040bf7cf6":{"kind":"REWORK","_number":15,"created":"2026-07-31 16:02:40.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/15"}}},"commit":{"parents":[{"commit":"78d38afc126c37e2ec65859e8cfc770eb5457075","subject":"Remove setting of oslo_policy[enforce_scope] flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-07-31 16:02:32.000000000","tz":60},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Added a config option to enable setting remote security group for VIP peer ports (default: Enabled)\n- Refactored code so that remote group ID would be added to VIP peer port security rules if enabled\n- If remote groyups are didabled then existing behaviour is preserved (remote_group\u003dNone)\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use remote groups\n- Updated peer port rule to prevent duplicated rules being added\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd9da4b8d50e4d2ccd615008ddbbfc6040bf7cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd9da4b8d50e4d2ccd615008ddbbfc6040bf7cf6"}]},"branch":"refs/heads/master"},"0b980e343c51b3dce68a0a3e32b17877f3047b04":{"kind":"REWORK","_number":16,"created":"2026-08-04 08:37:51.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/16"}}},"commit":{"parents":[{"commit":"78d38afc126c37e2ec65859e8cfc770eb5457075","subject":"Remove setting of oslo_policy[enforce_scope] flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-08-04 08:36:43.000000000","tz":60},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Added a config option to enable setting remote security group for VIP peer ports (default: Enabled)\n- Refactored code so that remote group ID would be added to VIP peer port security rules if enabled\n- If remote groyups are didabled then existing behaviour is preserved (remote_group\u003dNone)\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use remote groups\n- Updated peer port rule to prevent duplicated rules being added\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b980e343c51b3dce68a0a3e32b17877f3047b04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b980e343c51b3dce68a0a3e32b17877f3047b04"}]},"branch":"refs/heads/master"},"5b87e5b428b7e2a162033be003d4cc21a11ccfcb":{"kind":"REWORK","_number":17,"created":"2026-08-05 16:19:22.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/17"}}},"commit":{"parents":[{"commit":"78d38afc126c37e2ec65859e8cfc770eb5457075","subject":"Remove setting of oslo_policy[enforce_scope] flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-08-05 16:18:51.000000000","tz":60},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Added a config option to enable setting remote security group for VIP peer ports (default: Enabled)\n- Refactored code so that remote group ID would be added to VIP peer port security rules if enabled\n- If remote groyups are didabled then existing behaviour is preserved (remote_group\u003dNone)\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use remote groups\n- Updated peer port rule to prevent duplicated rules being added\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b87e5b428b7e2a162033be003d4cc21a11ccfcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b87e5b428b7e2a162033be003d4cc21a11ccfcb"}]},"branch":"refs/heads/master"},"dd154bf8a700188814a744cfa19d0844451226fb":{"kind":"REWORK","_number":18,"created":"2026-08-13 09:16:17.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/18"}}},"commit":{"parents":[{"commit":"78d38afc126c37e2ec65859e8cfc770eb5457075","subject":"Remove setting of oslo_policy[enforce_scope] flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-08-13 09:15:01.000000000","tz":60},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Added a config option to enable setting remote security group for VIP peer ports (default: Enabled)\n- Refactored code so that remote group ID would be added to VIP peer port security rules if enabled\n- If remote groyups are didabled then existing behaviour is preserved (remote_group\u003dNone)\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use remote groups\n- Updated peer port rule to prevent duplicated rules being added\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd154bf8a700188814a744cfa19d0844451226fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd154bf8a700188814a744cfa19d0844451226fb"}]},"branch":"refs/heads/master"},"b44b94c20fc3349891d6dcac2c3ded9a5c2f3ae4":{"kind":"REWORK","_number":19,"created":"2026-08-13 09:40:02.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/19"}}},"commit":{"parents":[{"commit":"78d38afc126c37e2ec65859e8cfc770eb5457075","subject":"Remove setting of oslo_policy[enforce_scope] flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-08-13 09:39:26.000000000","tz":60},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Added a config option to enable setting remote security group for VIP peer ports (default: Enabled)\n- Refactored code so that remote group ID would be added to VIP peer port security rules if enabled\n- If remote groyups are disabled then existing behaviour is preserved (remote_group\u003dNone)\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use remote groups\n- Updated peer port rule to prevent duplicated rules being added\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b44b94c20fc3349891d6dcac2c3ded9a5c2f3ae4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b44b94c20fc3349891d6dcac2c3ded9a5c2f3ae4"}]},"branch":"refs/heads/master"},"6ff1a3fd58ccfd0b085ac14c27691f2635c64413":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2026-08-13 10:20:36.000000000","uploader":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"ref":"refs/changes/08/971408/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/971408/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/971408/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/971408/20"}}},"commit":{"parents":[{"commit":"78d38afc126c37e2ec65859e8cfc770eb5457075","subject":"Remove setting of oslo_policy[enforce_scope] flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-01-12 09:17:10.000000000","tz":0},"committer":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-08-13 10:19:56.000000000","tz":60},"subject":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0","message":"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\n\n- Added a config option to enable setting remote security group for VIP peer ports (default: Enabled)\n- Refactored code so that remote group ID would be added to VIP peer port security rules if enabled\n- If remote groups are disabled then existing behaviour is preserved (remote_group\u003dNone)\n- The remote group ID is the ID of the VIP security group, this means traffic will not be allowed from outside the group for peer ports\n- Also updated security rules for VRRP and Auth header ports to use remote groups\n- Updated peer port rule to prevent duplicated rules being added\n\nCloses-Bug: 2133752\n\nChange-Id: Ib80e7b295827bd8521dcf4960cf8e825ab12314b\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nAssisted-by: Claude Code \u003chttps://anthropic.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ff1a3fd58ccfd0b085ac14c27691f2635c64413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ff1a3fd58ccfd0b085ac14c27691f2635c64413"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
