)]}'
{"id":"openstack%2Foctavia~972291","triplet_id":"openstack%2Foctavia~master~I8126f41e1a0ff1a77d1edacb1e45891c929c740b","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I8126f41e1a0ff1a77d1edacb1e45891c929c740b","subject":"Add support for Rocky 10","status":"ABANDONED","created":"2026-01-06 14:28:51.000000000","updated":"2026-01-08 14:06:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9ab205d25ee2b5d8ec1beb04925f81f483fe4c71","_number":972291,"virtual_id_number":972291,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-06 17:19:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1544e4ada48f475a2471245ad7b792517cf86633","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-06 14:28:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c5b801be1b6ba530811f1c55403cd66ce7ceec5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 17:19:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2103c0f10348485db5bf754f061ff7a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/410419cdda0340dcbba2de60c7e52e77 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b0cd78dd2de4128b7bf9e8c27912854 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0137973049347118b0060d637bb4e12 : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/edf722cab8154cc5a5f3b60378a77055 : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03ab08d3fda344fd98f0b643947ea3dc : SUCCESS in 6m 54s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/dba4db5634cf482d92062f5647de6d6c : SUCCESS in 4m 24s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/44204e44f48747959ebbf0dc060a0559 : SUCCESS in 8m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/912ee4d27406432097028667686a1957 : SUCCESS in 3m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e7b5823517a041739569d49cb5cb176a : SUCCESS in 7m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/37381b5aab6c42a494934f1df2ecb799 : SUCCESS in 10m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2a4f5ac5c97e4ab88294dcd8a552e03f : SUCCESS in 1h 07m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/dcf80c7a54f5429c8dbdfc9cf6afc2bb : SUCCESS in 1h 34m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9d31f9284a424d4abe69f349696bdb91 : SUCCESS in 1h 07m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9fd05a7e302e4d4bb800afd6d83fcfa0 : SUCCESS in 1h 41m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c51155473d2a4242be3ad7f46834eb6e : SUCCESS in 1h 18m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/de541ca3edab42ee85063efae6120684 : SUCCESS in 42m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e57a1229fd2040bfb927804d1b075dba : SUCCESS in 49m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8bb8e0fe97e1453c87440db1879796fa : SUCCESS in 35m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/096df47e9b6a4089b2975920c5c88359 : SUCCESS in 41m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8ead0156908f448db1d520990d7a0f8c : SUCCESS in 2h 46m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fcf8a1c79abb446a8c1c98bcb51b1474 : SUCCESS in 1h 40m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/fdbac9a388304c64a29a10cd5b44be88 : SUCCESS in 1h 24m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9db99d5af01f4660b830012f97fa695d : SUCCESS in 58m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec9aa47bff4cafd4cf2a6033b3e3d3599e828557","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-07 09:07:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"729de8d4de2b391765ea3c9bf2efc1dcf895269a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 11:51:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7939f049680d42629df45166bd346506\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e130e1b9d54245bd8974b166cace1cd9 : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bcd49aecdc54c9183541207a4d74b9e : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb6e7da5fd7f4d2eb4aac9555abb5f11 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f2f424aa564433b9552de284e664f7f : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/105064e82c8542f28475fad7d68e9bbd : SUCCESS in 12m 22s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ad2ce858d53d45ed8eb8d8a6551b88b1 : SUCCESS in 3m 31s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/b57fb5ce3860498f9d9b46957edbbe84 : SUCCESS in 8m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8ad1f61436fc443caadfecc93221da9c : SUCCESS in 2m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33143df461dc47a9ada5434c4201a95f : SUCCESS in 7m 12s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c89c7178aae740b6b64f9bfe41709c94 : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/959dc467e4fc46409f0ed37310d255f2 : SUCCESS in 41m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/dca66f80d5a843478d0b8bb7dfa2472f : SUCCESS in 1h 47m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2f2c3eebd8df4d4d844f83fa61392844 : SUCCESS in 1h 00m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d929b1e7af2a40bcb73a8bf0a507efb3 : SUCCESS in 2h 00m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3d29ffade6f34824999b679405eecede : SUCCESS in 1h 16m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/13085d9106594470ad41763e6bbbe336 : SUCCESS in 38m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/747ee6f9fa1d49439daf371327a8b1de : SUCCESS in 49m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/47de830308b24bbb92e7b7a463254bf4 : SUCCESS in 32m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2bb600cb97fe4b958bc411943012f577 : SUCCESS in 41m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1445bf0d5a41414d90218883af46a947 : SUCCESS in 2h 40m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/76ec386179094b4a967a492576f8de00 : SUCCESS in 1h 31m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d378d739d369348a2d8231745e48b6a271079b19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-08 13:47:12.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b2c46f2c355b072410b0872d30636a7da272a929","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-08 13:47:20.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"9ab205d25ee2b5d8ec1beb04925f81f483fe4c71","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-08 14:06:01.000000000","message":"Abandoned\n\nsquashed in parent patch","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5f14f44433b37cada061ebf05c5dd63d00402828","revisions":{"9eadc372f1627abbde579acdf19db2fdbfda37f1":{"kind":"REWORK","_number":1,"created":"2026-01-06 14:28:51.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/91/972291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/972291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/972291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/972291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/972291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/972291/1"}}},"commit":{"parents":[{"commit":"50ff38df3316174c997f7cc55f4ab7a444da6973","subject":"Add c10s jobs to the check pipeline (non-voting)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50ff38df3316174c997f7cc55f4ab7a444da6973"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-06 14:28:24.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-06 14:28:29.000000000","tz":60},"subject":"WIP Add support for Rocky 10","message":"WIP Add support for Rocky 10\n\nChange-Id: I8126f41e1a0ff1a77d1edacb1e45891c929c740b\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9eadc372f1627abbde579acdf19db2fdbfda37f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9eadc372f1627abbde579acdf19db2fdbfda37f1"}]},"branch":"refs/heads/master"},"861d7a1b1695c6e06eb3dc5cf8d4474d4e83342c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-07 09:07:29.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/91/972291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/972291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/972291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/972291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/972291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/972291/2"}}},"commit":{"parents":[{"commit":"f192625c6af89d966065e50b0ea5f942ef7d5af7","subject":"Add support for centos-10-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f192625c6af89d966065e50b0ea5f942ef7d5af7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-06 14:28:24.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-07 09:07:29.000000000","tz":0},"subject":"WIP Add support for Rocky 10","message":"WIP Add support for Rocky 10\n\nChange-Id: I8126f41e1a0ff1a77d1edacb1e45891c929c740b\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/861d7a1b1695c6e06eb3dc5cf8d4474d4e83342c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/861d7a1b1695c6e06eb3dc5cf8d4474d4e83342c"}]},"branch":"refs/heads/master","description":"Rebase"},"4c0b4291f375a86330fad0a724d6d26e5f1da981":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-08 13:47:12.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/91/972291/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/972291/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/972291/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/972291/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/972291/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/972291/3"}}},"commit":{"parents":[{"commit":"f192625c6af89d966065e50b0ea5f942ef7d5af7","subject":"Add support for centos-10-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f192625c6af89d966065e50b0ea5f942ef7d5af7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-06 14:28:24.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-08 13:47:12.000000000","tz":0},"subject":"Add support for Rocky 10","message":"Add support for Rocky 10\n\nChange-Id: I8126f41e1a0ff1a77d1edacb1e45891c929c740b\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c0b4291f375a86330fad0a724d6d26e5f1da981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c0b4291f375a86330fad0a724d6d26e5f1da981"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5f14f44433b37cada061ebf05c5dd63d00402828":{"kind":"NO_CHANGE","_number":4,"created":"2026-01-08 13:47:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/91/972291/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/972291/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/972291/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/972291/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/972291/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/972291/4"}}},"commit":{"parents":[{"commit":"297b61dd6d0f7a84f0b4ee1c2e17bd765eb1d4d9","subject":"Add support for centos-10-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/297b61dd6d0f7a84f0b4ee1c2e17bd765eb1d4d9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-06 14:28:24.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-08 13:47:20.000000000","tz":0},"subject":"Add support for Rocky 10","message":"Add support for Rocky 10\n\nChange-Id: I8126f41e1a0ff1a77d1edacb1e45891c929c740b\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f14f44433b37cada061ebf05c5dd63d00402828"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f14f44433b37cada061ebf05c5dd63d00402828"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
