)]}'
{"id":"openstack%2Foctavia~974037","triplet_id":"openstack%2Foctavia~master~I1dd437be6783051f4a641fd033d749c7ba77fb5a","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-02-12 14:16:49.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-12 14:42:50.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I1dd437be6783051f4a641fd033d749c7ba77fb5a","subject":"Fix octavia-grenade-skip-level job","status":"MERGED","created":"2026-01-21 11:07:44.000000000","updated":"2026-02-12 19:04:53.000000000","submitted":"2026-02-12 19:01:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974037","meta_rev_id":"4f69a982eb3b9e706fbb0133bc4bb3905fe3ce62","_number":974037,"virtual_id_number":974037,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-12 19:01:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-02-05 09:41:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":2,"date":"2026-01-21 17:45:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2026-02-05 16:30:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-12 14:16:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2026-02-05 16:30:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-21 11:53:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 17:45:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2026-02-05 09:41:21.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-02-12 14:16:40.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"827d72f171ac58c5faa05e58a3f44b9279fd427f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-21 11:07:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcfd71fc831e748ad00a9122a94dd1d80cb27c7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:53:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9e8d28e49b141f3988f82e5126887ef\n\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0aa3c28bd47c4565b14e9d0cd2064d11 : SUCCESS in 42m 31s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/424a18411ebc4b1c9d7b0932923a592b : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a8687279370f1d79cef54935410e1b8300eef49a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-01-21 12:27:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"edfd421528e93a747df5c79903f01125298d0744","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 15:02:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de6e0bb1c8f04232a4c408d9cbc537f7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cfd4d2ed9d144d69c52a5b772bc3297 : SUCCESS in 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b86cc74c75394710b85d3e0fcfd6474d : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac33f1a3f5b04e839ba9e80825b47b6f : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5881f45738a4bb7b363b3757b65c568 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44357c7fd194ccb8dd673230479464a : SUCCESS in 13m 11s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/22f02c95fd274d319264b8f5ccd81620 : SUCCESS in 2m 56s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/a8530578cf464394bcfe4c470e49d51c : SUCCESS in 4m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fcfa7962770c47fc98ab40d3938079c2 : SUCCESS in 2m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7711fa35f3a043028e9b8009626ba576 : SUCCESS in 7m 06s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/47113bf66f5e4101ac13fdbd6bbebc39 : SUCCESS in 5m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9847184ff158493c8d9f83b076f151d0 : SUCCESS in 1h 35m 23s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a71c9d26ad5e4728ad83272fbaf3b908 : SUCCESS in 1h 36m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0076f7b44f364777894cd1f3fc13b822 : SUCCESS in 1h 08m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/59db9fac7f04491abf003f1d1145455f : SUCCESS in 1h 41m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d79576fddba6499da54a4fd3812a8bb3 : SUCCESS in 1h 10m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c7ebb971d88c416c90dd1202a44df635 : SUCCESS in 38m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8030eb06bbef4f45a7e6b02dd07978d1 : SUCCESS in 44m 35s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/0e8d958ae71148f182bf65fa3302914a : SUCCESS in 45m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0a9b415e83164639971d94335c7899e1 : SUCCESS in 32m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e9ede42ef96f48a291b57c7e3313254d : SUCCESS in 36m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a3dcebf5aea34ddaabe803291227b852 : SUCCESS in 2h 33m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/00686904530943a0911e516afe416082 : SUCCESS in 1h 34m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e233a7b04ef9636ad74f8cb092a19fd9702b7138","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2026-01-21 17:45:36.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b17bcc7fbb30873e3c7a1a1e212fbdfd0156decb","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-02-05 09:41:21.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e1da0de941c60bdba19e939d135ed4b8f52249ce","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-05 16:30:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4d762fc26629ed6449c7cfe34414c3a432d1eff0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 16:31:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4503bf20d1e8655f887f73b1687193e386dbc2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 16:57:25.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ef0197a6a2345dc89f84ce327e0cbd8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4891d40608194b32a94a6c9344d4693f : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/983a74944b8c456da1bc9234633f5007 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/288d2fa5859e495f90c47007d2d26449 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7c48028ee4a4dfcb7f46fb69e9f959b : SUCCESS in 12m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e91409433c4446598a9195b0bae88f69 : SUCCESS in 3m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2d637930ef1d492dbc6dc09650c4d916 : SUCCESS in 7m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/99d9c2ec2c6446bca606d2ba2a1d6c59 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/32afff5453ca433690ad9c5157b1837e : FAILURE in 22m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/eadab1e6fc614c1d9528e5fd00f064e7 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ff65b638d4104717b4fe9432b29eefc0 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1e05793903b141e49da4ca3d5b3da7c9 : CANCELED\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2de1f6e2dd03475ea82e3b71b62461c9 : CANCELED","accounts_in_message":[],"_revision_number":2},{"id":"3e0b97b5c5b68c00d902a215db59cedbea8a188a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-12 14:16:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f29138528ff9f4f1028443b91d3878dcf612f617","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-12 14:35:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b1bd83cf27fea6056a35b60332544eb922186fb3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-12 14:42:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c3f2a88aba824dda0837a41aae705ff663ee1839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 17:13:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2740076a940e40aeaae98ab0f560e203\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bdf37a52050422e93414f204a1c433f : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34cb7e19e6b645fba77078fb8f61b152 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61e4acf02c494977a73b9121038e8a95 : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc259fa05bb84abea058496fe9c081e7 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14c88451c90644f4a8fa400e964475ec : SUCCESS in 12m 47s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/3b3695b30291493ab5d1e00fd7ecd03e : SUCCESS in 4m 49s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/58da4bd9126b45d381d94d670a1a34de : SUCCESS in 5m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5ea2258996b847ffa7d6f4ff0c3b20af : SUCCESS in 2m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/93e3f101a6a346aa825e058d3ceedcf4 : SUCCESS in 7m 29s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6405a621b7914e1e83fa9a40c787eb35 : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/779253ef30394495bc7cad3c9e4e05be : SUCCESS in 1h 48m 04s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/27ecce3482dc4a39a90e9a9882c62c2e : SUCCESS in 1h 46m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/415db90033f24887ba76386422cbe847 : SUCCESS in 1h 11m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/133ff67f4e4145b58e3ca0c7ec024d99 : SUCCESS in 2h 02m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ac683fde783f4cdcac232e5f824d6384 : SUCCESS in 1h 18m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1cf7aa10a5f946ca8757bf9a0c8c7f60 : SUCCESS in 44m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1a326b6b7ad24176a52b37a7cc872387 : SUCCESS in 50m 53s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8e31a81a365d40259debe0a5b766d43f : SUCCESS in 51m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7385d627fa4d4e68b1af3a0c18a2ae69 : SUCCESS in 42m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3a2f0ef3d8c344be8557f642723167f9 : SUCCESS in 43m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1727cbc83ae8429f88e20cde84f5465a : SUCCESS in 2h 51m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/50fdf06a87d2404bba197ca7a7589aed : SUCCESS in 1h 36m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"87bc8f05fbccf0a65063ea1c38be7fa5bc9cd869","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 17:13:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"18b439ba5b4102ccc583acdefb2f237e1bf55cc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 19:01:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b25403a2be54a08922d51cb117d6708\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af5e9ab9d893411cb08834f106ae81f5 : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4e6f60bccde45bc98703f0fc18dcd38 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/faf49518c155446e9ff424ccce249771 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e7a5465ec9e4015a8668b2cc85223c2 : SUCCESS in 14m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3ad9e72ce0f241788d738d1779d3feb9 : SUCCESS in 2m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/96f90ec7ef504a2ab5641a06de7f6b58 : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b269b16f3e7d4d0ebaec7abcc0ef288e : SUCCESS in 1h 10m 43s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/47e5b1c753684c00b2460c7aac3b5305 : SUCCESS in 1h 47m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3fe1c291f3e041e3b499144b25898066 : SUCCESS in 1h 04m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/542892a34b7842c59eb92186d9f345a8 : SUCCESS in 41m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6a1ef4551f1a49f79268ff08eb91bdce : SUCCESS in 47m 52s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/66a275fbc106492d94f9da78dd132b0b : SUCCESS in 48m 37s","accounts_in_message":[],"_revision_number":2},{"id":"2f0bd2bc2848e59ee33aaf01fca5f0c95ecd6652","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 19:01:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4f69a982eb3b9e706fbb0133bc4bb3905fe3ce62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 19:04:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d22b74cfd2949ceb610274eb1a285f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae833497a8ea462c8c433377956e1928 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cd688fc6054a5da4168050757a98fdd7e704eb43","revisions":{"71ad3bccd44efb3c318ca94edb1951976af920c0":{"kind":"REWORK","_number":1,"created":"2026-01-21 11:07:44.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/37/974037/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/974037/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/974037/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/974037/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/974037/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/974037/1"}}},"commit":{"parents":[{"commit":"e805a93381530bd1656eb2d2ec535d101d6fe505","subject":"Merge \"Migrate upper functional job to Python 3.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e805a93381530bd1656eb2d2ec535d101d6fe505"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-21 11:06:25.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-21 11:07:33.000000000","tz":60},"subject":"WIP Fix octavia-grenade-skip-level job","message":"WIP Fix octavia-grenade-skip-level job\n\nChange-Id: I1dd437be6783051f4a641fd033d749c7ba77fb5a\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71ad3bccd44efb3c318ca94edb1951976af920c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71ad3bccd44efb3c318ca94edb1951976af920c0"}]},"branch":"refs/heads/master"},"cd688fc6054a5da4168050757a98fdd7e704eb43":{"kind":"REWORK","_number":2,"created":"2026-01-21 12:27:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/37/974037/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/974037/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/974037/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/974037/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/974037/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/974037/2"}}},"commit":{"parents":[{"commit":"e805a93381530bd1656eb2d2ec535d101d6fe505","subject":"Merge \"Migrate upper functional job to Python 3.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e805a93381530bd1656eb2d2ec535d101d6fe505"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-21 11:06:25.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-01-21 12:26:58.000000000","tz":60},"subject":"Fix octavia-grenade-skip-level job","message":"Fix octavia-grenade-skip-level job\n\noctavia-grenade-skip-level was based on an unmaintained job.\n- fixed the definition of the job\n- make the job voting\n- add it back to the gate pipeline\n\nChange-Id: I1dd437be6783051f4a641fd033d749c7ba77fb5a\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd688fc6054a5da4168050757a98fdd7e704eb43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd688fc6054a5da4168050757a98fdd7e704eb43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
