)]}'
{"id":"openstack%2Foctavia~974870","triplet_id":"openstack%2Foctavia~stable%2F2025.1~I5e5f3b8dc8b2a94de927c547b187f9634c572c27","project":"openstack/octavia","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-02-24 06:18:32.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2026-02-24 13:52:53.000000000","reason":"Change was submitted"},"167":{"account":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"last_update":"2026-02-24 13:52:53.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-02-11 17:16:30.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-02-05 10:18:39.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I5e5f3b8dc8b2a94de927c547b187f9634c572c27","subject":"Fix database connection check","status":"MERGED","created":"2026-01-27 14:28:24.000000000","updated":"2026-02-24 13:53:53.000000000","submitted":"2026-02-24 13:52:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974870","cherry_pick_of_change":974212,"cherry_pick_of_patch_set":1,"meta_rev_id":"291bf267eddc8692a6ef05d086f96662859e612e","_number":974870,"virtual_id_number":974870,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-24 13:52:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-02-05 12:10:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":2,"date":"2026-02-05 10:18:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2026-02-11 17:16:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-30 11:35:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2026-02-11 17:16:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"}],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}],"CC":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-27 14:28:24.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"CC"},{"updated":"2026-01-27 17:37:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-30 11:35:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-30 11:35:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2026-01-30 11:36:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-02-05 12:10:01.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-02-11 09:21:35.000000000","updated_by":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"reviewer":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"state":"CC"},{"updated":"2026-02-11 09:21:35.000000000","updated_by":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2026-02-11 09:22:00.000000000","updated_by":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"db666f7bd8189ec891941864d2059bae5df322cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-01-27 14:28:24.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"60bb0f1c97877d9ca16777041f43b592d07be96a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 17:37:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b541faec5f08455098dd3e31d200616d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7547591b75cf4afcbb5bf288414e7042 : SUCCESS in 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62503e2cf30d43b3afeaf142440df21f : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6906e28f67f2444187f99d9f64cb390a : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/166ca3615ee6437a8177f43850475215 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34216f991a504067804842d4bb34768c : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27521a23a4c44f938f3e13e57e2f29d6 : SUCCESS in 5m 36s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/701ad19925de4f4d8008f2ebb1e1872b : SUCCESS in 5m 17s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/dbb710d112a749719fcffb75d271ad3f : SUCCESS in 7m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e8b298c0c0f24bbe95048e123a2c79ea : SUCCESS in 3m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0e1fe43a85e74fe5bee45795c724a77f : SUCCESS in 6m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bf3991803fe945a1b97f1aa00d7d1230 : SUCCESS in 7m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4c0c6b6343e2495eacba1f1f05367251 : SUCCESS in 6m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4a64edec52d44a358688f5c8e3e09d1f : SUCCESS in 1h 03m 06s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/cfa054b985a342e394dd47bfe235c7bd : SUCCESS in 2h 02m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ba78704f5b5d4fa88654b426ee808832 : SUCCESS in 1h 21m 55s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bb64758107941999897e56630588389 : SUCCESS in 2h 01m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29ba9d8cea904b2ca131d44d7fac9fef : SUCCESS in 1h 23m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/437d07a79e644274b55df7118d83e85e : SUCCESS in 2h 13m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e1a3f523740b42c2b4c03f2879a8e07d : SUCCESS in 1h 09m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6ab3b587e2f24543a4a1bcdf190f641c : SUCCESS in 36m 14s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3b58204a00844f87b313ac2deaf56e9a : FAILURE in 53m 01s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/07cbe82ea0ba43dea57b764fb9373dbe : FAILURE in 17m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9c058c5f2964d7db8106b77949cc863 : SUCCESS in 47m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b5f5e83f104b4a73af5d32e202995e36 : SUCCESS in 47m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9d0fe6c7361e46258a27531ec4041bdc : TIMED_OUT in 3h 06m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/baaeb234a56b40e6a8b1564a07a42fa5 : SUCCESS in 2h 00m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/35ccc4128ba044c3bece0f4252a45f7b : FAILURE in 15m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f26c1ed62aa59f6e93aef1c1aa2441bcc6a9f33","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-30 11:35:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"28cf7c7ed677f4fe9301a468096b2c86acce99b0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-05 10:18:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"667e9cf83d3156398e156f845ee2b6ab7f2d37f8","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-02-05 12:10:01.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d60e999233feec63c38e314e0e4b2d436bf617f3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2026-02-11 17:16:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75e235ddfb335c25ccad18065d6d3263696394bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 17:17:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aeb8f01af8f1d27c4c623d7abdbe4b589e266a7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 18:53:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f1e74f5a36f4e94bf1b95599f46168a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb6cdffb12634be5a7a8a0bfc14bd49f : CANCELED in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9397fb40bb344b0a8a31cca0e8068f6b : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/19299cc099664a2099fd49a45b236179 : CANCELED in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8b0e7fa56e3410f99642770bd72a223 : FAILURE in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a00fde965ae344f0a00baa608b1da976 : SUCCESS in 4m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/83a74c8ecdc14ee8bb649fdc073f558e : CANCELED in 2m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6618bb166bb84023a6aeee892ee49388 : CANCELED in 4m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/753138779b0347a7b636952af53505b1 : CANCELED in 3m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/be0eaac0977f4c2d822c72f328ad7a86 : CANCELED in 8m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4c74be1fd119424a93ecab8278eaef54 : CANCELED in 8m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/329a135447d04d90827e67d2d4403a3a : CANCELED in 8m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba15bfd9f9cd4c678c89d9f7f0dbe3aa : CANCELED in 8m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1b50dda06fe8fff0024328317842c425766323fc","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-12 07:48:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"572314c6886c99c2ca7bbe9a05d2964b5c7ddbb0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-24 06:18:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1960462fae86c3a85be09243a43cbe789ba9fdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 09:15:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a20f7ed41b54a90b30076f52309517f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09cc1fd0d0cf4665878603d2e398b7bd : FAILURE in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45ab6fbaf43e46dbb309ebb0a6ddb835 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58ffa0101320423385158a336f9b341b : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e5ff1652cf04c65ade118622253d4b3 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a87ebd69a84c4ef9bda30ed771334788 : SUCCESS in 13m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c1c1707c7074c6598f17ff00ae32968 : SUCCESS in 6m 11s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5f5873db89904c91916ac4e5db17c5a1 : SUCCESS in 4m 26s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/edce7a3bf80c4ebf92b38ddc7e0319fc : SUCCESS in 7m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3ad3ac68cfdf4382a3bbacdb013529ef : SUCCESS in 3m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5333a2bf01d842979752b4c84e1058f7 : SUCCESS in 7m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5c81e2cff2a242618d3d5c79432c6f29 : SUCCESS in 7m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fa2e1b77437947a2ad92403a47843dbb : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/213911f4b294473cbaedab42a817a2eb : SUCCESS in 2h 11m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/103f8c6654cd4f8f890022a28fdab3d0 : SUCCESS in 1h 35m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cd4a9b072f984e499a6179756f2657d1 : SUCCESS in 1h 02m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01362fd07a3340488506da3f18b547af : SUCCESS in 1h 24m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b0096239af5431d84e8c979c3dba50c : SUCCESS in 58m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fee58b1139fe4939869033968a627e26 : SUCCESS in 1h 35m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bf54be49e9f348d5a7c3668d13936a16 : SUCCESS in 1h 06m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7703da0cc9fd4b2f896c67e3a6bfccd0 : SUCCESS in 36m 07s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/262c7c8b081f4b15a0bc46a0dd91406c : FAILURE in 43m 47s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3bd9f9eeac374e75a5c0ba21d39d3fd6 : FAILURE in 17m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/deed434c58654e0498f2623db0e4a96e : SUCCESS in 40m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6eadd054cbba4b72b6688656bc2b6132 : SUCCESS in 40m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6cbf6d0702724c50825884a3f549f0e9 : FAILURE in 2h 55m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5257290ada3f4e34973a358bb5a311f1 : SUCCESS in 1h 32m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/557be684d1394efda18a6fc596b0a5b9 : FAILURE in 14m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e994fe84987ffeca8549f44adbd49ca167ea54d1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-24 09:34:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ee9cc98868041669f76ecf541c93735c63fd171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 12:17:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7465f5ba52bf48269bc7da5bd2ef1bf6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/717d6d03f97d468eb8179c35946fee84 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a75a4fce1ad4e2a8d8920185987fdc3 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ed8a8fcc4a244629649f4e0304323c0 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35e37a424f074abaa70e25aabecad45f : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740e2bbb6e2f4101b48091d3b1ed94cc : SUCCESS in 14m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/578fbf6f7a8648639ea983c2a4be89df : SUCCESS in 5m 12s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/86c4b244431e435996c7370c0d069bb0 : SUCCESS in 4m 10s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/b73a86de2022492cbebe121784aa8faa : SUCCESS in 9m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9ce23c6ed9224f8984b76fdc33698448 : SUCCESS in 2m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0cd996973da2459b953d265f7eaf985b : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/462a2cc3b37242369ac83d99d33d3823 : SUCCESS in 4m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/07853ba9574742278ad5f67c82695ec9 : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5757390a560a420a91e68b799c8d9c40 : SUCCESS in 1h 02m 07s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f43a2215c2ae40c9aab2bf6e561e59eb : SUCCESS in 1h 27m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2f363430d15549b6b357b048a439068d : SUCCESS in 1h 06m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25f6337f538d43cda610dffd5d359be2 : SUCCESS in 1h 24m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9ca1609100c4245b852ad22ad1862b9 : SUCCESS in 1h 02m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/68cfbb77722e435d95eec2e51291c6dd : SUCCESS in 1h 33m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4594908033314f17b9cbab396c1bc0a0 : SUCCESS in 1h 10m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dbcab036358d400baef6f5b91dc493c3 : SUCCESS in 41m 14s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9d00054e62be4ee7a95929549fdbafd8 : FAILURE in 38m 47s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6e848707f33d43e9a4c1269a500c8578 : FAILURE in 18m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2095ac1bfebc444aa75293f039399470 : SUCCESS in 34m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b4d500b15bb2449f9b9fc73d1b105867 : SUCCESS in 33m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8e7a1549aa6b4b1199ed97b502dfb5eb : SUCCESS in 2h 38m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b98b05653be04fcf8d608ba7261e60e4 : SUCCESS in 1h 28m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/92661208a3c64a0aaa13e85caedff8df : FAILURE in 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df7b37e67bedd92207763dc1233dd4e0a8cc8f1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 12:17:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7fe334989502939047a16bba2b800ac7926c8407","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 13:52:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f95c282c276f4b3b8850ae00413b599c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29cd59aa8c01425aaf0ca5f967db794d : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86fda24a14a742f997f5234927f8ca8a : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5726749911548798671da43a4de58fe : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02222e0dcf7f47abb46f8adcd1880e73 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59a530800b2f4018a4517b14bbc11dad : SUCCESS in 4m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f7cb94a2e6bf416b86a5fee7d6f20070 : SUCCESS in 4m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e751b16a308a44fe9fc59f97d92b6277 : SUCCESS in 4m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1a60f84de280455a9bb58c7db6154e5a : SUCCESS in 1h 29m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/915d93ea9be9405681eb57a2372adbf2 : SUCCESS in 1h 24m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6d16decc039b43d9a0c11b8e3e25b5e8 : SUCCESS in 1h 04m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b76a8b67d7e64b4f9d4edac7d4c89e5a : SUCCESS in 1h 24m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85dc8c8fc7a846c1bb5aedb5c4d4ac82 : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ae8fcd19b132f057f84a90e7964467ddfffc6de5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 13:52:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"291bf267eddc8692a6ef05d086f96662859e612e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 13:53:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feea70abb0564eeca4daafa54e9f6abb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfa6186f9ef34adaac39417bd2e5d433 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39587d542c8a4cc58352c7aba41a31a0 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8eb2e9cdd5eee7d28da452776ba8c7bf9bb11e48","revisions":{"8eb2e9cdd5eee7d28da452776ba8c7bf9bb11e48":{"kind":"REWORK","_number":1,"created":"2026-01-27 14:28:24.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/70/974870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/974870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/974870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/974870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/974870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/974870/1"}}},"commit":{"parents":[{"commit":"d67679c00b1c2e984fde45c298c9aa846c21a366","subject":"Merge \"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d67679c00b1c2e984fde45c298c9aa846c21a366"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2025-10-22 15:30:11.000000000","tz":60},"committer":{"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","date":"2026-01-27 14:28:24.000000000","tz":0},"subject":"Fix database connection check","message":"Fix database connection check\n\nSQLAlchemy 2 arrived in the Dalmation release. This removed the concept\nof \u0027Connectionless execution\u0027. See [1] for further details.\n\nPrior to this change, if the connection to the database was interrupted,\nthe Octavia Health Worker would go into an infinite loop due to the removed\nmethod / catch all exception handler.\n\n[1] https://docs.sqlalchemy.org/en/20/changelog/migration_20.html#migration-20-implicit-execution\nCloses-Bug: #2129562\n\nNote: test_api.py was partially created by Claude Code\n\nAssisted-By: Claude Code\nChange-Id: I5e5f3b8dc8b2a94de927c547b187f9634c572c27\nSigned-off-by: Doug Szumski \u003cdoug@stackhpc.com\u003e\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit 6fb0d272c78a101fca395afd875241e43318434c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8eb2e9cdd5eee7d28da452776ba8c7bf9bb11e48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8eb2e9cdd5eee7d28da452776ba8c7bf9bb11e48"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
