)]}'
{"id":"openstack%2Foctavia~975190","triplet_id":"openstack%2Foctavia~unmaintained%2F2024.1~I04e5198962639ac276b7042705e0deff898a5614","project":"openstack/octavia","branch":"unmaintained/2024.1","attention_set":{},"removed_from_attention_set":{"38797":{"account":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"last_update":"2026-09-16 23:17:47.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I04e5198962639ac276b7042705e0deff898a5614","subject":"Fix: Fixing the branch from which diskimage pulls the images, changing from stable/2024.1 to unmaintained/2024.1","status":"ABANDONED","created":"2026-01-29 18:01:21.000000000","updated":"2026-09-16 23:17:47.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a4701a6279c83c3dd764c3d338b369c31e4a3ebc","_number":975190,"virtual_id_number":975190,"owner":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-01-29 20:38:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-29 18:06:29.000000000","updated_by":{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"},"reviewer":{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"},"state":"REVIEWER"},{"updated":"2026-01-29 20:38:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-03 17:49:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-06-18 16:50:48.000000000","updated_by":{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"},"reviewer":{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"},"state":"REMOVED"}],"messages":[{"id":"1c66254fe188cb9237e3f65c24ca5fb0b9490576","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-01-29 18:01:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"468779d2cd95287079549b1a66b37467916c217f","author":{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"},"date":"2026-01-29 18:06:29.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1de99b2a1767d61582c1240d83de067dfc9ea709","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 20:38:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b2e9e0d5b4840a59225a4c3435ef57a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49190c315ef84e7ebda14bf39276072b : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/400931a113b74d11bcac5b49f6f3a65a : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58b2a956a4ce40878c71b578806cce5b : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b55e4e66249e4400a8854d7c94706dc1 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/743c6a1cd08245a5ac7b226ddc79ef7a : SUCCESS in 11m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c6f9006e74494180adc9894fe6749886 : SUCCESS in 4m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/241bfa5cb6e7480da019a90307a73744 : SUCCESS in 4m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/98f8cad009eb4977a8605e1942464447 : SUCCESS in 6m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/be06f2fb2e8c43849895f85e6a1f3330 : SUCCESS in 6m 43s (non-voting)\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/5b1d6dcacca64118a8ade21b05dae24c : FAILURE in 17m 40s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/34e042ff09d246c8848e13483ae3d4e7 : FAILURE in 26m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/fbfcf37cf658458aa7a623140bb50b32 : FAILURE in 25m 40s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/8822564c8dec4a719fba10b8a6ad3696 : TIMED_OUT in 2h 33m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/ae191d2a98bc4637a7b0d8781bf93d1f : TIMED_OUT in 2h 34m 28s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3bfb143d4afb4b51aeb1cd35735c03f9 : FAILURE in 26m 30s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/17c1c76835a348118b42ac9fb140f3a8 : SUCCESS in 38m 04s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/6815d890f91e477c839752b5709f6d55 : FAILURE in 36m 46s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-1 https://zuul.opendev.org/t/openstack/build/19f142028ac64fc7aea394c39140211d : FAILURE in 14m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d066027af34a4d92b6ed6a61c61c71ee : FAILURE in 28m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-1 https://zuul.opendev.org/t/openstack/build/7746e800dc734b4689ed7e1b90e7f360 : POST_FAILURE in 14m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-1 https://zuul.opendev.org/t/openstack/build/b628885fe3dd41e79a5de1bd0ab0b3b9 : RETRY_LIMIT in 4m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-1 https://zuul.opendev.org/t/openstack/build/15f056e6782142d6a23a18282a747390 : FAILURE in 27m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-1 https://zuul.opendev.org/t/openstack/build/cedf972a3b12415593324806cb050f48 : FAILURE in 17m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40f615c256b71e4c3bdff61f6b461aba165896fc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-03 17:49:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1d83616ba76521547f752917ba03c217d783c89","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"},"date":"2026-06-18 16:50:48.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_36824\u003e with the following votes:\n\n* Code-Review+1 by \u003cGERRIT_ACCOUNT_36824\u003e\n* Backport-Candidate+1 by \u003cGERRIT_ACCOUNT_36824\u003e\n","accounts_in_message":[{"_account_id":36824,"name":"Isaac Vicente de Medeiros Silva","display_name":"Isaac Silva","email":"isaacvicentsocial@gmail.com","username":"isaacvicente","status":"Cloud Engineer working with Openstack on Laboratory of Distributed System at LSD, Brazil"}],"_revision_number":1},{"id":"a4701a6279c83c3dd764c3d338b369c31e4a3ebc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"date":"2026-09-16 23:17:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b77c8a9110a4e8ac9c693a939ee65b52f1364123","revisions":{"b77c8a9110a4e8ac9c693a939ee65b52f1364123":{"kind":"REWORK","_number":1,"created":"2026-01-29 18:01:21.000000000","uploader":{"_account_id":38797,"name":"Eduardo Henrique Lima de Morais","display_name":"Eduardo Morais","email":"eduardo.henrique.lima.morais@gmail.com","username":"Eduardo-Morais"},"ref":"refs/changes/90/975190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/975190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/975190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/975190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/975190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/975190/1"}}},"commit":{"parents":[{"commit":"6422a0d27a0f48d18ac30e74c08b8748acffef9c","subject":"Update .gitreview for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6422a0d27a0f48d18ac30e74c08b8748acffef9c"}]}],"author":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-01-29 18:00:45.000000000","tz":-180},"committer":{"name":"Eduardo-Morais","email":"eduardo.henrique.lima.morais@gmail.com","date":"2026-01-29 18:00:45.000000000","tz":-180},"subject":"Fix: Fixing the branch from which diskimage pulls the images, changing from stable/2024.1 to unmaintained/2024.1","message":"Fix: Fixing the branch from which diskimage pulls the images, changing from stable/2024.1 to unmaintained/2024.1\n\nChange-Id: I04e5198962639ac276b7042705e0deff898a5614\nSigned-off-by: Eduardo-Morais \u003ceduardo.henrique.lima.morais@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b77c8a9110a4e8ac9c693a939ee65b52f1364123"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b77c8a9110a4e8ac9c693a939ee65b52f1364123"}]},"branch":"refs/heads/unmaintained/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
