)]}'
{"id":"openstack%2Foctavia~975193","triplet_id":"openstack%2Foctavia~stable%2F2024.2~I5e5f3b8dc8b2a94de927c547b187f9634c572c27","project":"openstack/octavia","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-02-25 16:19:51.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2026-02-25 16:19:51.000000000","reason":"Change was submitted"},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2026-02-25 16:19:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5e5f3b8dc8b2a94de927c547b187f9634c572c27","subject":"Fix database connection check","status":"MERGED","created":"2026-01-29 18:16:18.000000000","updated":"2026-02-25 16:21:05.000000000","submitted":"2026-02-25 16:19:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975193","cherry_pick_of_change":964549,"cherry_pick_of_patch_set":4,"meta_rev_id":"b40ba5e67b10f74dd3f78ca3af3e865f2c811f59","_number":975193,"virtual_id_number":975193,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-25 16:19:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-02-05 12:10:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":2,"date":"2026-02-05 10:18:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-11 09:18:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":1,"date":"2026-02-25 14:42:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}],"CC":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-29 18:16:18.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"CC"},{"updated":"2026-01-29 20:48:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-05 10:16:52.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2026-02-05 10:18:56.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-02-05 12:10:34.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-02-11 09:18:10.000000000","updated_by":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"reviewer":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"state":"REVIEWER"},{"updated":"2026-02-11 09:18:34.000000000","updated_by":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2026-02-11 09:18:58.000000000","updated_by":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"8cc169da7891573289bba7a9e7fc9eb8307d5d82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2026-01-29 18:16:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4c9689063337dd6e1510313a19be2808a0a21097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 20:48:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0cf6dc2b0a943ae9cc4302ddc5a70b3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fe85b0a5449402baa145509e1b3b526 : SUCCESS in 13m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5831af81d5914f3b8cb76c01815310df : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f39d7ae755a04c439e08590859de7c7d : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7f9dee607c944f996aa0105ccb16e48 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9bded98a322a4dfaae47bf6514ca1997 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72ca32888256431aba9a19831381009a : SUCCESS in 13m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81f97cefa9644c8fb126691605e2669a : SUCCESS in 6m 35s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9d639e512da04bb6991a5eba9f0a85f6 : FAILURE in 7m 05s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/588df78485eb435da6db8bcb3f2143a6 : SUCCESS in 10m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/036744d47b6645b69f0adcc5ddb52044 : SUCCESS in 5m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fca8d44e5e5f4d6eae0038f4ef5022b0 : SUCCESS in 9m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a74c669b02494421a8280eaf5155b290 : SUCCESS in 5m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bec596de294d47c0b91a2a4f9fadb679 : SUCCESS in 6m 59s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7c49442d3e2c4415878900bb6824fae8 : SUCCESS in 41m 21s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f374c96f196e4a4b978f8d90650d0881 : SUCCESS in 2h 16m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f81d6d9575c04302851a2368078bfb70 : SUCCESS in 1h 16m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/3af8502ebe1a4f1684f377e17a1734e4 : SUCCESS in 2h 21m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/02a00ebe481146aaaa5161930e2b4099 : SUCCESS in 1h 13m 24s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/77b1105841c745249a313f71442bb69a : SUCCESS in 46m 17s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/09859a481ea84183ab32abc13e397f47 : SUCCESS in 49m 00s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7e344208ac68454fad205bbc482e62ec : FAILURE in 29m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/729454d374834983917b014dc1c99b96 : SUCCESS in 33m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-2 https://zuul.opendev.org/t/openstack/build/47acee6604ff49ba9125309fc6b6385e : SUCCESS in 43m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-2 https://zuul.opendev.org/t/openstack/build/9faab78c165d4e6e9597fbe11246fd75 : FAILURE in 47m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-2 https://zuul.opendev.org/t/openstack/build/3aa0590aa99b41d399f61dee4d51f953 : SUCCESS in 2h 13m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f05482f90b3e44c1a3745ac002f0b044 : FAILURE in 14m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4be93855254136b67c33c357c69ca78e9d8f10e3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-05 10:16:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"92ce265daba576418acb2a429dd224cf24f3fa29","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-05 10:18:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"414c5532b7a1a69e6726f55e9241f69ebb7e3950","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-05 10:22:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a59718f9db9af4bd4753f8d82847262d9ea308ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 12:03:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce2c0da626cc40babbc141eb55d4a4ec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20aa0268feb74c819acdee86ea06562e : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeb69cf217fc460ca5fc71ba30ba2d59 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5a264b4096940ca83da28ed0c37040b : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/80cbb8714bf24fffb55d2e80178cf900 : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e8b75e6ad95450da0e525820771558c : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0861645e376f4b0b99c4ce0b80f09813 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51f88d85d9a74fc48c514d9fcd53515e : SUCCESS in 5m 34s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/468411a22c2f488384b2f3aa016c0079 : SUCCESS in 4m 53s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/b41a092fd608446395a74ed7add243de : SUCCESS in 8m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/480ebabee155419990896601b8392591 : SUCCESS in 2m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/83c22cf345ed4e5c874f6a7912d4aaa7 : SUCCESS in 6m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e9d44324d5b54090b6a1442afbb6debc : SUCCESS in 4m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/82e4f234ac294a6db5880621b83e330e : SUCCESS in 8m 16s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/73e4b16f60ab45899ddd6537d839256e : SUCCESS in 1h 18m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/1fd70bb5297447a6b69ed66ac171ad67 : SUCCESS in 1h 29m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b5cf80f91cc24a7fb07c94f1c7f86019 : SUCCESS in 1h 10m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/5343bc43f5c94e55b3f1b7c7dccc775e : SUCCESS in 1h 39m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/252a87777d754491a64813574f19b7e6 : SUCCESS in 1h 09m 03s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b1bb49c2efad4c76a6297e63d6f9e699 : SUCCESS in 40m 22s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/74b9037a21c94f6ba0f61cfb7d0acc96 : SUCCESS in 46m 16s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-2 https://zuul.opendev.org/t/openstack/build/31fa4b4068814e5bb08cc9bcd50fa59d : FAILURE in 17m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/2bb599c4f67241f398acff99d7f2e055 : SUCCESS in 37m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ca19ab4ad2d2423d8fa871eb34627c31 : SUCCESS in 29m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-2 https://zuul.opendev.org/t/openstack/build/e34d6e3b35ec48c9a940cd6cc1743a19 : FAILURE in 1h 16m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-2 https://zuul.opendev.org/t/openstack/build/1b8f2e903b9b4c678f50e8e0eea542e9 : SUCCESS in 1h 22m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-2 https://zuul.opendev.org/t/openstack/build/eaffd4a3c638487b9d19546f1e223314 : FAILURE in 15m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"844f242c4e54616b8227dacdd5ce377ff5bdbb4f","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-02-05 12:10:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c59631f3cd6656135b46cd79655c5ba3fe1977d","author":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"date":"2026-02-11 09:18:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6e3096afcfced2cf68f723195930089dba2714e6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-25 14:42:28.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10b15c7008af6827ddbd6edb5e182fcc2b7f3316","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 14:43:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed88e85e4c2c55c03153ff21eb89230e379bea0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 16:19:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/447c258ca95046c0a4590960e6482103\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cc769ce08de4ee2b9bbcb42cd1ae17a : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a57c9c145fe4825b84f882917edf21c : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4908a1a6931d42e2a9aeffefde097c4b : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/342e1d593be1486da9470c293f013e63 : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53d8c5d50ff7424ea2015f535c796e2d : SUCCESS in 5m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/88843a49057b480688ab5cc2d3d7380a : SUCCESS in 4m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fe4abdcba6854af19ac7c6fc481e6117 : SUCCESS in 7m 06s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/6e9c8285d2a54b8a838018856933cac4 : SUCCESS in 1h 19m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/bef7b3b7f30c453a966f798ab3c342d2 : SUCCESS in 1h 36m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4cadd5d555c7463bb9527280f9f9364c : SUCCESS in 57m 00s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/a1f27b70245543829fb3e26679797777 : SUCCESS in 39m 50s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b705e5879313492d80651ae86378d512 : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":1},{"id":"02b1016cef14379ba5aa841fa2997fd07ddb4335","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 16:19:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b40ba5e67b10f74dd3f78ca3af3e865f2c811f59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 16:21:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10f0ec4eac594bfa8566b9f20979ca77\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19bde50226114f2ea38e65f84c80f3a7 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c238ddeca400403b82de304675406a01 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd3a16731b8b893de86f748a4b10a5fb9c8c7482","revisions":{"cd3a16731b8b893de86f748a4b10a5fb9c8c7482":{"kind":"REWORK","_number":1,"created":"2026-01-29 18:16:18.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/93/975193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/975193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/975193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/975193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/975193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/975193/1"}}},"commit":{"parents":[{"commit":"ac102c9a4dfd60deac936310952cb34a938df595","subject":"Merge \"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac102c9a4dfd60deac936310952cb34a938df595"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2025-10-22 15:30:11.000000000","tz":60},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2026-01-29 18:16:18.000000000","tz":0},"subject":"Fix database connection check","message":"Fix database connection check\n\nSQLAlchemy 2 arrived in the Dalmation release. This removed the concept\nof \u0027Connectionless execution\u0027. See [1] for further details.\n\nPrior to this change, if the connection to the database was interrupted,\nthe Octavia Health Worker would go into an infinite loop due to the removed\nmethod / catch all exception handler.\n\n[1] https://docs.sqlalchemy.org/en/20/changelog/migration_20.html#migration-20-implicit-execution\nCloses-Bug: #2129562\n\nNote: test_api.py was partially created by Claude Code\n\nAssisted-By: Claude Code\nChange-Id: I5e5f3b8dc8b2a94de927c547b187f9634c572c27\nSigned-off-by: Doug Szumski \u003cdoug@stackhpc.com\u003e\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit 6fb0d272c78a101fca395afd875241e43318434c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd3a16731b8b893de86f748a4b10a5fb9c8c7482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd3a16731b8b893de86f748a4b10a5fb9c8c7482"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
