)]}'
{"id":"openstack%2Foctavia~975571","triplet_id":"openstack%2Foctavia~unmaintained%2F2023.1~Id9bb0e1716312cbce9164460e8c6a2df9ce6435b","project":"openstack/octavia","branch":"unmaintained/2023.1","topic":"unmaintained-job-cleanup","attention_set":{},"removed_from_attention_set":{"38360":{"account":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"last_update":"2026-02-05 22:31:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9bb0e1716312cbce9164460e8c6a2df9ce6435b","subject":"Update jobs for unmaintained/2023.1","status":"MERGED","created":"2026-02-03 20:16:10.000000000","updated":"2026-02-05 22:33:22.000000000","submitted":"2026-02-05 22:31:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975571-unmaintained-job-cleanup","meta_rev_id":"da63fb7b0f6de0582ec4b0ed159e06bdece4486c","_number":975571,"virtual_id_number":975571,"owner":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-05 22:31:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":2,"date":"2026-02-05 21:08:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2026-02-05 20:58:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2026-02-05 21:08:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-03 20:16:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-04 21:02:41.000000000","updated_by":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2026-02-05 20:58:50.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2026-02-05 21:08:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"352794e515e56334c6d0853d801b26351c6e3793","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-03 20:16:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fae5545c65d93dbce634e233346a2c95d4284650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 20:16:30.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/octavia on branch unmaintained/2023.1.  The\nproblem was:\n\n  Job octavia-v2-dsvm-scenario-amphora-v2-unmaintained-2023-1 not defined\n\nThe problem appears in the \"opendev.org/openstack/octavia\" project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - periodic-stable-jobs-neutron\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n  ...\n\n  in \"openstack/octavia/zuul.d/projects.yaml@unmaintained/2023.1\", line 3","accounts_in_message":[],"_revision_number":1},{"id":"707ea630d8d3913637827fce5a5a599ff6c465fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-03 20:19:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d52d3e55dde3363ce1bd3e184019968fc5e84b3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 22:40:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/855e4d9647854b23814bd2d3c27590a8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf08608e294447b5899309a7ee016c02 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bad92f8d9c44040abc75efcd0f6a339 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7b33749e121487a9cb4bbbaf61ec8ea : SUCCESS in 11m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/67f1b96345694520966923f97b52622d : SUCCESS in 2m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/748ed1e052b34c95832a1b7c7d0dce95 : SUCCESS in 8m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2014746cdf66437fac4b1d5bb30b880e : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/97abcc6620384d758675f3f668698379 : SUCCESS in 8m 52s (non-voting)\n- octavia-v2-dsvm-noop-api-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/a7cd54254db446d495e34749092176de : SUCCESS in 1h 03m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/8ed746dafaa0453299a864f19447a25a : SUCCESS in 1h 19m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/6bdbbc61ac5e43a5bf78bae81a2bc59b : SUCCESS in 55m 50s\n- octavia-v2-dsvm-tls-barbican-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/4dc591b65b884500ba117d75c5d131bb : SUCCESS in 32m 28s\n- octavia-v2-dsvm-tls-barbican-fips-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/39d453b25af5404ea3a209a81a4674cd : FAILURE in 18m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/85854488ebde47fbb16fad7d5874c313 : SUCCESS in 28m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/140f2c9dcaf241a799df2e714959f86e : SUCCESS in 31m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/20d5853a9f904a3f830bbcebad640684 : RETRY_LIMIT in 4m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/5d19b5cfae644465b160eb7d72ad2aeb : FAILURE in 19m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/a2840db7bde04d988d1fe0fde5d103cd : SUCCESS in 2h 03m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/8b0c5a6332c04da2b736c0a808715edc : SUCCESS in 1h 54m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/8c1ae4db7452422092f3419a6fe3f8f3 : FAILURE in 15m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d5e2dc4220a17a794ea42927161dff1aa134c455","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-04 18:28:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"796661e54c108913d285a3dff44fa82cc3d0abfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-04 18:30:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c2b0d92a5227ffc9e790ff155348942fbc9e8429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 20:02:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1a7ed8f14ba4347ae66b567b1cf414a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/699bec0210b44d539ed753b289fabeda : SUCCESS in 12m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abdb519e81774efa88c57050dc598d94 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c44674fee21454997de09f6af9d960d : SUCCESS in 11m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/829db30153d54340bb0c1323549614c5 : SUCCESS in 6m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/70bf32c2be4b45b8a3d0683bc08e36bf : SUCCESS in 8m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4f48715376f0485d9a1916adf797b595 : SUCCESS in 8m 07s\n- octavia-v2-dsvm-noop-api-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/e81920a4fd354216b8efe7cef73cfa7f : SUCCESS in 1h 09m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/7e835207c0f54637b6bb453adf922791 : SUCCESS in 1h 29m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/026d6f38b42a4a68832e18f2c903ee04 : SUCCESS in 1h 05m 48s\n- octavia-v2-dsvm-tls-barbican-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/5b9e5eb41d8c427aa949f42f6e28a936 : SUCCESS in 37m 56s","accounts_in_message":[],"_revision_number":4},{"id":"267fc534a388045b8617aca780099a91d07bff44","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2026-02-05 20:58:50.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c5d8d38a1f44e61f31c880066cf8a0cc41afd2d3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-05 21:08:15.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d8738745de72a19635845eecc0de9a02e40df162","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 21:08:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"09320ec9b3b1ed9740063976150bcad8ecad7439","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 22:31:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c53e22d977614d739888a6755b35b558\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2596819a0ac4d47aacdb9ac9ee58c35 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f5c21ae8c8946628391263d8fa2acce : SUCCESS in 12m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ee861a3c82ce495da2c726ff86c536ec : SUCCESS in 2m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ae662aec92c644baa8aee18ac7ce56c9 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/8e831e9265e4478bacbfff301016ef24 : SUCCESS in 1h 15m 39s\n- octavia-v2-dsvm-scenario-traffic-ops-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/3757fedfb1344e56806658090a8fb69d : SUCCESS in 1h 20m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/60b5da0cc889452388203b91d3e8fe21 : SUCCESS in 1h 00m 38s\n- octavia-v2-dsvm-tls-barbican-unmaintained-2023-1 https://zuul.opendev.org/t/openstack/build/2e1e6453efc24892afe7f918d0ca09b0 : SUCCESS in 33m 38s","accounts_in_message":[],"_revision_number":4},{"id":"cb180654e35611360b314477dc541b344469883a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 22:31:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"da63fb7b0f6de0582ec4b0ed159e06bdece4486c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 22:33:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d87a513f8a1e467d87d504de113bc9fc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c6c7535e50e4697b9ffc4bdabf7943b : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"47a59007c2eb3d8d10f14612481630ac66b28415","revisions":{"a26f8510d53ee5f33547fc358716cf03c1676976":{"kind":"REWORK","_number":1,"created":"2026-02-03 20:16:10.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/71/975571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/975571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/975571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/975571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/975571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/975571/1"}}},"commit":{"parents":[{"commit":"5975097dd4b021385178bad3815e2cc41aded276","subject":"Specialize v2-dsvm-noop-api job for 2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5975097dd4b021385178bad3815e2cc41aded276"}]}],"author":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-03 20:15:39.000000000","tz":-360},"committer":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-03 20:15:39.000000000","tz":-360},"subject":"Update jobs for unmaintained/2023.1","message":"Update jobs for unmaintained/2023.1\n\nA number of jobs on unmaintained/2023.1\nrequired specific options to be run correctly.\nSome of these were not running before but have\nsince been re-enabled due to\nI10b8b020668c49315723d6f0e464f6da8c76cb56.\n\nThis change specializes these jobs for\nthe unmaintained/2023.1 branch and attempts\nto clean up some unused jobs.\n\nChange-Id: Id9bb0e1716312cbce9164460e8c6a2df9ce6435b\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a26f8510d53ee5f33547fc358716cf03c1676976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a26f8510d53ee5f33547fc358716cf03c1676976"}]},"branch":"refs/heads/unmaintained/2023.1"},"eaf824581dbbc9f8c1957c4bf15a12e49397a603":{"kind":"REWORK","_number":2,"created":"2026-02-03 20:19:46.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/71/975571/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/975571/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/975571/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/975571/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/975571/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/975571/2"}}},"commit":{"parents":[{"commit":"5975097dd4b021385178bad3815e2cc41aded276","subject":"Specialize v2-dsvm-noop-api job for 2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5975097dd4b021385178bad3815e2cc41aded276"}]}],"author":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-03 20:15:39.000000000","tz":-360},"committer":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-03 20:19:41.000000000","tz":-360},"subject":"Update jobs for unmaintained/2023.1","message":"Update jobs for unmaintained/2023.1\n\nA number of jobs on unmaintained/2023.1\nrequired specific options to be run correctly.\nSome of these were not running before but have\nsince been re-enabled due to\nI10b8b020668c49315723d6f0e464f6da8c76cb56.\n\nThis change specializes these jobs for\nthe unmaintained/2023.1 branch and attempts\nto clean up some unused jobs.\n\nChange-Id: Id9bb0e1716312cbce9164460e8c6a2df9ce6435b\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eaf824581dbbc9f8c1957c4bf15a12e49397a603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eaf824581dbbc9f8c1957c4bf15a12e49397a603"}]},"branch":"refs/heads/unmaintained/2023.1"},"c2ba6377bfe243d2d8d80158af00f9e3008649b1":{"kind":"REWORK","_number":3,"created":"2026-02-04 18:28:32.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/71/975571/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/975571/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/975571/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/975571/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/975571/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/975571/3"}}},"commit":{"parents":[{"commit":"5975097dd4b021385178bad3815e2cc41aded276","subject":"Specialize v2-dsvm-noop-api job for 2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5975097dd4b021385178bad3815e2cc41aded276"}]}],"author":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-03 20:15:39.000000000","tz":-360},"committer":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-04 18:26:41.000000000","tz":-360},"subject":"Update jobs for unmaintained/2023.1","message":"Update jobs for unmaintained/2023.1\n\nSince 2023.1 is unmaintained we disable and remove\nall non-voting and unused jobs.\n\nAdditionally, a number of jobs on unmaintained/2023.1 required specific\noptions to be run correctly. Some of these were not running before but\nhave since been re-enabled due to\nI10b8b020668c49315723d6f0e464f6da8c76cb56. This change specializes these\njobs for the unmaintained/2023.1 branch.\n\nChange-Id: Id9bb0e1716312cbce9164460e8c6a2df9ce6435b\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2ba6377bfe243d2d8d80158af00f9e3008649b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2ba6377bfe243d2d8d80158af00f9e3008649b1"}]},"branch":"refs/heads/unmaintained/2023.1"},"47a59007c2eb3d8d10f14612481630ac66b28415":{"kind":"REWORK","_number":4,"created":"2026-02-04 18:30:25.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/71/975571/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/975571/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/975571/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/975571/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/975571/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/975571/4"}}},"commit":{"parents":[{"commit":"5975097dd4b021385178bad3815e2cc41aded276","subject":"Specialize v2-dsvm-noop-api job for 2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5975097dd4b021385178bad3815e2cc41aded276"}]}],"author":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-03 20:15:39.000000000","tz":-360},"committer":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-04 18:30:16.000000000","tz":-360},"subject":"Update jobs for unmaintained/2023.1","message":"Update jobs for unmaintained/2023.1\n\nSince 2023.1 is unmaintained we disable and remove\nall non-voting and unused jobs.\n\nAdditionally, a number of jobs on unmaintained/2023.1 required specific\noptions to be run correctly. Some of these were not running before but\nhave since been re-enabled due to\nI10b8b020668c49315723d6f0e464f6da8c76cb56. This change specializes these\njobs for the unmaintained/2023.1 branch.\n\nChange-Id: Id9bb0e1716312cbce9164460e8c6a2df9ce6435b\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47a59007c2eb3d8d10f14612481630ac66b28415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47a59007c2eb3d8d10f14612481630ac66b28415"}]},"branch":"refs/heads/unmaintained/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
