)]}'
{"id":"openstack%2Foctavia~975688","triplet_id":"openstack%2Foctavia~unmaintained%2F2024.1~I7026a5af7b2d97ecf902710a8dea232acd2be6e1","project":"openstack/octavia","branch":"unmaintained/2024.1","topic":"unmaintained-job-cleanup","attention_set":{},"removed_from_attention_set":{"38360":{"account":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"last_update":"2026-02-09 22:42:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7026a5af7b2d97ecf902710a8dea232acd2be6e1","subject":"Clean up jobs for unmaintained/2024.1","status":"MERGED","created":"2026-02-04 18:43:44.000000000","updated":"2026-02-09 22:44:39.000000000","submitted":"2026-02-09 22:42:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975688-unmaintained-job-cleanup","meta_rev_id":"c6695d841265eb9a9ab904f9d4082245d6c3ef65","_number":975688,"virtual_id_number":975688,"owner":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-09 22:42:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2026-02-06 15:03:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2026-02-09 21:01:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2026-02-09 21:01:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-04 20:28:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-06 15:03:28.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2026-02-09 21:01:39.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"9a85fb3e07795729e01ff463ad26e2be90e39b55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-04 18:43:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47e29a93883d8ed3e246c85052e27631fbccfcf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 20:28:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8286bdd4cd6f45f9b44e57c79f155b48\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1499d5798d7c4b16a5696c9313d0b47a : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/458174f1ee7d4d2a9ec7f9fc057c8fe7 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8439e840d1a4d8ca75fe4fb38aec346 : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6029de0c24c74eb59db63ae7a0141aa0 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848b002801804d00ae0723c1af78e9c9 : SUCCESS in 5m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d63432e3909e436a9fbe9830cccd771f : SUCCESS in 4m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/da38bab8659c47a791a62cc6acbfd702 : SUCCESS in 6m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4e6682d16e9247b5872cfbe06b482b0b : SUCCESS in 8m 30s\n- octavia-v2-dsvm-noop-api-unmaintained-2024-1 https://zuul.opendev.org/t/openstack/build/a418c33c7bfc47ebb1dd0c854d85e7d7 : SUCCESS in 1h 08m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-unmaintained-2024-1 https://zuul.opendev.org/t/openstack/build/33a80ee88e2c4908b8ab5d6276d593a9 : SUCCESS in 1h 39m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-unmaintained-2024-1 https://zuul.opendev.org/t/openstack/build/799b5a01241243a0b5fdbe20ac145a2d : SUCCESS in 1h 00m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-unmaintained-2024-1 https://zuul.opendev.org/t/openstack/build/9feeb2871655478cb704a90bdefe07e6 : SUCCESS in 1h 43m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-unmaintained-2024-1 https://zuul.opendev.org/t/openstack/build/1c17e0c5ad1d420f8f08ffa51a9518ab : SUCCESS in 59m 47s\n- octavia-v2-dsvm-tls-barbican-unmaintained-2024-1 https://zuul.opendev.org/t/openstack/build/45c17f140a0c4cc19abde1dc3dfe750f : SUCCESS in 34m 34s\n- octavia-grenade-unmaintained-2024-1 https://zuul.opendev.org/t/openstack/build/382affc5d6744d4ab69219f3d40705d6 : SUCCESS in 38m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ef881f7e8baba53e6573fd3885bc6bd0490a5d5b","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2026-02-06 15:03:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c183c26fbe735e8b90e2c40447611c05eb7150c4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-09 21:01:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7e7d4a8fe79777ceafce636c569aab50557bef3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 21:02:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6165bbc53ab204b52ca99777b59b81f79fba5b26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 22:42:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0c7757b428d4609900820005ab92d99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffb777e3e2594284b205fd87e6526ddf : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/602c9919306442c8831421f1a2b8ca4f : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef9e7db3f610403f93425099ba31a727 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a4d9bcd98d41a6b81efadb1f6d267f : SUCCESS in 11m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f3fe40464b83499784a6de15e9b77128 : SUCCESS in 4m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/172c5670d77d4612a6d3f43703d1ef79 : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api-unmaintained-2024-1 https://zuul.opendev.org/t/openstack/build/a4ffae3ec38740a88db86229222cf114 : SUCCESS in 58m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-unmaintained-2024-1 https://zuul.opendev.org/t/openstack/build/914d28f39a7640598f6fef73f54552ab : SUCCESS in 1h 38m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-unmaintained-2024-1 https://zuul.opendev.org/t/openstack/build/c8450a54b35b48e4a10037a95ab5e923 : SUCCESS in 1h 04m 06s\n- octavia-v2-dsvm-tls-barbican-unmaintained-2024-1 https://zuul.opendev.org/t/openstack/build/a020a1f6f2c641609fcdd9eea6b393ed : SUCCESS in 32m 55s\n- octavia-grenade-unmaintained-2024-1 https://zuul.opendev.org/t/openstack/build/086ff6b2532e4f4c82794a07104a7832 : SUCCESS in 38m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a8a1f7e70b1faa622b1ebf44bd5b939165b1df13","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 22:42:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c6695d841265eb9a9ab904f9d4082245d6c3ef65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 22:44:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f7f3ce0cba04339b40e15792de959f0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/307bb128bd254b3ba0e1cd9300ac98e6 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"435078680eeaf45e11785700d24d7206bf54cb7c","revisions":{"435078680eeaf45e11785700d24d7206bf54cb7c":{"kind":"REWORK","_number":1,"created":"2026-02-04 18:43:44.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/88/975688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/975688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/975688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/975688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/975688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/975688/1"}}},"commit":{"parents":[{"commit":"960efd74ea469193f407f3e4335f65bbef28d302","subject":"Merge \"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\" into unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/960efd74ea469193f407f3e4335f65bbef28d302"}]}],"author":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-04 18:42:48.000000000","tz":-360},"committer":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-04 18:42:48.000000000","tz":-360},"subject":"Clean up jobs for unmaintained/2024.1","message":"Clean up jobs for unmaintained/2024.1\n\nSince 2024.1 is unmaintained we disable and remove\nall non-voting and unused jobs.\n\nChange-Id: I7026a5af7b2d97ecf902710a8dea232acd2be6e1\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/435078680eeaf45e11785700d24d7206bf54cb7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/435078680eeaf45e11785700d24d7206bf54cb7c"}]},"branch":"refs/heads/unmaintained/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
