)]}'
{"id":"openstack%2Foctavia~976490","triplet_id":"openstack%2Foctavia~master~I770a70c534a5be0dd9abeb40ef3f503db63da6de","project":"openstack/octavia","branch":"master","topic":"fix/disable-failed-drivers","attention_set":{},"removed_from_attention_set":{"19298":{"account":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"last_update":"2026-03-03 00:38:38.000000000","reason":"Change was submitted"},"38360":{"account":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"last_update":"2026-03-03 00:38:38.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-03-03 00:38:38.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-03-02 13:09:59.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I770a70c534a5be0dd9abeb40ef3f503db63da6de","subject":"Remove failing drivers from enabled_provider_drivers","status":"MERGED","created":"2026-02-11 15:42:59.000000000","updated":"2026-03-03 00:40:25.000000000","submitted":"2026-03-03 00:38:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"976490-fix/disable-failed-drivers","meta_rev_id":"1f6a81c79307ce926b67caad6804a6104e6815f3","_number":976490,"virtual_id_number":976490,"owner":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-03 00:38:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-03-02 13:19:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-02 13:09:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-02 13:42:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-02 13:42:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":37881,"name":"Wesley Hershberger","display_name":"Wesley Hershberger","email":"wesley.hershberger@canonical.com","username":"whershberger","status":"Support Engineering @ Canonical"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-11 18:22:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-11 21:09:57.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"CC"},{"updated":"2026-02-13 14:03:29.000000000","updated_by":{"_account_id":37881,"name":"Wesley Hershberger","display_name":"Wesley Hershberger","email":"wesley.hershberger@canonical.com","username":"whershberger","status":"Support Engineering @ Canonical"},"reviewer":{"_account_id":37881,"name":"Wesley Hershberger","display_name":"Wesley Hershberger","email":"wesley.hershberger@canonical.com","username":"whershberger","status":"Support Engineering @ Canonical"},"state":"CC"},{"updated":"2026-02-13 14:37:16.000000000","updated_by":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2026-02-13 14:37:16.000000000","updated_by":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2026-02-24 19:20:50.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2026-02-24 22:33:51.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-02-27 08:49:37.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-03-02 13:18:48.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"CC"},{"updated":"2026-03-02 13:19:14.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"30954784b7dc38ffeb53960449037ffb26396967","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-11 15:42:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f13e8c0f6e7c686d94b1ddba0603214bbe7a38b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 18:22:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2c1afe9505444ffaefe4d9f2ab53533\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9e3e68b154d469e8bfa657ea7385037 : SUCCESS in 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06fc02155e1b4142a635c6c4870dcfce : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca084ff2355d4c2a9cbe0135ee8ea4f2 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eae509a752204dd89c19649a4b190b5e : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09ceeee55848420281bc2ee2fc929bfa : SUCCESS in 15m 06s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/fba6daec74fb417faa613a014a4f78a6 : SUCCESS in 4m 26s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/53e8da2a33fa4fb995a214499ddeddf5 : SUCCESS in 7m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/58ed7f3fb4b54c20bed4d574a7e95d2d : SUCCESS in 4m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c47731550cca4474974bb67c9014dbfb : SUCCESS in 7m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f13e1df1cc784d6784a5f8a22c4603bb : SUCCESS in 4m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72b6d2c6b64a4be8aea10bbbee841dcf : SUCCESS in 1h 23m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0d9ece76bd534d31a55176c546829b33 : SUCCESS in 1h 36m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b6c86af533894c0b9a2cc04c36e2ea19 : SUCCESS in 1h 02m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ae1a3b2f58264824b46fa4013c3006ad : SUCCESS in 1h 54m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fb2f756056dc4e0bac9a6c6e52444cb0 : SUCCESS in 1h 12m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c8ffd4749e304e51b3a461d90b6f8fd4 : SUCCESS in 37m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cfd0b7d7ef044d1298ff8ba16b58a04c : SUCCESS in 50m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/530183a122c94c9a83784b40db78cf98 : SUCCESS in 34m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1c8c6abb748d4c1eada7764857318784 : SUCCESS in 43m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fc1da82321b442f6b19754ee5e673e70 : SUCCESS in 2h 38m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c216735ce12b4902837f27f07a4456b6 : SUCCESS in 1h 36m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c800485cb49ebb249a00a21f715683c9b976d6ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-11 23:42:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"985ddb540bec7b866272d229e455ef8d3f697e18","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-11 23:47:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d036b01aae193ad4ffe519ad2c6a647f7d63e401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 02:25:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2816bc1c97784c1bb99cf5887c2d6a56\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5bd336382ef4ef8be2d730371d51a78 : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b139f48be0be416ca5ee30528e13e2b6 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/353fc386730e412cac95a0d8947cbe53 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e1b5e2ab7004730a61ccd5e0bff3cc4 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e2dcd111c98409a927e35918348eb9b : SUCCESS in 12m 54s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/3d4bc6ee7e15416c8fa090b7a5e89b4f : SUCCESS in 9m 04s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/41c4e9d9445b49cebb8d0d089fb25b66 : SUCCESS in 7m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d19ef78f50ee46b08c93fe231e66c99a : SUCCESS in 2m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/03da09c281424ebdbdf48200248c1d37 : SUCCESS in 7m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/22ab2c06aa404cd386fca8f0f41da8f2 : SUCCESS in 7m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b44fee2b042d44358952d74b2fcee14b : SUCCESS in 1h 34m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e56fe20d6244443b906cd584e763d559 : SUCCESS in 1h 38m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/62281234a7ac438aaff23bd531e78734 : SUCCESS in 1h 01m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cfb68bd571654af6b9caf094a946617b : SUCCESS in 1h 43m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8e5dd927082a4aab8c752697c786c2c6 : SUCCESS in 1h 17m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/119ef47464c444a4823ef1d4c2bcf4eb : SUCCESS in 37m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a2bd4fb16be94d3083ce7dcd4c63a394 : SUCCESS in 47m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b64ba844bc804f048d18437922b6b4b7 : SUCCESS in 35m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/32cf654991334784b9d5801d38bf0d43 : SUCCESS in 37m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8d53a317b2ff4f44804c7fe26a516fa2 : SUCCESS in 2h 42m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/faa07d1b4c964282889f454cfd7bd857 : SUCCESS in 1h 36m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"712e511c81605fca93320329d696835bdd2254ea","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2026-02-24 19:20:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3c2c0ff650ccd03340675be6e32ba33c0f68c974","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-24 22:33:51.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b8946da0762eb942dedd808a069c1d41ab769752","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-27 08:48:59.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"eca83bf01c2d2524de932a18a387af4301263778","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-27 08:49:37.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"25f0ed292020a3dc101d49b7863cccec7859d945","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-27 14:53:02.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9e173c39d94747f5a5b271a3967b47ec8316e377","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-27 16:27:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ab6d9f9aae5c147e8555b6275e9d8b38e5cd4dff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 19:18:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86c16eeba1ee4655864b3206b3548adb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4405a9e80eba47ae808cd7cbd47a8429 : SUCCESS in 13m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae41384e8d44baca1cf5737d9b7a68c : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1daa2908dec546aa9ef0de4efc81e62d : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55216348b78a49daa57fc7370cc0ea46 : SUCCESS in 6m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/775c256659364504a5b820549e477aab : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea41a730588841d8a16c4abe573a8052 : SUCCESS in 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40fa3588637f4d91bfd0c70c511e1cdb : SUCCESS in 4m 45s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/83bea1305b8d40b7a2c94b7a606b732a : SUCCESS in 5m 34s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/afbb13ac8bd44a0c891d0ccb86349c64 : SUCCESS in 7m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4cc49b7e5b974765b9091eaf58bebc9b : SUCCESS in 3m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/059800f33e56414c89315afbe12c4a9b : SUCCESS in 4m 40s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3993f3b92dcc4217acb9faf60ff6fd0f : SUCCESS in 6m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cf2dc013acde4f9783e40351f2da36c0 : SUCCESS in 1h 02m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/107a89318f014c2695439e6392cfa263 : SUCCESS in 1h 39m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/81aa2b70de8b471daf03cb2c83d77108 : SUCCESS in 1h 18m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a6eba5187f334488932ec34c55575bd4 : SUCCESS in 1h 46m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ffa9d535cf8e423ea7599e2f1e7622eb : SUCCESS in 1h 22m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e8135be19bf8452fbbd24728434ea139 : SUCCESS in 46m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8d59947a275f4e8aba815ce24ff5eb75 : SUCCESS in 47m 01s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/45714dae7445484c9787cc449019fe2e : FAILURE in 21m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac769ef3fd5b46dc9fbbd9701472d38c : SUCCESS in 36m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/45e4cac86c604412b790a948e34e01e4 : FAILURE in 53m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1f641cf273fb472c8bd8e933431a0150 : SUCCESS in 2h 50m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d5b496336d064f09ae1f88cc27152ef7 : SUCCESS in 1h 41m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/f4762af8b99a4f82b7898697b20e755b : SUCCESS in 1h 37m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e7a8f1f6ff2e4ab0a0bec02cc1dc1059 : SUCCESS in 1h 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6f356248ecc3e6e2a673db70984f6ef0ab809dbf","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-27 21:00:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ae29857b01702fbe4b2c697db1220fc6c11f0bdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 23:42:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b30cf136cc1405dadb840257c46ce27\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02f61019575b4f1f808f283a170d2f5a : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4da73adcc6e48e787b880abfa695fe4 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ea39ec887a442d8be8368bbc1b5432b : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7e9fa989d22401da030215817dde11e : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/11fd5a200fe84c66a4e2c38c012a5cde : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43ada13f1f6440019333d0abb3ace32f : SUCCESS in 9m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/833c470da88b497e941c4e2d153929d7 : SUCCESS in 4m 19s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/5dbf6fe97c624e7fbe4f433e2e7d01e7 : SUCCESS in 4m 16s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/c8eed39a098946afb9c47bd1b3e80f0d : SUCCESS in 8m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/119de0dad5774a51866618941d4b718e : SUCCESS in 3m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/caec3522f4a64eaf9b52fd79ea0d3130 : SUCCESS in 4m 10s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c815d075d9114c2e8e907ead4f94f81f : SUCCESS in 10m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3840ecdf0c174b879af916ef4cb7998e : SUCCESS in 1h 48m 38s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8521457c7cee408b868884b871faa2db : SUCCESS in 1h 38m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4732756435484ae684197bab2a17e299 : SUCCESS in 1h 19m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8f5c265015484676831a169596414c24 : SUCCESS in 1h 47m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6b8f89b9a14a46ce976f89d2ef48d9ca : SUCCESS in 1h 08m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/147a5d25ba89424fb485217ca8227af0 : SUCCESS in 39m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fd0a5faba8af49cea8451f384a2f5942 : SUCCESS in 43m 23s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7dcb45df00f54a1aa68ad9f00a86cb8e : SUCCESS in 45m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0fd934a5ea4044d5bd1591d909fee953 : FAILURE in 56m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bd80480590dc495399e33a005cd2020f : FAILURE in 44m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e1a0b4894f654cb98a1fec737964a62f : SUCCESS in 2h 39m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8511ebea0b7c4925bf5a7af835efa93a : SUCCESS in 1h 38m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/999b3aa9377741c0a4f2de54d09a0250 : SUCCESS in 1h 36m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6ce33e856940409d85843ae6f8d687e0 : SUCCESS in 1h 04m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f1558fa57461757bc20d554fe82452a926fb2171","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-03-02 13:09:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ffe124a96cc4104c086297eb12de6903225c6407","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-02 13:18:48.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"df62e8b4fee0c2e0acdad3ba0a1f0a26cc6edaf7","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-02 13:19:14.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"10600fe32e8b7186757ae93a9ec9ff3a8c64f751","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-03-02 13:42:09.000000000","message":"Patch Set 3: Backport-Candidate+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"954749f3bdfc458596adb6b79c7ffb07f7430de5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 13:45:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c68749f843c1d06efe78dd8bdbb1789509e6df4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 16:29:00.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/96669c59c5354a2489ffd235a332cd27\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bca44844a624094a6e864f117493abb : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be90c4e54d8a4786b41b05adbc0da489 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d0eea1922fb41b694921bdf0fd5cf80 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e52adc0aa2d94fbe98591ff3afc681b3 : SUCCESS in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9e8f03d1dd641f2b2f3d6f95a0f2758 : SUCCESS in 4m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4b6dd8ba299b4d5c825034af53bb6ee7 : SUCCESS in 5m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7f172ebf909049ba86396ae378501dba : SUCCESS in 9m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6a1bc19f2b134b308d4885cb6dc9d0e6 : SUCCESS in 1h 26m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4f4a4e36ef874e7897343b0ed5709560 : SUCCESS in 1h 51m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/714447b340ec4afd839f7e6d81f7c740 : TIMED_OUT in 2h 34m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/241612ea29d940069d39c443f7ad0851 : SUCCESS in 39m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5a649e7684914b84965d3e9f9a278946 : SUCCESS in 44m 13s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/0c742bd5c3df45b28cb78d344f67d374 : SUCCESS in 51m 38s","accounts_in_message":[],"_revision_number":3},{"id":"6c9beb49ed411d0707622fe3afaf6414509f2741","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-03-02 16:43:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0f55e89a4cf7d045f5e3b1015dfb7f927c367c9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 19:37:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f88c31fff7c48319c107811f6dc3796\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/306ccb5d9982481f996ca069201e9b3c : SUCCESS in 16m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85f774bfed4c4d93b5a29eca99032537 : SUCCESS in 6m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a42cd985caf4f84ab6ed2c904a1342c : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b91bf52537e14874aa7cd25b66a77ef2 : SUCCESS in 5m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/235d2ad90f3b446a9af713de97d3b389 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdee57e549c242308fcfc4453a3a2ca3 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67f0a23058df4fea812e1e762d23b871 : SUCCESS in 6m 06s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/8941b291cf6e49b88d0502be4ea6a0d4 : SUCCESS in 3m 52s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/b2bab9abac6f4452a6976b9421bf3d83 : SUCCESS in 5m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ff9e3aa4cc134affbd8d40795cdd8bf9 : POST_FAILURE in 5m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3bf7736a50c940e7bc26461d804adfaa : SUCCESS in 8m 10s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5ef8bbd19f4e49b9a8edd45370a0c6de : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/56c146086f5345d6bdd88ee737284b10 : SUCCESS in 1h 07m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6edb8b87b46945ad88fab05660403b35 : SUCCESS in 1h 45m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3cfc5f5bb7f0469ebe0512b4232c2e25 : SUCCESS in 1h 15m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b93e7a147ad54f1ab96df10f0009c36a : SUCCESS in 1h 50m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/63fc960502a849698aecd78456f58974 : SUCCESS in 1h 15m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8af9389dc3514c9394206464d760ed04 : SUCCESS in 42m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c8bd4fa9d5ff46268679ce0969ecf2b2 : SUCCESS in 46m 56s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b4727cd4df9a427790f26e3af2c67b24 : SUCCESS in 44m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e5e33048860467fb960f6ef9539a894 : FAILURE in 51m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6221c355d68b4b9c9b46bf4cf3c474e9 : FAILURE in 59m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2e2d3581e2e74298939c7bca7a79f6a0 : FAILURE in 2h 15m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ae9fd2879d19476e8403ceeafbde4d78 : SUCCESS in 1h 40m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/babc2449f67c4924be67822add9472ae : FAILURE in 58m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e5a58b6814324fb88e1a080b3e3e6489 : SUCCESS in 1h 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"72d61327d8a9b8a830cbb879c5643de8fa9dfc8c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-03-02 19:56:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4e857d1c620e6781b5c66391403ef68c34254544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 22:44:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34bdc4b06d7c40a7a794380dc82435e1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d5ac88a0a744f2381fb7bea74c2669f : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbe0f06e820e4441a30507bf23d37748 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2725aab951e74a29b14112a38bad01ce : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de325709b09c4dc18374fb4812d680c5 : SUCCESS in 5m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eeb725f1e20d4da0a3de01230f1bbf6e : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37cf9ce66f2647399bcf176a2be1222e : SUCCESS in 13m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7898767e4c1460e82bb101d28ca42ac : SUCCESS in 5m 12s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/e4f03cd7dff649f88a65bb2985f314e8 : SUCCESS in 4m 50s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/5251d6d29a1d41748edc72004395c8fb : SUCCESS in 8m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/af779b3148aa4901a63cb2a6f611bfb6 : SUCCESS in 4m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9ae56483e46e4058a556f1d980b1cbb0 : SUCCESS in 8m 36s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/623a78a675fc48d9957a91bf8e06113d : SUCCESS in 5m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f590aef92dc64dbf921836d99da6f3ce : SUCCESS in 1h 04m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/dbc4eb60a7a74e039272e19a511a6d91 : SUCCESS in 1h 36m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a61c4723655041009d24c99e383053f8 : SUCCESS in 1h 19m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/31b97ea892b2425a85f47cd189ec9436 : SUCCESS in 1h 40m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cd982c222570415389d9ca02ea4dbebf : SUCCESS in 1h 12m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/403ff0ca207d42a39fdf96a3c3b52233 : SUCCESS in 42m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/87d14d4308c04eaf8ab749e2b6f558b0 : SUCCESS in 45m 37s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c6073beeba1e4b88a05a8588f1a40b41 : SUCCESS in 44m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8471c1f1ed994ee794a384b2ad05a1ee : FAILURE in 51m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c09d2efdc246470f9c2323bd393192e2 : SUCCESS in 37m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c27618b1dbf24d8a8f7efc893dee7ae6 : SUCCESS in 2h 46m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/95030757c5ed4892b89784ce41fe47ca : SUCCESS in 1h 38m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/8e60825fb8fe4a97ae4042058c1cdd24 : SUCCESS in 1h 27m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c31d6c1c77c5478ab394667ab6bfbc08 : SUCCESS in 1h 09m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b9aed0879b1e637c89b6c4269f7596bc1da7bf0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 22:44:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c115e98fa7a934630680d2fde046e28483edfee2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 00:38:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb49ccbb573a4777887ebfe6e8104833\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/331cbb195063445cb76efba11aa54481 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10993c526e474feb8e9f1a232af4215d : SUCCESS in 6m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dc73658b1ab741cbbae235fe66945b77 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b9a7c4eb8b447b4b112c7a0e88508ae : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df34f44957e74305aa35bffe5d51f61a : SUCCESS in 4m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/99860429e8fc43acbe214e282d98078e : SUCCESS in 3m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c883162bc37a49b3b6bc2550b8d58cfd : SUCCESS in 3m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/270c4bba9aed46f8a2d59ec014e95b19 : SUCCESS in 1h 48m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/674e823141f54cafa6f5fba2595acbf0 : SUCCESS in 1h 41m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/488488415d314920b2ff856cb8d1a262 : SUCCESS in 1h 19m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c30d0efe64eb4831b9da98a76806f79d : SUCCESS in 41m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ffbbd1046bce495dbc2eedf20e56899b : SUCCESS in 45m 16s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/93b762193f6f445a80ea3f93d8f24161 : SUCCESS in 46m 35s","accounts_in_message":[],"_revision_number":3},{"id":"a1ec062af9c3309a81b7b561ec2252f12f0aaf3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 00:38:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1f6a81c79307ce926b67caad6804a6104e6815f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 00:40:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1de2f216c1e34ed8b79b1d2a3ee79176\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/861f16bd9af246d8874bc539e3bb387c : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c87c5bc0d8d4334bfa9e34f8adc727c : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6acaef1e47b610f1a8cd21d31b7000af960ff116","revisions":{"0252de2a60f4453674e33e65983fca3cd600ed7b":{"kind":"REWORK","_number":1,"created":"2026-02-11 15:42:59.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/90/976490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/976490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/976490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/976490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/976490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/976490/1"}}},"commit":{"parents":[{"commit":"7221e6f682c94302ab58d95645bccb956c23daf1","subject":"Merge \"Fixed missing mock in TestAmphoraAgentCMD.test_main\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7221e6f682c94302ab58d95645bccb956c23daf1"}]}],"author":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-10 23:31:24.000000000","tz":-360},"committer":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-11 15:42:53.000000000","tz":-360},"subject":"Remove failing drivers from enabled_provider_drivers","message":"Remove failing drivers from enabled_provider_drivers\n\nA previous patch (I34341e1aaad1524e3e0834309c5f6897f176af53)\nupdated octavia-api so that drivers which fail to initialize are\nskipped instead of causing the whole API process to error out.\nThe intention was to remove these failed drivers from the list\nof enabled drivers dynamically. While this works in unit tests,\nthey are not properly removed when the api is actually running\nas a WSGI process.\n\nThis patch switches to using set_override on the configuration\nobject, so that modifications to the enabled_drivers is\nproperly persisted for the duration of the process.\n\nDepends-On: I04a580fcb9c53dc2cab98846f7dde704cd317943\nChange-Id: I770a70c534a5be0dd9abeb40ef3f503db63da6de\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0252de2a60f4453674e33e65983fca3cd600ed7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0252de2a60f4453674e33e65983fca3cd600ed7b"}]},"branch":"refs/heads/master"},"bd25421aecd60fbc49068ab2fe7d0ad27460955e":{"kind":"REWORK","_number":2,"created":"2026-02-11 23:42:07.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/90/976490/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/976490/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/976490/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/976490/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/976490/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/976490/2"}}},"commit":{"parents":[{"commit":"7053687659591ff8eca4ce8d0c06720a2ad2ba20","subject":"Merge \"Fix issues related to pkg_resources module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7053687659591ff8eca4ce8d0c06720a2ad2ba20"}]}],"author":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-10 23:31:24.000000000","tz":-360},"committer":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-11 23:42:02.000000000","tz":-360},"subject":"Remove failing drivers from enabled_provider_drivers","message":"Remove failing drivers from enabled_provider_drivers\n\nA previous patch (I34341e1aaad1524e3e0834309c5f6897f176af53)\nupdated octavia-api so that drivers which fail to initialize are\nskipped instead of causing the whole API process to error out.\nThe intention was to remove these failed drivers from the list\nof enabled drivers dynamically. While this works in unit tests,\nthey are not properly removed when the api is actually running\nas a WSGI process.\n\nThis patch switches to using set_override on the configuration\nobject, so that modifications to the enabled_drivers is\nproperly persisted for the duration of the process.\n\nDepends-On: I04a580fcb9c53dc2cab98846f7dde704cd317943\nChange-Id: I770a70c534a5be0dd9abeb40ef3f503db63da6de\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd25421aecd60fbc49068ab2fe7d0ad27460955e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd25421aecd60fbc49068ab2fe7d0ad27460955e"}]},"branch":"refs/heads/master"},"6acaef1e47b610f1a8cd21d31b7000af960ff116":{"kind":"REWORK","_number":3,"created":"2026-02-27 16:27:36.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/90/976490/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/976490/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/976490/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/976490/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/976490/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/976490/3"}}},"commit":{"parents":[{"commit":"7053687659591ff8eca4ce8d0c06720a2ad2ba20","subject":"Merge \"Fix issues related to pkg_resources module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7053687659591ff8eca4ce8d0c06720a2ad2ba20"}]}],"author":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-10 23:31:24.000000000","tz":-360},"committer":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-27 16:07:03.000000000","tz":-360},"subject":"Remove failing drivers from enabled_provider_drivers","message":"Remove failing drivers from enabled_provider_drivers\n\nA previous patch (I34341e1aaad1524e3e0834309c5f6897f176af53)\nupdated octavia-api so that drivers which fail to initialize are\nskipped instead of causing the whole API process to error out.\nThe intention was to remove these failed drivers from the list\nof enabled drivers dynamically. While this works in unit tests,\nthey are not properly removed when the api is actually running\nas a WSGI process.\n\nThis patch switches to using set_override on the configuration\nobject, so that modifications to the enabled_drivers is\nproperly persisted for the duration of the process.\n\nDepends-On: I04a580fcb9c53dc2cab98846f7dde704cd317943\nChange-Id: I770a70c534a5be0dd9abeb40ef3f503db63da6de\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6acaef1e47b610f1a8cd21d31b7000af960ff116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6acaef1e47b610f1a8cd21d31b7000af960ff116"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
