)]}'
{"id":"openstack%2Foctavia~976723","triplet_id":"openstack%2Foctavia~master~Id8e6e0a89b92762ee36d15ea009c80afb8a1d90f","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-03-24 09:52:30.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-03-23 21:13:11.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-03-24 09:52:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id8e6e0a89b92762ee36d15ea009c80afb8a1d90f","subject":"devstack: install/uninstall redis - remove code for SUSE","status":"MERGED","created":"2026-02-13 09:28:45.000000000","updated":"2026-03-24 09:53:37.000000000","submitted":"2026-03-24 09:52:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976723","meta_rev_id":"be0577a44f83dac8b5d6787b27d7a7e729e32a9c","_number":976723,"virtual_id_number":976723,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-24 09:52:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2026-02-17 09:47:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2026-03-24 08:14:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 12:19:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-17 09:47:59.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-02-24 11:00:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2026-03-23 21:13:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REMOVED"},{"updated":"2026-03-23 21:13:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"a7cbeffa53c2aa1c200633f0ab5cf252ee043e90","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-13 09:28:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c2137d923edc769f900834e22b90c694008ca32","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-13 09:31:05.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f12f26146928043c7e76a498a0113e73cc12d279","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-13 09:31:22.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"94ed4e450c3bedb147354db5e9e6c3c394c441d4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-13 09:31:24.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"2842ae90029757dd4c70220949dc9e2346d20114","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 12:19:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d10b654968341b9843ae87f937fc069\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee07cad37d3f478fbd345b564224850b : SUCCESS in 12m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f237cb8c01b64ee5adaa9260d02b14e2 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b908a6e55fe4fb1adb2cada40255e89 : SUCCESS in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dbeb85fbeebd4c6d925530800ed34c88 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4274eb60b777467c895f3e65d99b77b1 : SUCCESS in 7m 23s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/6f9417b3232448e5a8f2c5dec743ce21 : SUCCESS in 3m 31s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/196ff11ccd69475dbcde47f548215dd6 : SUCCESS in 4m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1f69b5570b644383b81fa5ff59748b1e : SUCCESS in 2m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d07c9352f8a447668141df933fac72b9 : SUCCESS in 6m 25s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9bdb8c2b4b7d44368d1abb7a6fa8e21b : SUCCESS in 9m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8857400312e5433797f551578b234e75 : SUCCESS in 1h 11m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/57dd25e1f539405cb71210de02fbca93 : SUCCESS in 1h 34m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/22cfb1192edb437c9c4819d223a7decc : SUCCESS in 1h 12m 34s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/89a2316ec2874c9980f5fd37db2ad060 : SUCCESS in 1h 47m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/42f130592d0043398a1eb4e6f0f329e4 : SUCCESS in 1h 19m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ae915d34dac54d5083cecf5d3ec177cf : SUCCESS in 34m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/29f65a962eaf473ab7aa3e3b92dc7224 : SUCCESS in 46m 19s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/89f2a73c13f846abb9fa03f3e161740b : SUCCESS in 56m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/314fbe6299f74af38c60aabb3012ca81 : SUCCESS in 39m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2d485545746c4f35a5caa95aae2b099f : SUCCESS in 38m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3ebf25dc99c04bb4a8b7fee4b6e27b37 : SUCCESS in 2h 44m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0402b8d8c78f4a31ac7c6719b416df79 : SUCCESS in 1h 37m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b66dc92c3fac0ad7630e5156cb0a1c78dd9eb319","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-17 09:47:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"de434ea2b1ccc28e3aa124edbd79ac2a40864cbf","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-23 21:13:11.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_11628\u003e.","accounts_in_message":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"_revision_number":3},{"id":"c39e2978c37e4ee3d43acc91a6659ab11c1bdd63","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-03-24 08:14:41.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"571d6dfd252af997022de3573527423e721e5679","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 08:15:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"faa31dc810b6c67bdf667174240bd3e208ff1fb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 09:52:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d6f93308bfd4d5099127742f5b05d01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4b4061bb20f45fd948edc5fba22be12 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48d50bc97c814a218be333ce02243416 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38fa3431acae4703bdde53abb629dd9e : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b7273f5537a4c788f88d63eb7ce22fe : SUCCESS in 9m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e5896d10599d4781acb848d6a54f7aa7 : SUCCESS in 1m 56s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a2031d220cf342db83cb81c9b4be5773 : SUCCESS in 5m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ab59824e837547b59882fcdae3ded26a : SUCCESS in 1h 36m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5c97968915d049c4804ffafc8167736a : SUCCESS in 1h 36m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a23a21652d224181a650945ac39b0637 : SUCCESS in 1h 12m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7dc8b2c7f16d40d7975335e77eee97bc : SUCCESS in 40m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/84cd9b17fa0140ca8611dcdb07c6ca96 : SUCCESS in 46m 48s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/75ec4022fc34499d81c634824f9731c0 : SUCCESS in 43m 53s","accounts_in_message":[],"_revision_number":3},{"id":"e775e9fb4ab67a4498866e30f06aaa14edd70af1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 09:52:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"be0577a44f83dac8b5d6787b27d7a7e729e32a9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 09:53:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51aa9c6383144614af24cb89f6ff413d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1912e3c6a4c9411fae31cb569a7483e0 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d17be7daaca3d9fc156a505af7e348c29ec0de1b","revisions":{"e519e23bbb83172995560e411fb0b8c326bd8ccf":{"kind":"REWORK","_number":1,"created":"2026-02-13 09:28:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/23/976723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/976723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/976723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/976723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/976723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/976723/1"}}},"commit":{"parents":[{"commit":"2c4f284c1089b5a6d534f082d50cd520e09214c7","subject":"Merge \"devstack: Remove remaining logic for linuxbridge-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c4f284c1089b5a6d534f082d50cd520e09214c7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 09:28:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 09:28:45.000000000","tz":0},"subject":"devstack: remove code for SUSE","message":"devstack: remove code for SUSE\n\nnow unsupported\n\nChange-Id: Id8e6e0a89b92762ee36d15ea009c80afb8a1d90f\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e519e23bbb83172995560e411fb0b8c326bd8ccf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e519e23bbb83172995560e411fb0b8c326bd8ccf"}]},"branch":"refs/heads/master"},"384355de41ce6fb4b2345f0115c9d437dcf909b4":{"kind":"REWORK","_number":2,"created":"2026-02-13 09:31:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/23/976723/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/976723/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/976723/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/976723/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/976723/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/976723/2"}}},"commit":{"parents":[{"commit":"2c4f284c1089b5a6d534f082d50cd520e09214c7","subject":"Merge \"devstack: Remove remaining logic for linuxbridge-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c4f284c1089b5a6d534f082d50cd520e09214c7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 09:28:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 09:31:05.000000000","tz":0},"subject":"devstack: remove code for SUSE","message":"devstack: remove code for SUSE\n\nnow unsupported\n\nChange-Id: Id8e6e0a89b92762ee36d15ea009c80afb8a1d90f\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/384355de41ce6fb4b2345f0115c9d437dcf909b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/384355de41ce6fb4b2345f0115c9d437dcf909b4"}]},"branch":"refs/heads/master"},"d17be7daaca3d9fc156a505af7e348c29ec0de1b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-02-13 09:31:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/23/976723/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/976723/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/976723/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/976723/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/976723/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/976723/3"}}},"commit":{"parents":[{"commit":"2c4f284c1089b5a6d534f082d50cd520e09214c7","subject":"Merge \"devstack: Remove remaining logic for linuxbridge-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c4f284c1089b5a6d534f082d50cd520e09214c7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 09:28:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-13 09:31:22.000000000","tz":0},"subject":"devstack: install/uninstall redis - remove code for SUSE","message":"devstack: install/uninstall redis - remove code for SUSE\n\nnow unsupported\n\nChange-Id: Id8e6e0a89b92762ee36d15ea009c80afb8a1d90f\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d17be7daaca3d9fc156a505af7e348c29ec0de1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d17be7daaca3d9fc156a505af7e348c29ec0de1b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
