)]}'
{"id":"openstack%2Foctavia~977007","triplet_id":"openstack%2Foctavia~stable%2F2025.2~I7c699b59954acd5fe1c56e51e8b2c55fba09cdd8","project":"openstack/octavia","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-02-24 06:22:50.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I7c699b59954acd5fe1c56e51e8b2c55fba09cdd8","subject":"Fixed missing mock in TestAmphoraAgentCMD.test_main","status":"MERGED","created":"2026-02-16 16:57:54.000000000","updated":"2026-02-24 08:08:59.000000000","submitted":"2026-02-24 08:07:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977007","meta_rev_id":"78da76c9d797ff59c23b7cf2e2f389cf922ef7e9","_number":977007,"virtual_id_number":977007,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-24 08:07:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"all":[{"value":1,"date":"2026-02-23 13:48:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":1,"date":"2026-02-23 15:16:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":2,"date":"2026-02-23 08:28:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":1,"date":"2026-02-24 06:22:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-16 17:19:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-23 13:48:49.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-02-23 15:16:23.000000000","updated_by":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"reviewer":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"state":"REVIEWER"}],"messages":[{"id":"89bcba96ad0e5bc94ab4bf00df919012057cd24d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-16 16:57:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86003f46674824d1623355515b7bc2c8819f4fae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 17:19:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5737835f68a543318b09132dea5665d9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf511079c1c04864b23b3a1e67a114f9 : FAILURE in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb37c02585c54931914cdf345b4c97ec : FAILURE in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/269644e7f47b49d49dca270a18b49cb0 : FAILURE in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f91b0a1b00841da80066860798ef76a : FAILURE in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/93ff621a33a949aaa4cd94100840f31e : FAILURE in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c28f8cd28e944879450f8a2d11a35e4 : FAILURE in 4m 55s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3b84b9548f794f69aa1b6a31a9a7c150 : FAILURE in 3m 51s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/a292236a969c464087b4ab335ae36ea6 : FAILURE in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8695f581a450c627a3de1e3c5315b3e0ea17a507","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-23 08:28:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3dce4c6e4c2e693dde81b581768eb2294ca62137","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-02-23 13:48:49.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9ed139b9048a46de9658d9860cb326dd04d6ebc","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-23 15:16:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0ab3715b39acba3e3f6aa5f2d07d5fe5785c931","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-24 06:22:50.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"53bdd5f25a57a99946965cc752f6785b2b9fa250","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-24 06:23:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe072ce866dda4b4bf13bf1c6a2cb70fb67a25b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 06:35:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a606254cfaf4e5284611480a6d6a6db\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cd5368a26e34de5a767c1e6c73b8027 : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c7109b80b414418906f0d5ef0d230bf : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2dce514ea1a44b28b011c6eefb80513b : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84d7e89e2a6943d0a9fc9961c49b2aee : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/caed139d57a84176b5eb7442fd73fe5c : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d942be8a6e854a47af54885d8aaf3908 : SUCCESS in 9m 51s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5f22baa35d7b4bbb99ff2acc02e83f6b : SUCCESS in 5m 15s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/3588d3d4759147149c406a3cc9163dd7 : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1c7801288e4b0b4890227fb68254e436f0cef1ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 06:35:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7261f9ad70155a4a100bfb11bc62c19496d7d3f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 08:07:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04d1c6963b314b15985b305c9170630b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51599575d87d4cf2b4e18ad7eadb13db : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9984ea183ad34a5b8e4b80f86a6f6bbb : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28cc26ca36b04104a179b1e7b47fad8c : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6402c6e155e14aecb8ee22acde653769 : SUCCESS in 10m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/21f477a3b7624c5e8e41db83cf88592f : SUCCESS in 36m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6c948254754d40aba779adfbdab600d7 : SUCCESS in 1h 31m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/57f72ca5a36b407586844ab8bc4c87be : SUCCESS in 1h 00m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/803c60e8f86d4ae3a0fd3f5869b4da2b : SUCCESS in 33m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b95aada33925470db165f9f3f3560f4b : SUCCESS in 42m 56s","accounts_in_message":[],"_revision_number":1},{"id":"4a93b855d415c2b50390a03aade700799dfa93fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 08:07:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"78da76c9d797ff59c23b7cf2e2f389cf922ef7e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 08:08:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95aa15be2f5b48bf8550e035a01a6f3f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b25981120964be5b8da2e95b552fdd9 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ea1d33848e626c65acc114b85a6be3fa7a3224d","revisions":{"0ea1d33848e626c65acc114b85a6be3fa7a3224d":{"kind":"REWORK","_number":1,"created":"2026-02-16 16:57:54.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/07/977007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/977007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/977007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/977007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/977007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/977007/1"}}},"commit":{"parents":[{"commit":"39e6efbda987e4253e459da8733a989734edb9a4","subject":"Don\u0027t fail if a provider driver cannot be loaded in Octavia API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39e6efbda987e4253e459da8733a989734edb9a4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-12-18 07:02:00.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-02-16 16:57:44.000000000","tz":60},"subject":"Fixed missing mock in TestAmphoraAgentCMD.test_main","message":"Fixed missing mock in TestAmphoraAgentCMD.test_main\n\nThe lack of mock for multiprocessing.Manager in this test seems to\ntrigger random crashes when running the next test on the same process.\nIt also displays a backtrace related to multiprocessing:\n\nException ignored in: \u003cfunction _ConnectionBase.__del__ at 0x7f0cb663b400\u003e\nTraceback (most recent call last):\n  File /usr/lib/python3.10/multiprocessing/connection.py, line 132, in __del__\n    self._close()\n  File /usr/lib/python3.10/multiprocessing/connection.py, line 361, in _close\n    _close(self._handle)\nOSError: [Errno 9] Bad file descriptor\n\nChange-Id: I7c699b59954acd5fe1c56e51e8b2c55fba09cdd8\nCloses-Bug: #2136683\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit 406152a87fd7aebc25f810604b988899941475bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ea1d33848e626c65acc114b85a6be3fa7a3224d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ea1d33848e626c65acc114b85a6be3fa7a3224d"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
