)]}'
{"id":"openstack%2Foctavia~977293","triplet_id":"openstack%2Foctavia~stable%2F2025.1~I04a580fcb9c53dc2cab98846f7dde704cd317943","project":"openstack/octavia","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-02-23 16:46:49.000000000","reason":"\u003cGERRIT_ACCOUNT_38562\u003e replied on the change","reason_account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-02-23 16:46:31.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I04a580fcb9c53dc2cab98846f7dde704cd317943","subject":"Fix issues related to pkg_resources module","status":"MERGED","created":"2026-02-19 08:55:56.000000000","updated":"2026-02-23 18:14:38.000000000","submitted":"2026-02-23 18:13:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977293","meta_rev_id":"e64ece73b838b987da0b9b5729df2449d570bb2e","_number":977293,"virtual_id_number":977293,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-23 18:13:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"all":[{"value":1,"date":"2026-02-23 16:46:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":1,"date":"2026-02-23 16:41:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":2,"date":"2026-02-23 12:46:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":1,"date":"2026-02-23 16:46:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-19 08:55:56.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-02-19 10:57:27.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"CC"},{"updated":"2026-02-19 13:43:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-20 18:55:34.000000000","updated_by":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"reviewer":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"state":"CC"},{"updated":"2026-02-23 16:41:38.000000000","updated_by":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"reviewer":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"state":"REVIEWER"},{"updated":"2026-02-23 16:46:49.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"8df07c5eaa91e0117e154c598f559306ee369c74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-19 08:55:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a17ce51ae09fd08f784edb06f42e2d7e97f96cd","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-02-19 10:08:30.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce9fe2304fcba8aa0ab50b0b8f5acfa92f876108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-19 10:57:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"025325b7e92dbc89b61bfc63879e314aa5fd56e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 13:43:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c98bee7135f14d1780f003aa0fc763f1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79cfcef708d144c1845ebbf8f46e7c12 : SUCCESS in 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e55cfa900a254e54a7898a16e6af16f0 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acff7f9717ed41ef976c5be972af2517 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb29f29cd5f841e780109fb4cf49ce78 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1931afb8d4e41dc9aedee90eea17761 : SUCCESS in 13m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7e9d11f34a048b5818a7a14c7e897f8 : SUCCESS in 5m 51s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3547d8fa9fac497a83b4e8b482af411f : SUCCESS in 5m 04s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/1e1b28b9d11d4fd3a81909441ebb229f : SUCCESS in 6m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/532e39576f2a4c53bd549739533b56ea : SUCCESS in 3m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/da3678b23e7345c39d94159df6d9fcf6 : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c85b89a6a8d94c8390730e1c3a45f1cb : POST_FAILURE in 6m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/461f4a4b735f430b8100a09a96da1c37 : SUCCESS in 5m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/802d9182c04843f0afedab31d98aa59d : SUCCESS in 35m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0c6997d1d31c487393401ba7e47e57e1 : SUCCESS in 1h 26m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d67445e4b3524e978699f64f317c9daa : SUCCESS in 1h 08m 13s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b779bb448d7a4a4a85bb5d1cd1125db2 : POST_FAILURE in 1h 28m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78ee6f4e9a6a4b7195136a840b89b3a2 : POST_FAILURE in 57m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/28167338c7854910b574c894dcf4c314 : SUCCESS in 1h 33m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ff488c3040584fd7a62cf6be9abdb762 : SUCCESS in 1h 05m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/83c55343efb24e1b849b1fbf15efcf93 : SUCCESS in 35m 59s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/68505fba44514f67beafd6a4ed2a8874 : FAILURE in 17m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce066b978594438b8f87649144ca0e52 : SUCCESS in 34m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a9a6db129bd949408ef69dce4c47c053 : SUCCESS in 34m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8cb1e6dccb67449d8c9f648e96246719 : SUCCESS in 2h 45m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1abd464919fa4a34ac89889e31e58597 : SUCCESS in 1h 26m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/476f181132da471d855b7a5a7404fc22 : FAILURE in 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a00987138e75e82f557304d0401091f34f3a39c","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-20 18:55:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4b87e699678d05c51b04689ffdb738a97d70ac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 21:32:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b2a36b42ad3485ebd15d1cba1c0890f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce8c47bae8c944bc9fccf5ec8c89745a : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869b45ecfc9540e8ab76af1f7403ac01 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a71e99a0faf4e2bbf7def29e364f073 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d7826cd25ff64c368894a3054f8842a2 : FAILURE in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62237775ac4b40c8a3026895d4b05018 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e229d079320d4580bc8f143c408c5c19 : SUCCESS in 4m 09s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1c4af9798b7f419c855bafa3c06ac6fa : SUCCESS in 4m 27s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/b10285d380b3416da8601f8389af8dd3 : SUCCESS in 7m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4a61c7c4228a4a2299df1d318c996958 : SUCCESS in 2m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f67b8ea8e06c423e9509b899d98da0ab : SUCCESS in 7m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3c22e9c5d7264c108ced4c86fb0e7f77 : SUCCESS in 7m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/998eeb15204e4eb5aed597506813d14e : SUCCESS in 6m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72c444e52ff745a1adf04d6b88bda610 : SUCCESS in 1h 32m 48s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/061fc232acca47b48b90724e40b10dac : SUCCESS in 1h 30m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6b4b5a41b55a40e791b5222b18174c04 : SUCCESS in 1h 01m 34s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2952f88f68c405f868d38671483667a : SUCCESS in 1h 32m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00aac0493864412080fa8ae32cffa2d1 : SUCCESS in 1h 10m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4d1a52bb3dbe4a4a830b96e0bccf2c0d : SUCCESS in 1h 35m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a27165d59ff044a181f6d7c1b75046c3 : SUCCESS in 1h 10m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/25faf9ac1855439ba869ee314f4948b9 : SUCCESS in 38m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4947b1dc53fa4c8a9f09fa75cbe6d742 : FAILURE in 16m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba324e4b8eb74fb686f62525a8490fd2 : SUCCESS in 36m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/474d082023d44874b5cb6fe6b0fedd05 : SUCCESS in 41m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8be4be9c9ec7460fbf03676d791469f6 : SUCCESS in 2h 35m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f3531b9ea6144700ab5baf4ad324bbd1 : SUCCESS in 1h 25m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf55828cc6194cfaae964b9d8b786e88 : FAILURE in 13m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2221bb5e271f9a5a6a9f43900caf1baebebe6987","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-23 08:29:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"197880f659fbb0c495d598ee4a55a6d50f7b5e85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 10:56:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/685f20ed214e4219abb994d3cae69f87\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8886263800144ca9e747fac2fad61c6 : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dab9a1c2638a4c8eae47398c95876abb : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d490d97234e46ad91612ffe6b7542ef : SUCCESS in 6m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d5aedbf4e554937b1591b76cdbe30f1 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/132bbbf81d7448fbab6b0d62dc46db60 : SUCCESS in 16m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8709e26c25354552a36ce3e0c17a016e : SUCCESS in 4m 53s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9ff8172414804fd1ac7c07d027b33e75 : SUCCESS in 4m 06s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/48005ffad12043a4b6003f829a9b0169 : SUCCESS in 6m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d5ff31609415455a86dadec170c8b947 : SUCCESS in 5m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4f591fa8fd0a486ba1b2e2f8f490a986 : SUCCESS in 9m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2c881692f231464fb1196dc502045891 : SUCCESS in 6m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fa728ff184b34c0eaad49cc66fd02a7c : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f88e35fd89104a2aaaa31ab9f8822822 : SUCCESS in 1h 25m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/70cd9954c855499f92665283f71664ac : SUCCESS in 1h 28m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/37ff5449e50448ca9aa326faca7e17a9 : SUCCESS in 1h 20m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8723c0f2e2e487fb3e24c77f8c705c0 : SUCCESS in 1h 25m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b1ff2ef907e48df9820732961f322fb : SUCCESS in 1h 05m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/000c9c40a40a4a3b901306c7a55f1ad7 : SUCCESS in 1h 35m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9135e7a9063947ce9b72c8e0eff4f5cf : SUCCESS in 1h 11m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/76d098dbf4cb4ff19bf2ce094a160a10 : SUCCESS in 39m 03s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2f01485814ed4bfcb1ac2074593b6f1b : FAILURE in 16m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9317c755fd594c9bacbf650df2adee22 : SUCCESS in 32m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2db06b50077e4375b1962dbbdecbb446 : SUCCESS in 35m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/738bf824fbea4fd79d729a89c2936529 : SUCCESS in 2h 24m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7788d4e5789c42b889670429cffbe9ae : SUCCESS in 1h 30m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2005c8b47d6148cb89329b18a02f974e : FAILURE in 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ec49c68e40bf7e5b0c23779e66df0045d6f46b28","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-23 12:46:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ac6c87d8f9c3b4f8f78ed631a1dfd093714e8f73","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-23 16:41:38.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0595f147d5bd4e17ad5799fff660687333a80d57","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-23 16:46:31.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c6d2c669b0d8e400849e76baf2bf46d03f3a5436","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-02-23 16:46:49.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fd024542ffcd6c87811f7235e5f323a8d14e81c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 16:46:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"811ffee743877ec3c7f2b656111d060c1eb18910","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 18:13:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/371a6295689d48e6ae546e955c19eae6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69906602f80a4d95a54308553f69646f : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd3a20853cc24148b29fbeec2933ef88 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2660ecf3c09f47a998f43d7ea2cda579 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f11d275d00624ebb93a146993985f001 : SUCCESS in 13m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f263a00e6c984bf08c3701c7e9f805b9 : SUCCESS in 3m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cb5bc01c6e8641af9ab5526cd697cd83 : SUCCESS in 3m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/91632bb320014cceaf1093145caef825 : SUCCESS in 6m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/77529193ad144229ba31f863d5d17e76 : SUCCESS in 1h 04m 43s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d3cbb954c876473f932599b1e195b696 : SUCCESS in 1h 25m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4b58b95f914c47e69033face33ea0c8b : SUCCESS in 57m 55s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bc2cc3de43c401a8a5e33dbed1b5a81 : SUCCESS in 1h 24m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/128dc2a1e0af46bfa051216b3e543e4c : SUCCESS in 1h 03m 45s","accounts_in_message":[],"_revision_number":2},{"id":"6462ed8bd2fd45dda7995781faf5a735746990bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 18:13:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e64ece73b838b987da0b9b5729df2449d570bb2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 18:14:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48a48e63ef0e43fc951d79d189e527bb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac55b4ee40254beba7ba93693c3ec464 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a084ada42314e0d81a96986b498fdc3 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c826891aa08f60fadc37cb5c2ce36ef50f79a2c6","revisions":{"d125e8a20ae73f20b5302e2b0ba31bdc00513b61":{"kind":"REWORK","_number":1,"created":"2026-02-19 08:55:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/93/977293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/977293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/977293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/977293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/977293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/977293/1"}}},"commit":{"parents":[{"commit":"d67679c00b1c2e984fde45c298c9aa846c21a366","subject":"Merge \"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d67679c00b1c2e984fde45c298c9aa846c21a366"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-02-09 13:39:29.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-02-19 08:55:42.000000000","tz":60},"subject":"Fix issues related to pkg_resources module","message":"Fix issues related to pkg_resources module\n\nFixes issues with python setuptools removing pkg_resources\n\nUpdate unit tests to use importlib instead of pkg_resources\n\nPin setuptools (for pep8) in tox.ini\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/976865\nChange-Id: I04a580fcb9c53dc2cab98846f7dde704cd317943\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit 8ccea286e7619a73dba24086fc7a4eda9a97c7f7)\n(cherry picked from commit 70fc906832a72d4af1485d792c911af8fc744d71)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d125e8a20ae73f20b5302e2b0ba31bdc00513b61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d125e8a20ae73f20b5302e2b0ba31bdc00513b61"}]},"branch":"refs/heads/stable/2025.1"},"c826891aa08f60fadc37cb5c2ce36ef50f79a2c6":{"kind":"REWORK","_number":2,"created":"2026-02-19 10:57:27.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/93/977293/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/977293/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/977293/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/977293/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/977293/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/977293/2"}}},"commit":{"parents":[{"commit":"d67679c00b1c2e984fde45c298c9aa846c21a366","subject":"Merge \"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d67679c00b1c2e984fde45c298c9aa846c21a366"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-02-09 13:39:29.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-02-19 10:57:06.000000000","tz":60},"subject":"Fix issues related to pkg_resources module","message":"Fix issues related to pkg_resources module\n\nFixes issues with python setuptools removing pkg_resources\n\nUpdate unit tests to use importlib instead of pkg_resources\n\nPin setuptools (for pep8) in tox.ini\n\n\u003c\u003d2025.1: Also pin setuptools for docs job\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/976865\nChange-Id: I04a580fcb9c53dc2cab98846f7dde704cd317943\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit 8ccea286e7619a73dba24086fc7a4eda9a97c7f7)\n(cherry picked from commit 70fc906832a72d4af1485d792c911af8fc744d71)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c826891aa08f60fadc37cb5c2ce36ef50f79a2c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c826891aa08f60fadc37cb5c2ce36ef50f79a2c6"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
