)]}'
{"id":"openstack%2Foctavia~977294","triplet_id":"openstack%2Foctavia~stable%2F2024.2~I04a580fcb9c53dc2cab98846f7dde704cd317943","project":"openstack/octavia","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-02-23 13:06:52.000000000","reason":"\u003cGERRIT_ACCOUNT_38562\u003e replied on the change","reason_account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-02-23 16:46:38.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I04a580fcb9c53dc2cab98846f7dde704cd317943","subject":"Fix issues related to pkg_resources module","status":"MERGED","created":"2026-02-19 08:56:09.000000000","updated":"2026-02-23 18:14:41.000000000","submitted":"2026-02-23 18:13:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977294","meta_rev_id":"fa17dac4e208650162720e03600d07c5f7ef2072","_number":977294,"virtual_id_number":977294,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-23 18:13:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"all":[{"value":1,"date":"2026-02-23 13:06:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":1,"date":"2026-02-23 16:41:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-23 12:46:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-23 16:46:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-19 08:56:09.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-02-19 10:35:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-20 19:38:53.000000000","updated_by":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"reviewer":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"state":"CC"},{"updated":"2026-02-23 12:46:06.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"CC"},{"updated":"2026-02-23 13:06:52.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-02-23 16:41:51.000000000","updated_by":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"reviewer":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"state":"REVIEWER"}],"messages":[{"id":"e7eb00bbc0acd7a97a7295302d5b96a03dbed2a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-19 08:56:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b031c1e3ae40c74f14748f4b722dc3f7b1b6756f","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-02-19 10:08:06.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7bfe16647a03f16aee59717cd27693255b31400f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 10:35:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c643c7cd613e403eb96bb1348d3a9906\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52946436ba474fcdbe510cd9d580f492 : POST_FAILURE in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f18f7dcc8de04f61852b04d9bd2751a6 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b746b4e4eccf4c15a97cc051a81373e7 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8cc15ea5951d4cd49882ec9b7788a12f : SUCCESS in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e554c0e5df6a4102ab799b4735bf3246 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcc2f66e36e6487e960668a7c45b8e81 : FAILURE in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e3de8d25e63467086771aa07a969ae7 : SUCCESS in 6m 07s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/90ebf163790f4e5e9f5fcd3382791c1e : SUCCESS in 4m 30s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/eaf770450fb14acea976886509bdefa8 : SUCCESS in 9m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f3a5c6e4d1374e3b82668860e64d227e : SUCCESS in 4m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cd57587c29c740fabff7661ec8dc1983 : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/189b770e98644416bdccf499fff98d0c : POST_FAILURE in 7m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a5b7e1b059f1464eb0c4b55b0cedd9ae : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/a65d608196d84892958f24da4d04e0cf : POST_FAILURE in 1h 15m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ae4a79cdf5604fc1bdc791c5e18faef9 : POST_FAILURE in 1h 23m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/aa4378bb532340b2935280c108396998 : POST_FAILURE in 1h 08m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/483f817b23674960bee4fa474a781b30 : POST_FAILURE in 1h 34m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ccf2a186b58c41b6b3ce5a2340dc288f : POST_FAILURE in 1h 07m 31s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/2ffee867bd1c4467bafe0cb1a366de8e : POST_FAILURE in 33m 28s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-2 https://zuul.opendev.org/t/openstack/build/fe727ac8155d4ce98c8dd133f3acb2c6 : FAILURE in 23m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b867c57185eb4acf82056fc6947dc43f : POST_FAILURE in 37m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ae0d48c6c882468e872a6ad3b2e37c12 : POST_FAILURE in 39m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-2 https://zuul.opendev.org/t/openstack/build/a9fb76aaafae4ee2a15e4f3989121818 : FAILURE in 40m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-2 https://zuul.opendev.org/t/openstack/build/82f3a6ba5e024877abed401d26cf8e53 : SUCCESS in 1h 34m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-2 https://zuul.opendev.org/t/openstack/build/8f6314eb6bb149dbba3624bff4e14834 : FAILURE in 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c474bc4d711d7bce717767947a4bb3c68c8ead1","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-20 19:38:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5bf1315f9c061959ac0891c9ff9956dc9e240885","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 21:09:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46354a9e566f494a9fa0ac41946d2181\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1e7cd136056436ca36ac93e6d6d7d15 : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97518e846e894f3fa4cb75db42358284 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b17897c4acb74d328f46d0c7c6b52676 : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3568fd83bfe143689456c8cb9e139b09 : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9fb7f403c9424ce5bbf0d7fce1f31fa6 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f409b93bfbc64e6f9a1475e90ec5d0a2 : FAILURE in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b95088473844855b35c8391392e7ed6 : SUCCESS in 4m 42s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/57eaf5b481dd44f98b56dc0f2439fc15 : SUCCESS in 3m 02s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/a711fa5f50e148ae879a6c6e83dfca0b : SUCCESS in 8m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a874c1535b1f4e3385229e53eaa05b92 : SUCCESS in 3m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/88cc9a2b2b894f78995d494b1713d8c3 : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3064bb8da87b4a3e90e92085018291b3 : SUCCESS in 6m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e313f1275f2c462d8fe485c8d605db1d : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/27addbf2fa8045eca6be49043f1dadd8 : SUCCESS in 59m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/c32a307f502b4212973e1388d24c5043 : SUCCESS in 1h 22m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/8d9c4678e16a430084287eb3ab0323db : SUCCESS in 1h 00m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/fe554c5232304fff96f107146e221907 : SUCCESS in 1h 28m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b83c8f63be064d5ebb02c1de2d8f486c : SUCCESS in 1h 04m 50s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/121e5151c1e241329d0d6bcc1f6af2f9 : SUCCESS in 30m 40s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-2 https://zuul.opendev.org/t/openstack/build/bd45ed36ef93423ea2963dff09468d0e : FAILURE in 16m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/1b07474eeed64dd5beb1a7cbde857409 : SUCCESS in 27m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4c921bccdb0f4750a5310bf9fc4052b5 : SUCCESS in 29m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-2 https://zuul.opendev.org/t/openstack/build/45b7e158122f477397057917933d3683 : FAILURE in 40m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b55f6a3575674bde8e530bbc732a8697 : SUCCESS in 1h 21m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-2 https://zuul.opendev.org/t/openstack/build/12047ce6c4fc48269bfabca6eef16b20 : FAILURE in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d85e868460f6c9721e4499ba8bd99081a609ed4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-23 12:46:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"68f22d547cd80746b20bb5e9d78c042c9487c860","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-23 12:46:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4b052ef5a781e4b03b0b9a773d426d8d59375412","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-02-23 13:06:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"af0fd23740b72d53d1c57850d145f48c49f252cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 14:13:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7b3634523f8449d8805b0dce6fdf354\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/303af78e8981403395f8328c8d38f621 : SUCCESS in 14m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c4a610bb8d4db8b35add6314524187 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3489d3bdb52449029cc8c5bb200ae234 : SUCCESS in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce6f57d89aad442aa9589a65f688a2dd : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/639a999462d64c8e8cc281910adef397 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2b639f4c13b4befa9b6eac24a7b25ed : SUCCESS in 11m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebe8046a8888435b9fcf06821bf7e6dc : SUCCESS in 5m 38s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bacbc96914c14748ad53f9f6d8d24cc1 : SUCCESS in 5m 17s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/8688ed529cfc4c89b2b3f430f9d4cd67 : SUCCESS in 8m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4d96b79e5bd849f89edb3e2d2667ed8e : SUCCESS in 4m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/966f7b604d4940609724cfc437799bd0 : SUCCESS in 4m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/15b59eed727f445998fbf1d66d91db49 : SUCCESS in 8m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2367383a5f274b1baa33b235a7434b2a : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/2e55a69e0d9d453c8a6616ac2ee3ce03 : SUCCESS in 1h 09m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/fcae51d09f494d3596241737ef7db46f : SUCCESS in 1h 21m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b3b9a025365f4641aebfed3afd771826 : SUCCESS in 1h 09m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/e8d830eaea724ff8a5afd3276a6b566c : SUCCESS in 1h 24m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7c69884c2b2142279210b49907236c30 : SUCCESS in 1h 04m 40s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/564721f14b2743dfb494c9d57bdc1ff7 : SUCCESS in 38m 17s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-2 https://zuul.opendev.org/t/openstack/build/70bb472211194ba185f9d041c5685754 : FAILURE in 19m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7902030257f442aca68e140c43410da6 : SUCCESS in 35m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-2 https://zuul.opendev.org/t/openstack/build/90bfae7a5d864158bc998f6388c7b1ed : SUCCESS in 34m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b37f96b31474464f95d04c33d6073589 : FAILURE in 44m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-2 https://zuul.opendev.org/t/openstack/build/665fed6799bd45f3921fcb6da353d93d : SUCCESS in 1h 25m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-2 https://zuul.opendev.org/t/openstack/build/befb6357085f46e0bfd5f7269dfbbb38 : FAILURE in 19m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d35a8c75468438362562db8deb0c21a8c4d9d10b","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-02-23 16:41:51.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d2cadc4fc4f102d6eec6d95bc88e93a965b0cd2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-23 16:46:38.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5187ca15b201603aef73e0828e10f6213c88ea65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 16:47:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1e297943bc2bafb851d20ffaf9a988b5f22f52b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 18:13:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5748d0d0d5c24b198f6ec84e45196c23\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f000ac461f1242d0871bd27d5f00c458 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b032c1bc28694e83a3bb0c5493ec1c1d : SUCCESS in 6m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b08eadde711f4c7d8c6f72157fc8a19f : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c61d019aed64b09b5306f676fd56a03 : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07beed9ca6944a118b858e407c074267 : SUCCESS in 4m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/18e8aaf39a4f4e07aff90278ca54098f : SUCCESS in 4m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ce5673f110404afba40b14d13a0e6036 : SUCCESS in 6m 03s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/9ab9b54c49ab483a8445ffe8c96af4ba : SUCCESS in 57m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/8e7451c0ac6d47d3bcdb295a033f4371 : SUCCESS in 1h 23m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/db02f171765f45469cd4e5a6ad60bb6c : SUCCESS in 55m 07s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/50e68994f2e1415b9af16e14a0b4d41c : SUCCESS in 29m 37s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/de1881764a4842d28c05fa16b5c72269 : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":2},{"id":"a6c64c614de7c757f532b84b35895435da02907f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 18:13:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fa17dac4e208650162720e03600d07c5f7ef2072","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 18:14:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23e42954b72d48f7ac25cc1950872c46\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce9d47aa1d574d21a3ef1e35fbcdc6c5 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/386f27a4aef24e06a69f219590e41542 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a680d59e6092bebe448c00d46e4a2096a70a374b","revisions":{"33b0aea8fde295dac1fdc5dfc06b91c463f9c630":{"kind":"REWORK","_number":1,"created":"2026-02-19 08:56:09.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/94/977294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/977294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/977294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/977294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/977294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/977294/1"}}},"commit":{"parents":[{"commit":"ac102c9a4dfd60deac936310952cb34a938df595","subject":"Merge \"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac102c9a4dfd60deac936310952cb34a938df595"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-02-09 13:39:29.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-02-19 08:56:03.000000000","tz":60},"subject":"Fix issues related to pkg_resources module","message":"Fix issues related to pkg_resources module\n\nFixes issues with python setuptools removing pkg_resources\n\nUpdate unit tests to use importlib instead of pkg_resources\n\nPin setuptools (for pep8) in tox.ini\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/976865\nChange-Id: I04a580fcb9c53dc2cab98846f7dde704cd317943\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit 8ccea286e7619a73dba24086fc7a4eda9a97c7f7)\n(cherry picked from commit 70fc906832a72d4af1485d792c911af8fc744d71)\n(cherry picked from commit d125e8a20ae73f20b5302e2b0ba31bdc00513b61)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/33b0aea8fde295dac1fdc5dfc06b91c463f9c630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/33b0aea8fde295dac1fdc5dfc06b91c463f9c630"}]},"branch":"refs/heads/stable/2024.2"},"a680d59e6092bebe448c00d46e4a2096a70a374b":{"kind":"REWORK","_number":2,"created":"2026-02-23 12:46:06.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/94/977294/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/977294/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/977294/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/977294/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/977294/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/977294/2"}}},"commit":{"parents":[{"commit":"ac102c9a4dfd60deac936310952cb34a938df595","subject":"Merge \"Don\u0027t fail if a provider driver cannot be loaded in Octavia API\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac102c9a4dfd60deac936310952cb34a938df595"}]}],"author":{"name":"Richard Cruise","email":"rcruise@redhat.com","date":"2026-02-09 13:39:29.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-02-23 12:45:48.000000000","tz":60},"subject":"Fix issues related to pkg_resources module","message":"Fix issues related to pkg_resources module\n\nFixes issues with python setuptools removing pkg_resources\n\nUpdate unit tests to use importlib instead of pkg_resources\n\nPin setuptools (for pep8) in tox.ini\n\n\u003c\u003d2025.1: Also pin setuptools for docs job\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/976865\nChange-Id: I04a580fcb9c53dc2cab98846f7dde704cd317943\nSigned-off-by: Richard Cruise \u003crcruise@redhat.com\u003e\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit 8ccea286e7619a73dba24086fc7a4eda9a97c7f7)\n(cherry picked from commit 70fc906832a72d4af1485d792c911af8fc744d71)\n(cherry picked from commit d125e8a20ae73f20b5302e2b0ba31bdc00513b61)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a680d59e6092bebe448c00d46e4a2096a70a374b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a680d59e6092bebe448c00d46e4a2096a70a374b"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
