)]}'
{"id":"openstack%2Foctavia~978204","triplet_id":"openstack%2Foctavia~master~Ib0393fab6ceb932f3ae5d93889b6d6f08469266e","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2026-03-24 10:28:01.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-03-24 08:48:28.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Ib0393fab6ceb932f3ae5d93889b6d6f08469266e","subject":"Test LB API giving HTTP 500 when get_subnet fails","status":"MERGED","created":"2026-02-27 14:30:43.000000000","updated":"2026-03-24 10:29:14.000000000","submitted":"2026-03-24 10:28:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978204","meta_rev_id":"94fef3a7e9710d8b2ef19841526d05290eb6e773","_number":978204,"virtual_id_number":978204,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-24 10:28:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-03-02 10:09:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":2,"date":"2026-03-24 08:48:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-02 23:10:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":1,"date":"2026-03-24 08:48:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-27 14:45:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-02 10:09:44.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-03-02 23:10:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-02 23:10:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"5df003789b47cc50efe2ac3c4d67d7076cb2a13c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-27 14:30:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6700297604508f561998fe1046731e07e22a3f23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 14:45:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74b2c93373be44a9abb0064a86d767fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/424f0963315c4fbcab7334a30b78e9c6 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f6f1a93c27741b69f400970b5a5830c : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba87ad6b37304c8999f5f1270836c29c : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8e10fb4084f4c95a11282823156f769 : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ed00206440944ac9a01cc885c6f1984f : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/791c6f1b1c6848598c2822fe11a26a4b : SUCCESS in 12m 35s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/424e20f5938a448b94d54c76bdf39d3e : SUCCESS in 6m 36s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/ff95b8d48d934896b88d98b687240d8b : SUCCESS in 4m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6ad5a27176504e588e989f737335714b : SUCCESS in 9m 53s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9bd4bca14c494ef2ae48536125769190 : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8a78274c0227a618e62779e3471e8beabb40852f","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-02 10:09:44.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e666309f90409d5f7a7948cd4fe55aeb176995c8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-02 23:10:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d70d711c86bd7b8f7b50c714043f0a31f9db3954","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-03-24 08:48:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cff4421d129b4913ae8ef530083c23b57c9d76e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 08:49:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"669df36b7e3478c85984852a08108bf56d877342","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 10:28:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1861ad2be8547d6b1983f0e17925fc6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/300c9935aff545db901d8598dc12d703 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b17f97f81f3b41bebad1a06525e35994 : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ea502a6643ae4e8a8e6e825bb985242c : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d5260c0d479407b85706da1ae539cad : SUCCESS in 13m 25s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0cfc4709c8b1434baa02fe59577f148e : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b06b35d954ab44fcb30c73526b41189a : SUCCESS in 1h 24m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3fadd464535f4fd5bc030a5c50445da5 : SUCCESS in 1h 38m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/96c7072df05d4a04b26bbc76ccfd6968 : SUCCESS in 1h 32m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/38d1fadc09184145a7d1d1050466654b : SUCCESS in 53m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fbe45093fd65495a8616cce7c83289ad : SUCCESS in 52m 10s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5dfe519cf20a45cba2e117744c4f49ee : SUCCESS in 44m 50s","accounts_in_message":[],"_revision_number":1},{"id":"73f63453e2c127895c207aa9d01e5c882d6d8fe9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 10:28:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"94fef3a7e9710d8b2ef19841526d05290eb6e773","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 10:29:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cbbc77a7b094b8cb13839e5a744bbc8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30aaa3019a564b31992e647745c951ea : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cacd6b56a69954744887a373c527c4e06a287b74","revisions":{"cacd6b56a69954744887a373c527c4e06a287b74":{"kind":"REWORK","_number":1,"created":"2026-02-27 14:30:43.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/04/978204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/04/978204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/978204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/978204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/978204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/04/978204/1"}}},"commit":{"parents":[{"commit":"225f28e469ef4d848cbfa5ae0eec303f46ffa8be","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/225f28e469ef4d848cbfa5ae0eec303f46ffa8be"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-02-27 13:55:43.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-02-27 14:05:59.000000000","tz":60},"subject":"Test LB API giving HTTP 500 when get_subnet fails","message":"Test LB API giving HTTP 500 when get_subnet fails\n\nIf the get_subnet method in the network driver fails\nwhen we try to create a load balancer with a VIP\nnetwork and without a subnet ID specified we will get\na HTTP 500 error from the API.\n\nThis checks that this error condition is true\nand will be fixed by a future patch.\n\nChange-Id: Ib0393fab6ceb932f3ae5d93889b6d6f08469266e\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cacd6b56a69954744887a373c527c4e06a287b74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cacd6b56a69954744887a373c527c4e06a287b74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
