)]}'
{"id":"openstack%2Foctavia~978771","triplet_id":"openstack%2Foctavia~stable%2F2025.2~I770a70c534a5be0dd9abeb40ef3f503db63da6de","project":"openstack/octavia","branch":"stable/2025.2","topic":"fix/disable-failed-drivers-stable/2025.2","attention_set":{},"removed_from_attention_set":{"38360":{"account":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"last_update":"2026-03-19 09:35:13.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-03-19 08:00:45.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I770a70c534a5be0dd9abeb40ef3f503db63da6de","subject":"Remove failing drivers from enabled_provider_drivers","status":"MERGED","created":"2026-03-03 15:20:51.000000000","updated":"2026-03-19 09:36:20.000000000","submitted":"2026-03-19 09:35:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978771-fix/disable-failed-drivers-stable/2025.2","cherry_pick_of_change":976490,"cherry_pick_of_patch_set":3,"meta_rev_id":"e9225487bda6e201a1b2a401e2056680578d0db8","_number":978771,"virtual_id_number":978771,"owner":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-19 09:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-03-04 17:31:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2026-03-10 11:31:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":2,"date":"2026-03-19 08:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-04 19:21:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":1,"date":"2026-03-19 08:00:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-03 17:05:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-03 18:08:06.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2026-03-04 17:31:31.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2026-03-04 19:21:48.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-03-04 19:21:48.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-03-04 19:21:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-10 11:31:44.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"bd023e82d717fc25c80eee07271b16ffb467bc3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-03-03 15:20:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d1e671cdbd8269942c59e49fc902cf0a126d341d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 17:05:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59dab9bebc9446b285bf3c22fc944694\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8443ed00a8c44f8598e7b1d96fb01346 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70890623e4a4bde82c8341f54038c41 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee49d5dac83e479caeadcfd5661869a4 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/682c1d71894c4658a773a3b1c24a566e : SUCCESS in 6m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d2c8e0d976af46e391acd15346584c12 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67ba025e84ba466eaf3bee77f9fddc34 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c62665484e94fe28c81b00680f2c5e2 : SUCCESS in 4m 08s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1ae3028186dc4b8c9fa87dd38c67ca73 : SUCCESS in 6m 14s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/dfbefa6449da4b3da5ecc11efde3766c : SUCCESS in 7m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1cd19c581dbe41dba5e6af8f08d60a6e : SUCCESS in 2m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cfafb6928b384dd2a6e6830b2ef46c70 : SUCCESS in 7m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7afba2cee03c4057ae012e44b817aafd : SUCCESS in 8m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/95d268ea872d44d8b8bf5a341c6bcf40 : SUCCESS in 58m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/37510d83763542b89ffbf23e0fc99768 : FAILURE in 26m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f9ec74e4a97b43348b9fa6a4d335b9fb : FAILURE in 26m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5e59fe581dd54e54956bb7863e3830a4 : SUCCESS in 1h 42m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e0567af5f52640559b35ede534b664b8 : FAILURE in 22m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/967e34aadf6140268ca3015eae17d538 : SUCCESS in 36m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6617a20d89e5410699c8b8c4b7d9c10c : SUCCESS in 46m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/05b70011f5f148d4970affbddbd8ecaf : FAILURE in 25m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7695fb9328e446c987b9de981ac105ff : FAILURE in 26m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4551d80209004a85b03bfadd00729e30 : FAILURE in 31m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8fa8ad48d9af485dbff2fcd171ed8c32 : FAILURE in 26m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30c6a8a785f6cd683908a8431eacc48db0caca05","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-03-03 18:08:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba6612565ebf14f97ba323f213324dde46d0beb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 20:53:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7175e5407c9a4411bb50ac5d921b9824\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/343a7fa5fc4d4df4aba4cb9fb2b3b83b : SUCCESS in 16m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd4482e58f694efeb26ae5a879f0d61d : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e816667bb1be4d9aaae5165f45400592 : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b7feec737b5420f8f514fb194c9c983 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bac3ec095cd3408899e6210fbbfba904 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d31f524813884d45b533f2c39443c872 : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6dbc21e3232423fabe3ddcc2730cf3d : SUCCESS in 5m 09s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/dbf3c395e3624f2685bacc793a31aec9 : SUCCESS in 5m 12s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/4078b425dfd144f9a13961f32813b3ff : SUCCESS in 6m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/14a70bb75de64859ba301bb4022af2fb : SUCCESS in 2m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9429b69c6b764c3f8bf2336f69cbfb4c : SUCCESS in 6m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0d932bb909e84a9e889e709e883a0636 : SUCCESS in 5m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a7c30b7da45a42b794d08ba229469eef : SUCCESS in 1h 26m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4f22c20c2402491d8b3c4438fb64e660 : SUCCESS in 1h 36m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/147b734fbca24db981263df5c0777775 : SUCCESS in 1h 17m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/23aa738b69414fe9963d29ccd13a353f : SUCCESS in 1h 39m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6b57e0772509473db4120190f9c19b62 : SUCCESS in 1h 11m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/862b7345ee384cc8b138b9d35fa4c8cc : SUCCESS in 39m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e3f5d81fa83b4a0fb7b73e1fe9bb2209 : SUCCESS in 48m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8f51e43802b46d08673b8c12cc748e5 : SUCCESS in 36m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/447774534ff64f959760370e3371383a : SUCCESS in 37m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6fcca1e19b2d402ca08121fa782847cd : SUCCESS in 2h 44m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0247cccfc4ea471bbd2370c3c105a482 : SUCCESS in 1h 35m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e11febce44ac763a42a315c1f2aa774aedead75","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2026-03-04 17:31:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ed8d64d95b7b663bf6718703cccd7a8918c9bed1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-04 19:21:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"34d8cd49d2517d728a40c91d77115ebe86b4a31c","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-10 11:31:44.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83af0ceb30fa9c1a2b35eae89071d8720cf80785","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-03-19 08:00:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea8152a3fe3871af51c208171b28f98d86dac81b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 08:01:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13afb66c96122cdf9dc01de60606c55afe53a901","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 09:35:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c15da9112094c9a9a6417d0979a6c1b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ca0f3b4972241f8a5b21848e0d9291f : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8694d575c164bdaa108548a5505b5ea : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c2b0b35e13f84ab598a343f33dd0bbcd : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40c9b8dd259e4544be410d21df4e87fe : SUCCESS in 13m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b48f7b745c3d4140b7f8d4180766ab8e : SUCCESS in 5m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fcbf9edaba34413498157f05b9a64634 : SUCCESS in 2m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/93d9cbdfc43b4411bac1e23c645bfe3f : SUCCESS in 8m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/27bb335585a9442988629a666f1fa8d5 : SUCCESS in 1h 28m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1fb17d2ba6904a97becb74c03f572eee : SUCCESS in 1h 31m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3c993de9dbde415581350fcf0526c0a5 : SUCCESS in 59m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9f6b43d9e01045be979d680176e75c81 : SUCCESS in 33m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/61072192cfcc4f0abeece5675176335f : SUCCESS in 42m 43s","accounts_in_message":[],"_revision_number":1},{"id":"368f211f5c4a0a383371cf7584171b1df00eb7a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 09:35:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e9225487bda6e201a1b2a401e2056680578d0db8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 09:36:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/126c7dfd20204483b672abb2575f6886\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c74db8151c83436abb00e744b4599822 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/168414e6c22645ebaefdd628f0e4eb30 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e8376a0bf4eb4b7a6f4306ff9095e04ae5ff4f7","revisions":{"5e8376a0bf4eb4b7a6f4306ff9095e04ae5ff4f7":{"kind":"REWORK","_number":1,"created":"2026-03-03 15:20:51.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/71/978771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/978771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/978771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/978771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/978771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/978771/1"}}},"commit":{"parents":[{"commit":"8ef93fbcb6fdaf20b1eed825f384c7e3fde92088","subject":"Merge \"Fix database connection check\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ef93fbcb6fdaf20b1eed825f384c7e3fde92088"}]}],"author":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-10 23:31:24.000000000","tz":-360},"committer":{"name":"Zachary Mark Raines","email":"zachary.raines@canonical.com","date":"2026-03-03 15:20:51.000000000","tz":0},"subject":"Remove failing drivers from enabled_provider_drivers","message":"Remove failing drivers from enabled_provider_drivers\n\nA previous patch (I34341e1aaad1524e3e0834309c5f6897f176af53)\nupdated octavia-api so that drivers which fail to initialize are\nskipped instead of causing the whole API process to error out.\nThe intention was to remove these failed drivers from the list\nof enabled drivers dynamically. While this works in unit tests,\nthey are not properly removed when the api is actually running\nas a WSGI process.\n\nThis patch switches to using set_override on the configuration\nobject, so that modifications to the enabled_drivers is\nproperly persisted for the duration of the process.\n\nDepends-On: I04a580fcb9c53dc2cab98846f7dde704cd317943\nChange-Id: I770a70c534a5be0dd9abeb40ef3f503db63da6de\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n(cherry picked from commit 6acaef1e47b610f1a8cd21d31b7000af960ff116)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e8376a0bf4eb4b7a6f4306ff9095e04ae5ff4f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5e8376a0bf4eb4b7a6f4306ff9095e04ae5ff4f7"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
