)]}'
{"id":"openstack%2Foctavia~978901","triplet_id":"openstack%2Foctavia~stable%2F2025.1~I770a70c534a5be0dd9abeb40ef3f503db63da6de","project":"openstack/octavia","branch":"stable/2025.1","topic":"fix/disable-failed-drivers-stable/2025.2-stable/2025.1","attention_set":{},"removed_from_attention_set":{"38360":{"account":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"last_update":"2026-03-19 09:35:16.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-03-19 08:01:01.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I770a70c534a5be0dd9abeb40ef3f503db63da6de","subject":"Remove failing drivers from enabled_provider_drivers","status":"MERGED","created":"2026-03-04 16:24:22.000000000","updated":"2026-03-19 09:36:26.000000000","submitted":"2026-03-19 09:35:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978901-fix/disable-failed-drivers-stable/2025.2-stable/2025.1","cherry_pick_of_change":978771,"cherry_pick_of_patch_set":1,"meta_rev_id":"7ba9879b2c9bdd0f1ef7fe52d632b56ea3d63cb5","_number":978901,"virtual_id_number":978901,"owner":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-19 09:35:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-03-04 17:31:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2026-03-10 11:31:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":2,"date":"2026-03-19 08:01:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-04 19:22:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":1,"date":"2026-03-19 08:01:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-04 17:31:40.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2026-03-04 18:51:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-04 19:22:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-04 19:22:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-03-10 11:31:24.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"f53a4e06542cff1b580c893c72f11da1a2d76169","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-03-04 16:24:22.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"6576b7cb20ab6c098cae6db8975c349b02debcc1","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2026-03-04 17:31:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0ca1b027334cd20b30927d505ca9dba9f513e097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 18:51:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/448ce3d634544eb48a1d46a1e821b842\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04c3da480cba4b49bcdbe4abef4f030c : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a42b3c60579b46cfa7d048f25c8b5a5d : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5701b34ccd974ffa90deaec41ebb2d36 : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c79ec6fa70d44fe8a968bb217373203b : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaa802a969bf4215892ed70f80e681f1 : SUCCESS in 13m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9ce2ae399774186a19b470cc19ac2c9 : SUCCESS in 4m 11s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4a7ee0727eb24072956354aeb7582f90 : SUCCESS in 4m 37s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/e247ce0a870b46cdb88ee36f76071d8f : SUCCESS in 5m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e4d0ffdabab04b65ac54303f6653c1a7 : SUCCESS in 3m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aaedb7ef3f86432f8e7f7c661da7a9d6 : SUCCESS in 8m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b58b9c1dc93d4eb1897c48a5141bdecd : SUCCESS in 7m 45s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/15f7e8976a044525b959d452c7bd744e : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5d5b08252f0b40efb994d43940e8f274 : SUCCESS in 1h 46m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3a5b04762efd40e3b855a7a29aa47b87 : SUCCESS in 1h 29m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/489f86f17aa74a1884c3cff44d7fde10 : SUCCESS in 1h 04m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bad1f16860b4448f98d82b808980217e : SUCCESS in 1h 32m 16s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abcdb60d551e44a2b6378f32eb890c6e : SUCCESS in 1h 13m 51s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/07c65b7e5de64a55b62d875f7a863219 : SUCCESS in 1h 36m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/51017bdfe1d44cf69f6f85121c11dd5a : SUCCESS in 1h 34m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d3be80cfa0a64cda8a4f9f5ddc257c76 : SUCCESS in 43m 18s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5e89e1d8701a443f9c7fa3938e3fe8d9 : FAILURE in 38m 19s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3aad2015d1eb42d1b6e0c21411c42569 : FAILURE in 18m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e14c767eb88d416e8b78ba7cadf7a2d7 : SUCCESS in 39m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/630a417bdfb544829364305bf6135848 : SUCCESS in 38m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6fa2d99561824fd09a724d5d686d9e75 : SUCCESS in 2h 25m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/46d8c0122aa14e609e984d4e249a8375 : SUCCESS in 1h 32m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e869d57237764dce8de292342d8e9a5f : FAILURE in 15m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45528f86efd00136a02d772f51de1b03c0b80e75","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-04 19:22:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cb63c602b2741d7d54d52503442e57194df62030","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-10 11:31:24.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7020be4a0d55305598a0333c6959e0299a119f86","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-03-19 08:01:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"53ad70d58f68fce37ba0e4fe5460d6379ea53209","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 08:01:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b36dee075832edf5667b0d8d668af7618089c532","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 09:35:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81c740088cd141dba235f1a84813960a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58a5b878f71d4980a18a59dc7850ffbf : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8048ec3832a447a9714b8ea27b633c6 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c07d6e0e7b0474f96177674379dc7bb : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac70cbb9adb34dabb87fdc436830f70a : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e13c1a7e6ea4c698cbfb9f495d477eb : SUCCESS in 4m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6354ea5609e848ab94fad6039dad3420 : SUCCESS in 3m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/769eb6ca8f004178818d51bcfa08a6de : SUCCESS in 8m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e42050a33ed74018834c23f25052dcbf : SUCCESS in 56m 11s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8b3dd358dee14c33862c4ef744feee38 : SUCCESS in 1h 31m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d631b890f04145fbb044b34c37037528 : SUCCESS in 54m 16s","accounts_in_message":[],"_revision_number":1},{"id":"858f277482c8b73d4d69bbd369d08d5ba9c5ce2e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 09:35:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ba9879b2c9bdd0f1ef7fe52d632b56ea3d63cb5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 09:36:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9592a68cd176429986f59c2d63e5d3af\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7327ad5e8d61429aa58c23aa971b965b : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58136338bde84974953ee6ae4d4924df : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ac3a9de0de662d1351c15afb39c2fbe28502279","revisions":{"3ac3a9de0de662d1351c15afb39c2fbe28502279":{"kind":"REWORK","_number":1,"created":"2026-03-04 16:24:22.000000000","uploader":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"ref":"refs/changes/01/978901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/978901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/978901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/978901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/978901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/978901/1"}}},"commit":{"parents":[{"commit":"66e191aa0b4371928331ba3fe4a477ffc7bfa24b","subject":"Merge \"Update diskimage-create.sh to use noble as default\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66e191aa0b4371928331ba3fe4a477ffc7bfa24b"}]}],"author":{"name":"Zachary Raines","email":"zachary.raines@canonical.com","date":"2026-02-10 23:31:24.000000000","tz":-360},"committer":{"name":"Zachary Mark Raines","email":"zachary.raines@canonical.com","date":"2026-03-04 16:24:22.000000000","tz":0},"subject":"Remove failing drivers from enabled_provider_drivers","message":"Remove failing drivers from enabled_provider_drivers\n\nA previous patch (I34341e1aaad1524e3e0834309c5f6897f176af53)\nupdated octavia-api so that drivers which fail to initialize are\nskipped instead of causing the whole API process to error out.\nThe intention was to remove these failed drivers from the list\nof enabled drivers dynamically. While this works in unit tests,\nthey are not properly removed when the api is actually running\nas a WSGI process.\n\nThis patch switches to using set_override on the configuration\nobject, so that modifications to the enabled_drivers is\nproperly persisted for the duration of the process.\n\nDepends-On: I04a580fcb9c53dc2cab98846f7dde704cd317943\nChange-Id: I770a70c534a5be0dd9abeb40ef3f503db63da6de\nSigned-off-by: Zachary Raines \u003czachary.raines@canonical.com\u003e\n(cherry picked from commit 6acaef1e47b610f1a8cd21d31b7000af960ff116)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ac3a9de0de662d1351c15afb39c2fbe28502279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ac3a9de0de662d1351c15afb39c2fbe28502279"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
