)]}'
{"id":"openstack%2Foctavia~980907","triplet_id":"openstack%2Foctavia~master~Ifd9973f126a20c35b960ced2851f75a4f8961b35","project":"openstack/octavia","branch":"master","attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-03-23 10:16:15.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-29 17:54:49.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ifd9973f126a20c35b960ced2851f75a4f8961b35","subject":"Use specific service user for inter-service communications","status":"NEW","created":"2026-03-17 12:09:35.000000000","updated":"2026-06-29 20:51:27.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0f9e9c31cc6203ebe7f4cac31ac4985d7fa1e259","_number":980907,"virtual_id_number":980907,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-29 20:51:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-17 13:43:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-19 18:21:06.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"55e7e03e54d8e448cbbec50fed0d29f34449cb98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-17 12:09:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c8df85e4fac47cfe933198d87fc167c01032c5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 13:43:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a8fa14860684ef0ba2bb5c35d211d63\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5d04432955140b18ff6138552364cdd : FAILURE in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c10f075afef34b05a80fc3b9d7d74911 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/985ab365895d4866a314aa250c34e6b3 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36a0074dab154124ab93e63687f4e3a9 : SUCCESS in 3m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d346454b55164bf6bc942547ca0b7168 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31dc33e8817848baba4f4286dfc40265 : SUCCESS in 11m 10s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/756db7d65290472b9f7d11eef590b8b8 : SUCCESS in 5m 18s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/6b61ffbf9d3146498a8a2d06873162ce : FAILURE in 3m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/85f4aa16ed6c47cf8ca586ceef2a44ac : SUCCESS in 2m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/063e6d2c66404a5fad13db13d34aaecc : FAILURE in 4m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9607d72caf9647d8a1f371a657f1e79a : FAILURE in 4m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/720195a6e6e14060bc6e2865717da14a : SUCCESS in 1h 28m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d43e76866e294b14a8b6a34d0739e7dc : FAILURE in 29m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6af6527f6beb43c6ae75001b7d4cbd90 : FAILURE in 26m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/867b31fa26544ef6af80dc6971650a1b : FAILURE in 28m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e5697c3453f848c5b191bdf8594ca1f6 : FAILURE in 28m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/01aec82d994f4bd29bc64f806f2682e2 : FAILURE in 30m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e155086252eb4df3abd3dd1b97779966 : SUCCESS in 45m 13s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/69de61bddeac45bea2d985bb0c0a86b0 : SUCCESS in 44m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e366f432cdc4e0091be65140a8e5fbd : FAILURE in 27m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8dbc8bf4820349369de3be6cee8d90be : FAILURE in 29m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/36fd6cf0edf3496e9e626066d6921591 : FAILURE in 41m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7b2e5e82a24b4f38ac509cc679f6b6c8 : FAILURE in 29m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/28f18f27f7fd42f7b4b5698ddd9a6073 : FAILURE in 32m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7383e2c0aa134c6bbe3860424ca7e68e : FAILURE in 31m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98529b25fe2891e23ba2b0b07299a47fd80eaaf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-17 13:51:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"26ac20f235bb62bc1e45ae5c3e8168a6a7ac1684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 15:29:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f53f43d6bded44549476439b399ea4e3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4a882c8877d44739f67edcccdfd90d1 : FAILURE in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0645d03217c4324b3c643c4896c0eb0 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78917e4a1d4e43a984f68992d3fb674a : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa9fde8657f9441bad3a67ed84e3e4ab : SUCCESS in 5m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d47d9d9398974a9fafee186e0b9a8827 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d23875f63034450c8b4a19c8234b92e7 : SUCCESS in 12m 50s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/63fa9c23bc09423f8b513fade719dbc5 : SUCCESS in 4m 13s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/736e45598a6046a2af9200135c369e3b : FAILURE in 5m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/eaa43baa532a4978aaf3b61233d313b1 : SUCCESS in 3m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/367900ffe29e4f928bc4dea3ae59326d : FAILURE in 4m 45s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3f6678b9ffc043a08b89a090fd168d24 : FAILURE in 4m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53a1ddf3d3f644149e3ce5846e42b093 : SUCCESS in 1h 30m 19s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b05d7d9f61ae49babe3d8edeb50fe2de : FAILURE in 29m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6a92196df4344a9987926047031446a1 : FAILURE in 27m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bd57a7f1260744a3b2185a83a6c61afd : FAILURE in 26m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9e05d9271c0b4adaa1fb43613d93d504 : FAILURE in 30m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/508fa26aae5243ca9c859254aa3a3dda : FAILURE in 30m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/032dea9df0834dc28273175263f3d5a9 : SUCCESS in 44m 58s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/054df8f0893d4fdf8479d58cb37ee2de : SUCCESS in 44m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a191413e222d41028f1e17de2eadd888 : FAILURE in 25m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3e14801d90164d83b1e9fa662a9770a3 : FAILURE in 28m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/61c548b25d734c6482c166ddf2426ec5 : FAILURE in 44m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8d9d4695ee524dc98012183ec90310ad : FAILURE in 27m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/b39c5f9f43ad47ab8c0921b78ed2d0bf : FAILURE in 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ca9fc65d141a4a11b5124ae2ec3ee3fe : FAILURE in 33m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b25c375dd5c77efbd39df0d2abdf1f162d495ff3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-17 15:41:13.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"11e4a9536540e355ed561645547c4274204707b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 18:07:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/238de7c5da3f4964949f11d8acb92582\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/468f349fff5e47e6a511733b6e362778 : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eab05debe17c4986904f1b23b4c52946 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c60e622bc3b04b68a7f33f6014a08fd1 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ba806cfebc048ac8d9afc448a7b325d : SUCCESS in 5m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae482e190d2745c6b5ca6c377c139643 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19e6eb82479b488bbbd20b7f75eb4198 : SUCCESS in 7m 48s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/372fcc2b912f48b3934f189eca8cc12e : SUCCESS in 6m 46s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/b5559a3649b04cff954e49532ca0ea13 : SUCCESS in 7m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0cfc1bff358a4a60b50d91a347c7e585 : SUCCESS in 5m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/52f6068321a6455eb671b42bd0c7a338 : SUCCESS in 8m 03s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0a81a08466bd44ef99cbbe0e44f93a6d : SUCCESS in 8m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ab85e1732cba4583b1366c260f3d65bc : SUCCESS in 2h 15m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ac67473af11142d3a81d01273fea44ff : FAILURE in 20m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/319e261ae6094076982674fc7554642f : FAILURE in 19m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7956ecfeba384968be4d8f40d3e96660 : FAILURE in 13m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3ddcdc638aa04732a4cc735d44ed6ce4 : FAILURE in 29m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2bbe728bd06941b088ea906b6bd01ac1 : FAILURE in 38m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2d41714dc7934cdda530edcf04bf9c18 : FAILURE in 22m 47s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a023ec56777444d99955febcd0a13766 : SUCCESS in 59m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fdfa51f8720d4da298cda6a46aa58d9c : FAILURE in 37m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9c70ce9f2fde4b12aba2ded3eb080130 : FAILURE in 52m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e3bbdace1d5f4f83afa338efa69b24a2 : FAILURE in 1h 33m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1d9eaf34deb641dc966079e1d02a4e48 : FAILURE in 14m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/992a589f473a4456a2a539188b257463 : FAILURE in 35m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/eaaaf22a971c4124ac86835066d78023 : FAILURE in 32m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"befa3118334f492eeab5f180cc65e6cd95f9ce30","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-18 01:23:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"56aa17abb0ca07bd0e251699a9218e6c25161161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 02:10:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e9885173a0d4c07b46cd717fbdb71ae\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6ea2013a56e4e51bbad31273eab92f7 : SUCCESS in 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee1c9f2b82748938d40e28b0b19ee0c : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23749320bbc743a7a81b89f2b3e5d5ed : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38eeb16125744f5d8fce2cb11e26f035 : SUCCESS in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/474515eb68d1443f978af9f777e09523 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf485d677d8249f39714080285e9bf10 : SUCCESS in 10m 28s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/14ba80baa68747a6b3f847da54a08d03 : SUCCESS in 2m 44s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/429c01b63ce44ec185cc86d08bb9f180 : SUCCESS in 7m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3fac6332cd564b22b0b95fdd05b0d854 : SUCCESS in 2m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/48979a7c4e0b41ce806b48f8a569497d : SUCCESS in 7m 11s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9a46eb3852484e1d91780cc303b89a59 : SUCCESS in 8m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f77a132e00be48df9f007c8181a74c8e : SUCCESS in 42m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/002c065b86374609a264621af4197f04 : FAILURE in 26m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9109ea7d35694135b7703416b6bad616 : FAILURE in 27m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2ac752d82234498eb63319cd26313812 : FAILURE in 25m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e9b124b9b2cf474dad6d1ae7eda54a6f : FAILURE in 26m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/04ee7090fac64cf994e01d51a0e41be7 : FAILURE in 29m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e0dfe1d4d0d64900a337d9b56b461c93 : SUCCESS in 43m 42s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2bb653f3f37a4a8793b3af4f94177d0a : SUCCESS in 40m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/267ffd4c56f648608ca8a8a9bc7f5895 : FAILURE in 26m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9a5734cda8984cd3a92c3f80eab2bedd : FAILURE in 28m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2fe5c1a9206f47f2b1a28c88e9a1a7b2 : FAILURE in 42m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c8438a0e77ac4f0f95a20a0dfa900901 : FAILURE in 26m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/64a5d733d5724896acca4ab90170e632 : FAILURE in 30m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/14496f6c105a4e6bb28c1e57c6b779e5 : FAILURE in 31m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e62477fd45b342b76f15ce62a0f3c286e1e3c8e4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-03-19 18:21:06.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8fbcb702ae9b0f45f141d0d3670532df9d8acc2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-23 10:16:15.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a13f11a0c4c587bf025ad73a6ada73afea299f48","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-23 10:35:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4dfcd0f3276ff3c9a7f45b7e8fe46c9c9f53a720","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-23 10:35:17.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"ad70674bafc0aeaaea647cff1e3f572c0c5753ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 11:37:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d183d802632f445d996dbd0f509f4d73\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8c180fc9efd40f0ac0775210b78159e : SUCCESS in 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58b82de93f5c4143a2b6067dba210016 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e5fe70e3e8f456689c5ffae75418172 : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/476d301333ae49518d5e2b718bb5947c : SUCCESS in 5m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dcc2eac6b74846279c88471e73a9e76b : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec968dabcb944b8db20ddc1f28e29cc5 : SUCCESS in 12m 31s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/d68e2e0977be442a8bd47d5ccc8eae3a : SUCCESS in 7m 34s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/a35ee2292c12477faf58194dc62a69c0 : SUCCESS in 9m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e94b05ce95724f52b8dfcaa662b8e589 : SUCCESS in 4m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b2c5a630019b47259960ae7249f24d6d : SUCCESS in 7m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d45666042a204be39a1d3858087d0f9f : SUCCESS in 7m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eecc269f18de48b9ae8050c7563f7699 : SUCCESS in 48m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4f29b94c0fc54557ae2e606e6c0708df : FAILURE in 29m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fa33c031c823472bb0ec4b34b02fd440 : FAILURE in 28m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d499eb059286421886cf8cd5782670bb : FAILURE in 30m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8fdaa0fba2574fdb9d7af0ad1beb2f65 : FAILURE in 37m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/949ba38b9faf498b9b2f9374fd152f61 : FAILURE in 28m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/46c0026473f545b8b715e6ff512a2ac6 : SUCCESS in 43m 03s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f3ea7d493efe4ec8bdc341ce5c919e5d : SUCCESS in 46m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/491b55f88e034ec48e7296af2af99a2f : FAILURE in 31m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5136a09433fb486d8c271396dd3a5071 : FAILURE in 30m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a63a9454429a46319f40d7bcc45352cb : FAILURE in 44m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3e08f791b893473ba65a50af7a1848c2 : FAILURE in 28m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/c612f727efd043b4ac0ff327f6fc430a : FAILURE in 33m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d0fb684a139d4db6a393c762def9e2fb : FAILURE in 34m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"32303a0ae7da78f0385d6127c0f951b6a8fc4b98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-04 14:17:21.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"b0bdb2302b52fc4a8067eb32812f286ee7470602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 17:17:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d35b9823452a46f6aeb1e562a053e15d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c4a933bf7d44578b815e90f2a8742f0 : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec8f16899eb0463b9e9c0afcfd41980a : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c644f02270f4d578b60cbb9dc6c0ce1 : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8cc8923c622f4c309fa96532d506abc7 : SUCCESS in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eb4835a518be4da989c45ff17a3b5bd7 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fffc549d4294c088e4ef88486470fc7 : SUCCESS in 13m 14s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/3eb3b770313f440a8245cedc22638942 : SUCCESS in 5m 26s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/c8547c18a8784e9f9facbf81257c9924 : SUCCESS in 6m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/32877bd6884848b59f5e0ee091b1cd3d : SUCCESS in 3m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7d92be28ed024c239021d662d8b93faf : SUCCESS in 6m 10s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/58133dc0687641b1b05fedf975543428 : SUCCESS in 5m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/79e89f4c15d042b9aa8978368691121b : SUCCESS in 1h 04m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/24d7c7b8f9684a72b31f87cb281cbe35 : FAILURE in 45m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e1f7a173613b49d78f01aaf70ccfd866 : FAILURE in 37m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/85129b56655c4f43937ec06fffc4e9bd : FAILURE in 26m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4f6f882c7b1a45f897d0ccc85e294957 : FAILURE in 39m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/615bba84112644d9bef606d4b7da5399 : FAILURE in 29m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/86a887d472af433fbfadec02f4a20984 : SUCCESS in 1h 10m 14s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/78d383cbab2240659a7b5e082a9d12c7 : SUCCESS in 43m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e797b5308b94115aa6a82acec158234 : FAILURE in 36m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ba5c190337244ed287e206103b3fd6ef : FAILURE in 19m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/028442a2b8a74562bb3171f35e70925c : FAILURE in 2h 55m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/253ca3f726d045afb802a52c86d68c35 : FAILURE in 25m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/e2c546b4504d49d8be1cc811ceaf4b42 : FAILURE in 28m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/063ccdfdde6c4c97a1770416328f2966 : FAILURE in 55m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1b5121c44922f2d395a359b58ca0fb583b58665f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-04 17:32:28.000000000","message":"Patch Set 6: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cd5555de1479d6b2870753468189996b50766333","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-04 17:40:08.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"27e60dac6df439d01f1cb722880a98280ff5050f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 18:32:06.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63ac5027fe14499f96233767aed0f277\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8b15f36c2a94a7396001982bed7318c : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a616451227e144d593d81425240575d2 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b783f908dba444f9a21f6fd0b773955c : SUCCESS in 6m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98982d5794084642a62c8ea26abc8d96 : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/38bd62884af5482682f7c942c9f2dba7 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8ba6743b3d345a698b90df1e0ab1989 : SUCCESS in 12m 27s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/2cb9ffa549674a4faec7a6e60a98b5f5 : SUCCESS in 6m 08s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/00be477fd52745dc826cb4118035f60c : SUCCESS in 5m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4ab23fe6d8bc4d9c98561328e218768f : SUCCESS in 4m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/58a58ae2590447aab5e9125d28bf61ea : SUCCESS in 6m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/01e0bb3edcde4c69bedcd222cb06b351 : SUCCESS in 5m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d3e24b2d16754d018eb45898daeb1407 : FAILURE in 16m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3f3c33907f8d465eb2c1a5b2dbee2e1e : FAILURE in 19m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c2ab557e99624007b4e7def0bfd1d654 : FAILURE in 21m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7f16694b3180479aa9591107db7cc943 : FAILURE in 20m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f0ab4562257c40acaa4f1cb5373d8695 : FAILURE in 20m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d3e2242c979f4f6ca25e47fd8bc58ae8 : FAILURE in 18m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/179bee4bfff841239bbc0a9e2f23bf9c : SUCCESS in 49m 17s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f3906f6c7b0b4f758e4f4fe1cda46645 : SUCCESS in 47m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02a7889403ab41f3bafb2a2c73699a2c : FAILURE in 17m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9dadebae60ce44e6abf636a63b55fd64 : FAILURE in 18m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/007ee18d7479408bb8a80b8518b3523e : FAILURE in 20m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/85c21b5e5ad942538c5228e1ed1d2702 : FAILURE in 15m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/8d6856d635884bbfb99c3042ecabda1f : FAILURE in 20m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6aedf04c35334475a9728cef236ea3d7 : FAILURE in 22m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4403656f30f0278d4c93572e876fa17c7467cff2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-04 18:41:08.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"48f3fe5f26dda34a9c51285056d5915beee949bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 21:45:51.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c58674c7589c436f98763886653a8c87\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/384d438aeae94b678dc150cf048c8a1c : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f62d3e11399445fa966c3c1c7b0e7873 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32fb5739e046445e95267af84cf05993 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/882d757fe2d4482787ced6911d7d6573 : SUCCESS in 3m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9bbd9b2f609648c4b676712e3f355383 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63c6268836b04cce89e6661ec36f65f9 : SUCCESS in 7m 21s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/1ee45e8417e7495c90f365fb17d37c02 : SUCCESS in 5m 49s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/abb3a9f771a1437292710766562d712b : SUCCESS in 5m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4f1f79f35647418f8e592364fe1124bf : SUCCESS in 2m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e33e3e27dc7244539373dfc3b42f781a : SUCCESS in 6m 11s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8f7c8af60ba843c7a319c929ba0cfb4d : SUCCESS in 5m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cf036d47f3fe4f5faa14ba1cb0481f55 : FAILURE in 19m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5becb279ae3b498fb19a8c9bcfccc89f : FAILURE in 24m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4a1d2266c31f458bbd9303328069fb4f : FAILURE in 23m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2688c4ab77114befa46138d56f8f8d71 : FAILURE in 22m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/576c5ad7974f49a2b3e7291c74dd82d6 : FAILURE in 23m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bd48a541e70f4685b8ec22aa39e225cf : FAILURE in 24m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1cf967fd208d4714a60fdd3e02f94738 : SUCCESS in 42m 56s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8919f2a26d234955a6b4898b5993b58e : SUCCESS in 42m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0edd5344fbbe45f3b993d5f0e18a24b1 : FAILURE in 22m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a49e632055234f198f538e3a44e166d5 : FAILURE in 22m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4e0e3ede9dfe4c8099c0b70f61f2df81 : TIMED_OUT in 3h 03m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c5b2d53e8f5b446680f524aa97661946 : FAILURE in 23m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/e25d19a332ed462a9153b3b5b561b6f2 : FAILURE in 30m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f36b641a05c5469b9fce768763d02b9e : FAILURE in 28m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0cb3c67903493e59437c09df985fbc0a2890a7b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-05 08:41:53.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"65eda3456de4c1f46e9a9e74a447d0675ef147e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 09:53:31.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fefb345470694945ab430a4a2c1f2b44\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e110afce17714ffc93211637aa7b8fcf : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63f075ecc6e245239e22fdc579104b69 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6664c29d49634b589435848e526cebfa : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e74c58a0543741cb806b054f95d0a1b8 : SUCCESS in 2m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a0640b9d474b475f934a041ab71e4ac8 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a0bc33664a04ce4880e7450c288ed16 : SUCCESS in 5m 51s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/d739dce7bdd34d6c8f4867b73da5f54c : SUCCESS in 4m 10s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/16ced6ebaf5045d5a0c1830d5d379be9 : SUCCESS in 6m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0df54c4b0d964905a6ce1dbae2273294 : SUCCESS in 3m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/92615a5a5f594d20ba2e1e8362adc0b7 : SUCCESS in 5m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/dd509e31db1e48ac9a678bd7e03cb288 : SUCCESS in 3m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8f637b95908d4053927ec08151b2555d : SUCCESS in 1h 06m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/278544446abe4bf8b45b6c1397016b86 : FAILURE in 31m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f489aa04470341c0be8a7ab98a1ac01f : FAILURE in 32m 40s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7bd6658b2a3e4757bf0c6452a88918b6 : FAILURE in 27m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b2a29649008d461b97e66e3881a1b2e6 : FAILURE in 31m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ac03949a4604455b9c82251e3b20ea6 : FAILURE in 33m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/03b7b6ac58c1428da9554cc271353f3b : SUCCESS in 46m 29s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ff3d23e0357a482ebbd7f768fc021731 : SUCCESS in 46m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5d0868456de48c484ef3935f76a4256 : FAILURE in 27m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5770d2f2b8034fffb907024b72b2983d : FAILURE in 34m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4cc8b8f7becb4f918f3cf1f653c63fe5 : FAILURE in 48m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a3954f07fc11486a92d6af78f5f39083 : FAILURE in 32m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/ea2bc0cf3daa4c20bfdcc880fee2eeba : FAILURE in 36m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3140a5c9a9fa43deb1e372ffbc1582f8 : FAILURE in 38m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"03fb6bfd9ac190b99c1496ce223b409b0e5c4b83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-05 11:12:25.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"079644d4ac7c9761d0dc86967b33d0bc880b921e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 13:47:39.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2300fd41d4a04bc496db278b1a1f7ebe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1316b47e1c9e427783334f1316267903 : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ef40a479f504af2bc73cf53add62e6d : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd8f73fc71264f58953c612be359d2fe : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65c7211818a048ff9b6760d832e7ea70 : SUCCESS in 4m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b28664095c8a47c2a3cadfbe2f2be2ae : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/660dc9c2073a41d5b0f3efc28c2e9fd3 : SUCCESS in 13m 16s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/ad44be2432804b488d234050ea440010 : SUCCESS in 4m 47s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/b9394af176fe400a8fe694fd14287a9a : SUCCESS in 5m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d00d44406bbe4e00b45cceca513773ab : SUCCESS in 3m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dadcc08970d646e79aa8a747b1572240 : SUCCESS in 5m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8545a9027bb24ae59cc1824cf14f0ce0 : SUCCESS in 10m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9ab696fef6a14d7da5212c820c7336e4 : SUCCESS in 1h 02m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1ca409bcc2c94463b865b5aa21a49b71 : SUCCESS in 1h 36m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/754bbf4057ec4993839b1770494ec1a9 : SUCCESS in 1h 01m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2b1ffc7a10c040068d4fcaf282efa193 : SUCCESS in 1h 37m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9e26144d4b04424f8da2a99e90e2cf4e : SUCCESS in 1h 10m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/55d8279e6ed54de184d17b91f11661c6 : SUCCESS in 40m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/27eb879991a840d5be7fe4a93d55bddd : SUCCESS in 54m 25s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/52cfe745cbb0482ea78e27a55c6405f6 : SUCCESS in 46m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ca9186d798d42d899c53e3c2ffc0658 : SUCCESS in 37m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ffda94f928c54ef39b22933c502a5616 : SUCCESS in 38m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/44d3123999d64ae5bbf31d3265fac73a : SUCCESS in 2h 33m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/940567ed1d274826af8940787d89a104 : SUCCESS in 1h 33m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/f3669b21e7b040879ad1f2ede2df1842 : SUCCESS in 1h 32m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/20ae298c8ace421b926e244e9bb2e3fb : SUCCESS in 1h 01m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d78162810dd40fc771a2f41627070f2a99ec56cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-29 14:42:45.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"c8fec5f5b328dc66acf7c40687410fb884e9fd9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 17:48:05.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6614c7a31b44b019cc23cf1b19de354\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b818db9727de4f0c8bf710a03bb3e5f7 : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb87cb166187425aa40c1345b9c31787 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1d41fe0254a42faa4a6f5eb0c765203 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/992454dde3eb41e1b49493cf1bb4f2f8 : SUCCESS in 4m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/36b3031f3593462bb841fd70009b4a5e : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c84ef5403f548438ee97c0c2ca6194d : SUCCESS in 14m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89c69b8ddba44a56973dd164bebe7fa8 : SUCCESS in 5m 09s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/6459416094cd4ad18d17748a906aa7eb : SUCCESS in 4m 52s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/f07aa2e59706402bb99faefd203f228f : SUCCESS in 4m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d2b72e1af00049fa9086b6ebdec28c8d : SUCCESS in 5m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bb88efab2c294120a961f8d9d2cd84a2 : SUCCESS in 4m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bfbd9ceacca24a22a969bedff491db46 : SUCCESS in 58m 43s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4071fcbaa3a545d581d8e7da48c43463 : SUCCESS in 1h 31m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/970fd447585c4301b11308c5eb27527e : SUCCESS in 59m 17s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/472b18e4fe5745beb671aebf39c6ae26 : SUCCESS in 1h 33m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/26684003aff445478fd4d65df1fe5f35 : SUCCESS in 1h 01m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1d8bc9fd596245a9930bf9e6d5b41724 : SUCCESS in 36m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c4d296537f7645c59bcc30c187dd0e0b : TIMED_OUT in 3h 03m 33s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/81af0ad4278a4ad2b503d5544914a863 : SUCCESS in 42m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ed1d3b74ee149af8ae8f2bf1d607266 : SUCCESS in 33m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/71cbb10ae20c46e9986343a5cab3d9cb : SUCCESS in 36m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/03af3a245a844cbd9a7965b850abe33b : SUCCESS in 2h 17m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8a7eda96399f4b998e5e9c966c35c05e : SUCCESS in 1h 29m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/08690385db514abfae88272af2720434 : SUCCESS in 1h 25m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/67988fe829514351bd5660882dc514ee : SUCCESS in 1h 04m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8b246f5b10c102c0397f10d8eb690e687faae1ad","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-29 17:54:49.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"0f9e9c31cc6203ebe7f4cac31ac4985d7fa1e259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 20:51:27.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43628608b7674a2aa7065f90ba452ea1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47af871fade74e37864ba04a2ff8e1ee : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/936525d2919c4347be7d9350b406d248 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a04c249919a24aca9fbbdef866dc1c3d : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b0468a2097a14f3c9efb39d5df3387df : SUCCESS in 4m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8b996f36e4464bd2a975e342c8d9e77e : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0fbc66f3b164c6e97975f5348c0f77c : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8da00344ea274ff3b6945eacd7c0710c : SUCCESS in 10m 26s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/5d9230af46cf40438bff689f0b71fff1 : SUCCESS in 4m 28s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/09369216ae4744b997d1b119c18f9211 : SUCCESS in 5m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f8236f5e8fa947218301b53a5a9e5df5 : SUCCESS in 3m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/825464e0aff84286ad0f2fa7222d88ad : SUCCESS in 5m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/84874ac62cc147ca9b1470e89a08aa29 : SUCCESS in 1h 18m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4a79a7076db94f0cb4aad6c914457df9 : SUCCESS in 1h 36m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/47a4e1ad67c14a89988d17b2e74ad80f : SUCCESS in 1h 07m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/52b1f5e6d4564efa924c94293e738093 : SUCCESS in 1h 41m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d25493e6e5a049e588837fe8d70fe6c3 : SUCCESS in 1h 01m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e281b61c777e4f218253c7ae63c5c9cd : SUCCESS in 33m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7c1efe792b784e9b8bb863442311ddcf : SUCCESS in 48m 02s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/4ba82b2f7b114f6d828f05eb19ca45f3 : SUCCESS in 41m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84dff82548e24db1b5634b1edd509bc0 : SUCCESS in 40m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/194d201918014a8298eb759ea91c4924 : SUCCESS in 33m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/24889f3db6b34e2d81ea27df1fd71a78 : SUCCESS in 2h 55m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5db9363b19704a99b50a7cc161f396be : SUCCESS in 1h 35m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/67a734670eb7471ab6af05e340d638ff : SUCCESS in 1h 31m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/80eaf67cf5104129828cb209caa91011 : SUCCESS in 1h 02m 16s (non-voting)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"ce92fff16e8b6f5b199be5c2ef33726582c339fd","revisions":{"047276e905362e71e92c9051239d8de8ff7ee501":{"kind":"REWORK","_number":1,"created":"2026-03-17 12:09:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/980907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/980907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/980907/1"}}},"commit":{"parents":[{"commit":"94e25df57912fadbaa77fba4abcbb38770c60d30","subject":"Hard-code service name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/94e25df57912fadbaa77fba4abcbb38770c60d30"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:06:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:08:29.000000000","tz":540},"subject":"Use specific service user for inter-service communications","message":"Use specific service user for inter-service communications\n\n.... instead of the admin user.\n\nThis is the popular architecture in production deployments.\n\nChange-Id: Ifd9973f126a20c35b960ced2851f75a4f8961b35\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/047276e905362e71e92c9051239d8de8ff7ee501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/047276e905362e71e92c9051239d8de8ff7ee501"}]},"branch":"refs/heads/master"},"16e34420d10334054bc90cc62b22cdf46a99d909":{"kind":"REWORK","_number":2,"created":"2026-03-17 13:51:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/980907/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/980907/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/980907/2"}}},"commit":{"parents":[{"commit":"94e25df57912fadbaa77fba4abcbb38770c60d30","subject":"Hard-code service name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/94e25df57912fadbaa77fba4abcbb38770c60d30"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:06:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 13:50:46.000000000","tz":540},"subject":"Use specific service user for inter-service communications","message":"Use specific service user for inter-service communications\n\n.... instead of the admin user.\n\nThis is the popular architecture in production deployments.\n\nChange-Id: Ifd9973f126a20c35b960ced2851f75a4f8961b35\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16e34420d10334054bc90cc62b22cdf46a99d909"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16e34420d10334054bc90cc62b22cdf46a99d909"}]},"branch":"refs/heads/master"},"b255cc830021a90e7a8597c9e4d3d35afc2fd69c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-17 15:41:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/980907/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/980907/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/980907/3"}}},"commit":{"parents":[{"commit":"2dca2d89009fa23935ca68152054a62964960845","subject":"Hard-code service name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2dca2d89009fa23935ca68152054a62964960845"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:06:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 15:41:13.000000000","tz":0},"subject":"Use specific service user for inter-service communications","message":"Use specific service user for inter-service communications\n\n.... instead of the admin user.\n\nThis is the popular architecture in production deployments.\n\nChange-Id: Ifd9973f126a20c35b960ced2851f75a4f8961b35\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b255cc830021a90e7a8597c9e4d3d35afc2fd69c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b255cc830021a90e7a8597c9e4d3d35afc2fd69c"}]},"branch":"refs/heads/master","description":"Rebase"},"1c2c882ffaa0d0b246c0e45b80b5744e2038fdd9":{"kind":"REWORK","_number":4,"created":"2026-03-23 10:16:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/980907/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/980907/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/980907/4"}}},"commit":{"parents":[{"commit":"94e25df57912fadbaa77fba4abcbb38770c60d30","subject":"Hard-code service name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/94e25df57912fadbaa77fba4abcbb38770c60d30"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:06:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-23 10:15:10.000000000","tz":540},"subject":"Use specific service user for inter-service communications","message":"Use specific service user for inter-service communications\n\n.... instead of the admin user. This is the popular architecture in\nproduction deployments.\n\nNote that the admin role is added to the octavia service user due to\nlack of permission to manipulate security group. (This needs to be\nfixed in neutron)\n\nChange-Id: Ifd9973f126a20c35b960ced2851f75a4f8961b35\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c2c882ffaa0d0b246c0e45b80b5744e2038fdd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c2c882ffaa0d0b246c0e45b80b5744e2038fdd9"}]},"branch":"refs/heads/master"},"6f359b17f50280b4e17c09abd1f7d2ee51fd80ca":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-23 10:35:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/980907/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/980907/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/980907/5"}}},"commit":{"parents":[{"commit":"2dca2d89009fa23935ca68152054a62964960845","subject":"Hard-code service name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2dca2d89009fa23935ca68152054a62964960845"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:06:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-23 10:35:17.000000000","tz":0},"subject":"Use specific service user for inter-service communications","message":"Use specific service user for inter-service communications\n\n.... instead of the admin user. This is the popular architecture in\nproduction deployments.\n\nNote that the admin role is added to the octavia service user due to\nlack of permission to manipulate security group. (This needs to be\nfixed in neutron)\n\nChange-Id: Ifd9973f126a20c35b960ced2851f75a4f8961b35\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f359b17f50280b4e17c09abd1f7d2ee51fd80ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f359b17f50280b4e17c09abd1f7d2ee51fd80ca"}]},"branch":"refs/heads/master","description":"Rebase"},"f1052c694a7003bc9a252094f07761fbe8809857":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-05-04 14:17:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/980907/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/980907/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/980907/6"}}},"commit":{"parents":[{"commit":"fa315ca8c0c832958cf104b3967951d4a2cab53a","subject":"Hard-code service name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa315ca8c0c832958cf104b3967951d4a2cab53a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:06:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-04 14:17:21.000000000","tz":0},"subject":"Use specific service user for inter-service communications","message":"Use specific service user for inter-service communications\n\n.... instead of the admin user. This is the popular architecture in\nproduction deployments.\n\nNote that the admin role is added to the octavia service user due to\nlack of permission to manipulate security group. (This needs to be\nfixed in neutron)\n\nChange-Id: Ifd9973f126a20c35b960ced2851f75a4f8961b35\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1052c694a7003bc9a252094f07761fbe8809857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1052c694a7003bc9a252094f07761fbe8809857"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"6f359b17f50280b4e17c09abd1f7d2ee51fd80ca","theirs":"fa315ca8c0c832958cf104b3967951d4a2cab53a","contains_conflicts":false}},"472434d9fff558d8ae3836322bde7f68942ecefc":{"kind":"REWORK","_number":7,"created":"2026-05-04 17:40:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/980907/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/980907/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/980907/7"}}},"commit":{"parents":[{"commit":"fa315ca8c0c832958cf104b3967951d4a2cab53a","subject":"Hard-code service name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa315ca8c0c832958cf104b3967951d4a2cab53a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:06:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-04 17:39:53.000000000","tz":540},"subject":"Use specific service user for inter-service communications","message":"Use specific service user for inter-service communications\n\n.... instead of the admin user. This is the popular architecture in\nproduction deployments.\n\nNote that the admin role is added to the octavia service user due to\nlack of permission to manipulate security group. (This needs to be\nfixed in neutron)\n\nChange-Id: Ifd9973f126a20c35b960ced2851f75a4f8961b35\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/472434d9fff558d8ae3836322bde7f68942ecefc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/472434d9fff558d8ae3836322bde7f68942ecefc"}]},"branch":"refs/heads/master"},"cb7396c292f1bae8c3c9b6add78a9691f2db64f5":{"kind":"REWORK","_number":8,"created":"2026-05-04 18:41:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/980907/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/980907/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/980907/8"}}},"commit":{"parents":[{"commit":"fa315ca8c0c832958cf104b3967951d4a2cab53a","subject":"Hard-code service name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa315ca8c0c832958cf104b3967951d4a2cab53a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:06:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-04 18:40:44.000000000","tz":540},"subject":"Use specific service user for inter-service communications","message":"Use specific service user for inter-service communications\n\n.... instead of the admin user. This is the popular architecture in\nproduction deployments.\n\nNote that the admin role is added to the octavia service user due to\nlack of permission to manipulate security group. (This needs to be\nfixed in neutron)\n\nChange-Id: Ifd9973f126a20c35b960ced2851f75a4f8961b35\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb7396c292f1bae8c3c9b6add78a9691f2db64f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb7396c292f1bae8c3c9b6add78a9691f2db64f5"}]},"branch":"refs/heads/master"},"049f302c6a197a3b733539e758569f2ad7016d44":{"kind":"REWORK","_number":9,"created":"2026-05-05 08:41:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/980907/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/980907/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/980907/9"}}},"commit":{"parents":[{"commit":"fa315ca8c0c832958cf104b3967951d4a2cab53a","subject":"Hard-code service name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa315ca8c0c832958cf104b3967951d4a2cab53a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:06:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-05 08:41:37.000000000","tz":540},"subject":"Use specific service user for inter-service communications","message":"Use specific service user for inter-service communications\n\n.... instead of the admin user. This is the popular architecture in\nproduction deployments.\n\nNote that the admin role is added to the octavia service user due to\nlack of permission to manipulate security group. (This needs to be\nfixed in neutron)\n\nChange-Id: Ifd9973f126a20c35b960ced2851f75a4f8961b35\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/049f302c6a197a3b733539e758569f2ad7016d44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/049f302c6a197a3b733539e758569f2ad7016d44"}]},"branch":"refs/heads/master"},"8d8b9128aa5c65e888830c6ef19c86eeb3348f4a":{"kind":"REWORK","_number":10,"created":"2026-05-05 11:12:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/980907/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/980907/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/980907/10"}}},"commit":{"parents":[{"commit":"fa315ca8c0c832958cf104b3967951d4a2cab53a","subject":"Hard-code service name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa315ca8c0c832958cf104b3967951d4a2cab53a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:06:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-05 11:11:21.000000000","tz":540},"subject":"Use specific service user for inter-service communications","message":"Use specific service user for inter-service communications\n\n.... instead of the admin user. This is the popular architecture in\nproduction deployments.\n\nNote that the admin role is added to the octavia service user due to\nlack of permission to manipulate security group. (This needs to be\nfixed in neutron)\n\nChange-Id: Ifd9973f126a20c35b960ced2851f75a4f8961b35\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d8b9128aa5c65e888830c6ef19c86eeb3348f4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d8b9128aa5c65e888830c6ef19c86eeb3348f4a"}]},"branch":"refs/heads/master"},"ce92fff16e8b6f5b199be5c2ef33726582c339fd":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-06-29 14:42:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/980907/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/980907/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/980907/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/980907/11"}}},"commit":{"parents":[{"commit":"c53013237649d790e0c056ae2bf9d9d9b6944384","subject":"Hard-code service name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c53013237649d790e0c056ae2bf9d9d9b6944384"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:06:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-29 14:42:45.000000000","tz":0},"subject":"Use specific service user for inter-service communications","message":"Use specific service user for inter-service communications\n\n.... instead of the admin user. This is the popular architecture in\nproduction deployments.\n\nNote that the admin role is added to the octavia service user due to\nlack of permission to manipulate security group. (This needs to be\nfixed in neutron)\n\nChange-Id: Ifd9973f126a20c35b960ced2851f75a4f8961b35\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce92fff16e8b6f5b199be5c2ef33726582c339fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce92fff16e8b6f5b199be5c2ef33726582c339fd"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"fa315ca8c0c832958cf104b3967951d4a2cab53a","ours":"8d8b9128aa5c65e888830c6ef19c86eeb3348f4a","theirs":"c53013237649d790e0c056ae2bf9d9d9b6944384","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
