)]}'
{"id":"openstack%2Foctavia~980908","triplet_id":"openstack%2Foctavia~master~I157cd2324cf67f5e9dc0495090a3bf72995941ee","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-29 13:26:38.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I157cd2324cf67f5e9dc0495090a3bf72995941ee","subject":"devstack: Use common function to manage keystoneauth options","status":"NEW","created":"2026-03-17 12:09:35.000000000","updated":"2026-06-29 16:55:27.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"645584a23ab6f68c8a494b3cdb869cc8e8460b73","_number":980908,"virtual_id_number":980908,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-29 16:55:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-17 13:45:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13a142395bc8ae05f0efec033e0da43c2030fe67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-17 12:09:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9872a16d008d853a5dc5c5f253d05c1c7e244235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 13:45:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40f123fc104d4f5d8a9c256aff0cd15b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4526f9c82cf44aca0cef77ab6332dfb : FAILURE in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04fa2134ffe647bdafebaa3dd0530904 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0282781a75644f7b4998a63addae5ee : SUCCESS in 6m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84c2d18d42134523a5ff95c90a3b214c : SUCCESS in 6m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/658ec04bebb2471586c5fee1c08344de : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a85413301a542258c043dc4fcd77217 : SUCCESS in 13m 09s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/63526a197aa14bbdab1e7647694287e7 : SUCCESS in 5m 12s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/25182bdc3b674c44845fa2e8ac71ab46 : FAILURE in 4m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/39fc1c43359044ecb7cb216316a1bd38 : SUCCESS in 3m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1cce8f854c2e4ea7aca02a3d77e28394 : FAILURE in 6m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bcca721461f244e988fe6d09cc60cb08 : FAILURE in 4m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c42343985e3f40258a5cc57a72404568 : SUCCESS in 1h 30m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7c4f5d4b99214f2c9844af49d2e1b939 : FAILURE in 27m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c4d568baab284b3f8e37f192b53523a1 : FAILURE in 29m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/414895c8378b4045abf281ea2d148ebb : FAILURE in 26m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/30994a994eeb478ea9aead34da77d8cb : FAILURE in 26m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/20b5c6e33c0141ac9846935fd78367cf : FAILURE in 27m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ae62c720f5da4ac9b5d3e84d9eb206ce : SUCCESS in 43m 25s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/fc284778581a4b29a5cbecc7e071b7d2 : SUCCESS in 44m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28e45e49249049e3b7ef76298dd0db9b : FAILURE in 28m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e03125aa43c740749508221eb2c9fac0 : FAILURE in 29m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2fe8157ead584b41bfadadeb7e44e288 : FAILURE in 47m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/af8db0451037486186a59949e759bc3d : FAILURE in 27m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/e5363202ce59405285b4407e94793106 : FAILURE in 32m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3662143651274b5c83291f65109806c5 : FAILURE in 31m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ea7991ba2a853dd8bef865d903477ddf88d64dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-17 13:51:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6471ed88e20ab8a8e3203f391f135052736bb42d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-17 15:41:18.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f5dd6e8ba2c816038dc8d5547450272ba89387fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 16:42:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66265ae1f84d483b92b088bace937a9e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6868b8f638e745d4bbef2bdda3851aca : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05083499e2ab409483103cdd4d19fcf5 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f53eb835c9847d9adfa44298550c166 : SUCCESS in 6m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba4453801d60432eaf45cbbeaf109743 : SUCCESS in 6m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/77046150e9a446db8b459ac4e60b1c71 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8317823f08448eb930fb3ca99ddc440 : SUCCESS in 15m 37s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/0124fd494d0a480fa7e58114056d6d07 : SUCCESS in 5m 31s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/e55fd07e950b4dd5abfb6ed41bb5d40b : SUCCESS in 5m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5a7af8c5d0074594906c42e499923fe8 : SUCCESS in 4m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1c4f4d22e67040249a12660a36676452 : SUCCESS in 5m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a1e9d445e66f43c5a14c4c82ec7e392f : SUCCESS in 5m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2a9c0871887345dfa87926890b7e6867 : SUCCESS in 49m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d0b8d217446245fcb94f80cf13080809 : FAILURE in 26m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/06506118abab4ed1bd78dc4d98174b5a : FAILURE in 52m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f140fccc3b7346bf9e612317ed97350e : FAILURE in 13m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b3c95fedccbb4e95a71e1274bf587fce : FAILURE in 21m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e5f70f81f0de4df48b971e7b30e1cbfc : FAILURE in 34m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d30f73490b7547369ae12bb75d443a70 : SUCCESS in 50m 10s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b55e1e1ec38f453b9a1cbef17416baca : FAILURE in 15m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/385fae50189f48199a99afb884cf3450 : FAILURE in 26m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ed6d387bc3c546f9aebc88edf901cf67 : FAILURE in 20m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bd2a03fff0cc4b9ab1f6748b7450e671 : FAILURE in 36m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d0935807aadc42ac8f22950a3a403147 : FAILURE in 20m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/3e898eff929a4908a8f1428e4d3ed042 : FAILURE in 32m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/542e873f10e548208eb2d6f15edc0f20 : FAILURE in 37m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6ac39b37f84ddf98c99d66ba4a22a15045a4b5a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-23 10:16:15.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"12e2bf42c40a11aca7b8df823973ffca42fe2850","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-23 10:35:22.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"dd95e4fea62471734770f5ab4f2b639b34437c5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 11:53:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57e96695de7448d282d50b19b8a9794e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f54a3bdcc2c440fea19f2f9b2e3035a5 : SUCCESS in 13m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f22664b80346e882703d4e39fb5c80 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/feefb8bfbd8f4e3eb8b188eb4cf4fba4 : SUCCESS in 6m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9dfb1bbb9014147ba34e0fa38473f54 : SUCCESS in 5m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c667e61fd6e7431c983ed8a294719dce : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/900b14100c7e43b4b02c1dbb7fffae1b : SUCCESS in 12m 40s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/c96c413048834e6181e580157e80478e : SUCCESS in 5m 12s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/db2ccbde16aa425687d679f0f985052e : SUCCESS in 7m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9900bd8d8458405da67a4447b4cd402a : SUCCESS in 4m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1614eee4c0db43388c8a022873b571cd : SUCCESS in 9m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5af1e1cb93e547c58058135e4785f542 : SUCCESS in 5m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bfaacfe51ebd487985a951b7842d0086 : SUCCESS in 1h 05m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/57acaf35ab554cf2aa79a03a31e7909b : FAILURE in 28m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9310d5ad1bc74f6dbfea0dbfd409d006 : FAILURE in 19m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f6981c6d9fdf46358ded02e913eeaac0 : FAILURE in 36m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a7162dabdd074a3fb3312d922159c703 : FAILURE in 35m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/be9ffe8d38e34d00b066e755fb229813 : FAILURE in 38m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/52ce3219082a4911bd70ed7a7efc974d : SUCCESS in 55m 24s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9c5e57e5ef214ba79e60224bd37ff11f : SUCCESS in 42m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4d8535f722f4db18507a0196887f270 : FAILURE in 31m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5d067e3ce55e48158ec20f18d99bf454 : FAILURE in 32m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c25b021e53f54d30953d80ceef9bdfc5 : FAILURE in 51m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/46061003b4f747389ca73b11e6fb5203 : FAILURE in 27m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/d513922631d0484ba20d93e567ba3ae7 : FAILURE in 53m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/58591f49fc9747aab0306b797db0adb4 : FAILURE in 35m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5bdaa16b7fe5d5943b2391baa9c9a0b6e2d6ba76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-04 14:17:30.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"dda78ee44b9ba310409f65abee80745b18eebbe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 17:27:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99d7b102c3c340babe24f2b3ffb9c973\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e073be3d986340ef93ba8d491ea9a5f9 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c584e7bdb5146278c94b24cd616b4aa : SUCCESS in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e3c85837c214b4e9dbe3a3d2dd59af5 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48fc85ed92bc4543843370f9971fbe3c : SUCCESS in 5m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f1d32431197941bcbec9e189edc2c4fa : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea938b439c6e40b98f37a008c79abd42 : SUCCESS in 13m 14s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/46ba113d28ba410e92c3d2b61ebd5e05 : SUCCESS in 5m 11s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/687bdec5ebf045059f853d039c41ae2d : SUCCESS in 6m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0e1684cf46d7451c9086444c9a63d60b : SUCCESS in 3m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5c15ca1c5ae940eda71e7cd15928fbcf : SUCCESS in 6m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2dbf6579962c4ddb9207a4cae4a08615 : SUCCESS in 5m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c25d0418b1ae492996c0568a2e04de83 : SUCCESS in 47m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5dacc7e5742c4710b190cd22b3b3d7be : FAILURE in 25m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fd2eb5b9b07c4833ba3ee4d1fba67035 : FAILURE in 25m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3b57a6d572ad4eff80a4168b092b2903 : FAILURE in 25m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9c4dc27354924f1180e724a70ee592ed : FAILURE in 29m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1c6f987db52f43e4982c3fea3cdb3337 : FAILURE in 38m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fcf0cf17e2904bef839c0ffa6ce3b4c4 : SUCCESS in 1h 08m 54s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f0d696f15ff84af388e4b703dbb234f7 : SUCCESS in 38m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f17b122c70f74ad2925397beff7a768a : FAILURE in 28m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d72134b6a49f466e8c8f2b86d529d115 : FAILURE in 20m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/71304476583446a88ee1af3d62ba37c0 : TIMED_OUT in 3h 05m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2236e805fca34c18bbcedaf71aa5daa3 : FAILURE in 25m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/c70e08629b8a4b819cee718ebdbc9c02 : FAILURE in 28m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/08df41cd344d4b289959085a082e68d6 : FAILURE in 56m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5bea1c5ccc52f785f572c132734ce0fd49808520","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-06 03:51:45.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"267b18b453a6fb72391fbb20630fd0d2453b6149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 06:31:43.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8e76ecb84554eda941c2ccbcab12b27\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ebad6991e9f4dd29533e9aaefeb434e : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c7b322474bf479189a0a980da92a63e : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ed9b761ae5b44619e02924b5fa87f37 : SUCCESS in 6m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b1b5664a54864189b0784603d71f90fe : SUCCESS in 4m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ff6ea886ad5a4fc1a9670af4230f90ae : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b6cdbdd156449c3babbcb0fe61dced3 : SUCCESS in 12m 35s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/d8265881349e4f4cbf291e09272aeb0a : SUCCESS in 4m 49s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/c7485b64267d49a9bb7387250f82e1f7 : SUCCESS in 6m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8111fea328a64afab5f85ac07e40ad40 : SUCCESS in 4m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4b1ba7d2a2d2467a9c2d40dbcbe95ff6 : SUCCESS in 7m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/760672fa00d9404dba4f40b3ae7942d9 : SUCCESS in 4m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f4e48ff12794293968d3a69b65504e7 : SUCCESS in 1h 12m 07s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/22b042d3f0774386aa56d45978a4f882 : SUCCESS in 1h 31m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/359787b9c58a4100b22362970ddfb5f6 : SUCCESS in 58m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1ad5296aa09e4533aa7c6630534cda7d : SUCCESS in 1h 37m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/182e47176cad4b05954b2bdfe5721456 : SUCCESS in 1h 04m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/06f077e066084a02b0bee988529d6c89 : FAILURE in 42m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7e87fe4fadb7401ab92439d16e9f1793 : SUCCESS in 56m 07s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/1728c12e7a2b488fa6405f03fb931d0a : SUCCESS in 1h 08m 33s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aba80158b16b4220abe5e23a84eaac48 : SUCCESS in 36m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bf3fcfaea59b4f0f8436564ec432b663 : SUCCESS in 37m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c9deb125160448c896e3ec6d23cbb2b0 : SUCCESS in 2h 38m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b242ccb361ae4373a9e26ccb55951a5d : SUCCESS in 2h 09m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/92ab61c6e8034680ad519609869f0003 : FAILURE in 36m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b7139ddcb08842c982cb5eb33edadc42 : SUCCESS in 1h 04m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9fb30eecd60fc98a70569d7809a06b8c949ba21c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-29 13:26:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7b5829f5e79234750ccd8e18d04d271027f53eb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-29 14:42:50.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"645584a23ab6f68c8a494b3cdb869cc8e8460b73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 16:55:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c1df0d434ab47dd992bf7a9933e5691\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c671aa6d11214501b8beba4c686f8cdb : SUCCESS in 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b87d942386d74f26929d83c8e3824642 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6817c6a617445b08c9bb4ff471d7ebf : SUCCESS in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/40d0d9ed12e64a96833bb0e73ffe1f07 : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f86ba673067d475eb1643b710a3d1587 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12dbe81463b54ff9914b5c8b15dac637 : SUCCESS in 11m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/364bd1ea6b7c4683be332dac6f0c01a8 : SUCCESS in 8m 02s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/623ec4aecda9433c9052ee7711f17d70 : SUCCESS in 3m 38s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/5ba2e9f8f90d4ff4afd637ad368901f2 : SUCCESS in 4m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8542fd8e59be453183146919d9f90960 : SUCCESS in 3m 45s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2208a4dfd6094037a4c6ac60ebb18c4e : SUCCESS in 6m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c0d4556ff0d6438d82e49e3b0c722c0e : SUCCESS in 1h 05m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a90e8a1b6c36418ab72e1c2f3bcc261b : SUCCESS in 1h 32m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7ddc921f366c49b0b53c41728d85306b : SUCCESS in 1h 00m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9dfa92d5360c45b78fb1f805c09e653d : SUCCESS in 1h 37m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fcba3cdb5c0643bda6a0e4aed53e739d : SUCCESS in 1h 11m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39756bf08b0d4b0885a518ca74383a66 : SUCCESS in 36m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/39d2b5c55fce4a79868df8e48700faf5 : SUCCESS in 51m 33s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/da503ea67fd949c2b6d9caa31e7ae446 : SUCCESS in 46m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc8fe8f8b3da45f182d17a609eb96ff4 : SUCCESS in 32m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ea16a081ab9f44dfafd595e2e4a92359 : SUCCESS in 36m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5e15e63714e049858011a5de6f741308 : SUCCESS in 2h 10m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7b1f8d763afe41049157184c56cf40cc : SUCCESS in 1h 30m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/39b4b1c1cd6247ab9b1fee83ba709792 : SUCCESS in 1h 24m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/68f48ca781d64acbbaa00f7a4ebf8a92 : SUCCESS in 1h 06m 59s (non-voting)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"f9de08e485eef6bac9892440a35725e3d48c7162","revisions":{"ab5225867b580b0ef58e1ffaa5936b9a9dbdd2e8":{"kind":"REWORK","_number":1,"created":"2026-03-17 12:09:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/980908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/980908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/980908/1"}}},"commit":{"parents":[{"commit":"047276e905362e71e92c9051239d8de8ff7ee501","subject":"Use specific service user for inter-service communications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/047276e905362e71e92c9051239d8de8ff7ee501"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:09:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:09:01.000000000","tz":540},"subject":"devstack: Use common function to manage keystoneauth options","message":"devstack: Use common function to manage keystoneauth options\n\n... to simplify own logic.\n\nDepends-on: https://review.opendev.org/954751\nChange-Id: I157cd2324cf67f5e9dc0495090a3bf72995941ee\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab5225867b580b0ef58e1ffaa5936b9a9dbdd2e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab5225867b580b0ef58e1ffaa5936b9a9dbdd2e8"}]},"branch":"refs/heads/master"},"1bb44f864d272731d70e2c8ac79cd9b1d6327ceb":{"kind":"REWORK","_number":2,"created":"2026-03-17 13:51:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/980908/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/980908/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/980908/2"}}},"commit":{"parents":[{"commit":"16e34420d10334054bc90cc62b22cdf46a99d909","subject":"Use specific service user for inter-service communications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16e34420d10334054bc90cc62b22cdf46a99d909"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:09:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 13:51:00.000000000","tz":540},"subject":"devstack: Use common function to manage keystoneauth options","message":"devstack: Use common function to manage keystoneauth options\n\n... to simplify own logic.\n\nDepends-on: https://review.opendev.org/954751\nChange-Id: I157cd2324cf67f5e9dc0495090a3bf72995941ee\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1bb44f864d272731d70e2c8ac79cd9b1d6327ceb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1bb44f864d272731d70e2c8ac79cd9b1d6327ceb"}]},"branch":"refs/heads/master"},"22c780be6b0534641e147440591e519f94973a4e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-17 15:41:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/980908/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/980908/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/980908/3"}}},"commit":{"parents":[{"commit":"b255cc830021a90e7a8597c9e4d3d35afc2fd69c","subject":"Use specific service user for inter-service communications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b255cc830021a90e7a8597c9e4d3d35afc2fd69c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:09:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 15:41:18.000000000","tz":0},"subject":"devstack: Use common function to manage keystoneauth options","message":"devstack: Use common function to manage keystoneauth options\n\n... to simplify own logic.\n\nDepends-on: https://review.opendev.org/954751\nChange-Id: I157cd2324cf67f5e9dc0495090a3bf72995941ee\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22c780be6b0534641e147440591e519f94973a4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22c780be6b0534641e147440591e519f94973a4e"}]},"branch":"refs/heads/master","description":"Rebase"},"099dc0eb7234bd210049c5da83e6b3ab9b4fe983":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-23 10:16:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/980908/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/980908/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/980908/4"}}},"commit":{"parents":[{"commit":"1c2c882ffaa0d0b246c0e45b80b5744e2038fdd9","subject":"Use specific service user for inter-service communications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c2c882ffaa0d0b246c0e45b80b5744e2038fdd9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:09:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-23 10:16:05.000000000","tz":540},"subject":"devstack: Use common function to manage keystoneauth options","message":"devstack: Use common function to manage keystoneauth options\n\n... to simplify own logic.\n\nDepends-on: https://review.opendev.org/954751\nChange-Id: I157cd2324cf67f5e9dc0495090a3bf72995941ee\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/099dc0eb7234bd210049c5da83e6b3ab9b4fe983"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/099dc0eb7234bd210049c5da83e6b3ab9b4fe983"}]},"branch":"refs/heads/master"},"6cc6e0c0c2a125bcc473356f8539c525759cb40f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-23 10:35:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/980908/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/980908/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/980908/5"}}},"commit":{"parents":[{"commit":"6f359b17f50280b4e17c09abd1f7d2ee51fd80ca","subject":"Use specific service user for inter-service communications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f359b17f50280b4e17c09abd1f7d2ee51fd80ca"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:09:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-23 10:35:22.000000000","tz":0},"subject":"devstack: Use common function to manage keystoneauth options","message":"devstack: Use common function to manage keystoneauth options\n\n... to simplify own logic.\n\nDepends-on: https://review.opendev.org/954751\nChange-Id: I157cd2324cf67f5e9dc0495090a3bf72995941ee\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cc6e0c0c2a125bcc473356f8539c525759cb40f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cc6e0c0c2a125bcc473356f8539c525759cb40f"}]},"branch":"refs/heads/master","description":"Rebase"},"62828d638d3414489c3228d56ffd7668e94e035a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-05-04 14:17:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/980908/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/980908/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/980908/6"}}},"commit":{"parents":[{"commit":"f1052c694a7003bc9a252094f07761fbe8809857","subject":"Use specific service user for inter-service communications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1052c694a7003bc9a252094f07761fbe8809857"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:09:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-04 14:17:30.000000000","tz":0},"subject":"devstack: Use common function to manage keystoneauth options","message":"devstack: Use common function to manage keystoneauth options\n\n... to simplify own logic.\n\nDepends-on: https://review.opendev.org/954751\nChange-Id: I157cd2324cf67f5e9dc0495090a3bf72995941ee\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/62828d638d3414489c3228d56ffd7668e94e035a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/62828d638d3414489c3228d56ffd7668e94e035a"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"6cc6e0c0c2a125bcc473356f8539c525759cb40f","theirs":"f1052c694a7003bc9a252094f07761fbe8809857","contains_conflicts":false}},"696ff2b0a944690c786998b3b911cced621eaa19":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-05-06 03:51:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/980908/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/980908/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/980908/7"}}},"commit":{"parents":[{"commit":"8d8b9128aa5c65e888830c6ef19c86eeb3348f4a","subject":"Use specific service user for inter-service communications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d8b9128aa5c65e888830c6ef19c86eeb3348f4a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:09:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-06 03:51:45.000000000","tz":0},"subject":"devstack: Use common function to manage keystoneauth options","message":"devstack: Use common function to manage keystoneauth options\n\n... to simplify own logic.\n\nDepends-on: https://review.opendev.org/954751\nChange-Id: I157cd2324cf67f5e9dc0495090a3bf72995941ee\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/696ff2b0a944690c786998b3b911cced621eaa19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/696ff2b0a944690c786998b3b911cced621eaa19"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"62828d638d3414489c3228d56ffd7668e94e035a","theirs":"8d8b9128aa5c65e888830c6ef19c86eeb3348f4a","contains_conflicts":false}},"f9de08e485eef6bac9892440a35725e3d48c7162":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-06-29 14:42:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/980908/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/980908/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/980908/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/980908/8"}}},"commit":{"parents":[{"commit":"ce92fff16e8b6f5b199be5c2ef33726582c339fd","subject":"Use specific service user for inter-service communications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce92fff16e8b6f5b199be5c2ef33726582c339fd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 12:09:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-29 14:42:50.000000000","tz":0},"subject":"devstack: Use common function to manage keystoneauth options","message":"devstack: Use common function to manage keystoneauth options\n\n... to simplify own logic.\n\nDepends-on: https://review.opendev.org/954751\nChange-Id: I157cd2324cf67f5e9dc0495090a3bf72995941ee\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9de08e485eef6bac9892440a35725e3d48c7162"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9de08e485eef6bac9892440a35725e3d48c7162"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"8d8b9128aa5c65e888830c6ef19c86eeb3348f4a","ours":"696ff2b0a944690c786998b3b911cced621eaa19","theirs":"ce92fff16e8b6f5b199be5c2ef33726582c339fd","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
