)]}'
{"id":"openstack%2Foctavia~981880","triplet_id":"openstack%2Foctavia~master~Idaf84dcd6946418ed8baf55d6596b2de6064e9f7","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-06-30 06:57:42.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-30 06:57:42.000000000","reason":"Change was abandoned"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-06-11 15:11:30.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Idaf84dcd6946418ed8baf55d6596b2de6064e9f7","subject":"Bump testing dependencies","status":"ABANDONED","created":"2026-03-24 12:36:05.000000000","updated":"2026-06-30 06:57:42.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"d2928256cec3fb3d11a207f91b6283841b9a004d","_number":981880,"virtual_id_number":981880,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-10 21:23:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-10 21:33:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-10 21:33:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-24 15:40:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-10 18:42:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-06-10 21:33:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-06-10 21:33:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-06-11 13:46:32.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"19072f590c6aba80c89e4455dbc31bc19fdc5093","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-03-24 12:36:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62a1e23faba5f50ef3129beacca86c7488f47658","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 15:40:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d5b8f7125c94a34b0c9af3a23a43920\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e7e0b442500d47088ff1fc9b1bfce945 : SUCCESS in 4m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7f31aaeddd91409a863afd8025d492c6 : SUCCESS in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5dc515be49a644aa90e422087a851fe4 : SUCCESS in 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f80dc640394421ab033606c1e8f55bd : POST_FAILURE in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa013eed923342908c51d57afee6802a : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/23de8c757de14d7588a7cbe54d4f3fd6 : POST_FAILURE in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/45bb9f1d6afb4dda9fb4aab229dec561 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11ae2c2aeeb141239ffee6f94aed7050 : POST_FAILURE in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8521fb6ddd4849e2a8be4d5c512b423a : SUCCESS in 5m 19s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/883c77207530457a94a2fbad93e0a979 : SUCCESS in 3m 04s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/a3bdf8e9cdd44acfa80deacdaa059aae : SUCCESS in 8m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a5c530392d76496bb6a23bd2b43ac3bc : POST_FAILURE in 4m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a1259f72a8ce45ebaca90fe20a5a0943 : POST_FAILURE in 4m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a1466d6ddf4b4253a471f5ff7c58d7b0 : SUCCESS in 10m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c4ce916269fb422698b69f75bbad5d7d : POST_FAILURE in 49m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/188e93d48abd4e7db34f5f3e9fb2918a : POST_FAILURE in 1h 33m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/514c212d9dca44f696ac1951b65221dc : SUCCESS in 1h 17m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3182816a44ab4c9a96f1df4c065da472 : SUCCESS in 1h 38m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5e835b602a03431cb6898ce25fe4c16b : POST_FAILURE in 1h 22m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3f7daf402be142fdba7a177ac703e460 : SUCCESS in 36m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4e82aac139334c59a8e46d1a9fb7f964 : FAILURE in 51m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1ed4fd543c664444abca0df2d99c68f8 : FAILURE in 51m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/24adc6df26b343e8b114d8ea87aa8823 : TIMED_OUT in 3h 03m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1d84c9c24f6d4f02a7676355226ec6bb : POST_FAILURE in 1h 36m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/6316977b1225450c901325f830da2dc3 : POST_FAILURE in 1h 30m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0a6478388d144fbf9deca06e0acd7f91 : SUCCESS in 1h 04m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8627eae694c7dcefde990193ae59de7e4de44fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-10 18:42:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_29244\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":2},{"id":"e6e1e932d90180fd513546a71ba27e912666fe3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 21:23:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec0455250b40426f9e75ff64b765e095\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fe2cc00a0fff48dabdf72fa59e08c956 : SUCCESS in 3m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b4c3ca7dbd6c4721b74a097554ca2cae : SUCCESS in 1m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fefec8fa105a4b81880b0fbc1276c9a1 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4a103df7f844b4a8fa4b58d3d94c75b : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32ca62b3a8544e7ca668a15e4be23cde : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9602c92acac540a4b47c7605437eed03 : SUCCESS in 3m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a345502118c34c7db79ad10ed43b76d4 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6261b9e6dfb146e3b4ece82f97296859 : SUCCESS in 12m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8ab6ccfe7c746b29fef31cc545a5c53 : SUCCESS in 4m 23s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/c3f8f0ecaca84b399c698ed2814fe860 : SUCCESS in 3m 13s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/2afc2d8e71a64ed98af4c57d7d6f48b1 : SUCCESS in 5m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9c37cb38c71a4cc58b8efd35abcd1c55 : SUCCESS in 2m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5a3cd2676165481d84ac1f0c3f54fa1a : SUCCESS in 5m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/06d8f5988d844fab84ab15151b3d9035 : SUCCESS in 1h 02m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5c327cfad5f2456298a4e8e2f125ab14 : SUCCESS in 1h 32m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/80b48cba71cb481bac7151d33fa7bd52 : SUCCESS in 1h 08m 13s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c9f67fabe2a5499595744e8a4ad85ec7 : SUCCESS in 1h 38m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/770746543989433aa2673d0d102d0e9a : SUCCESS in 1h 06m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/91f77da9f8c64fdab516c313ba7c52f6 : SUCCESS in 40m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39f176b8b90b4884a6342553b6bae1ea : SUCCESS in 38m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b0fc838407ca457482d8dd651109990c : SUCCESS in 38m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a493bc4a7cb74f1a8adbcdd6ef1d7f37 : SUCCESS in 2h 39m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3b8289a652554c068fcb6d91df0af382 : SUCCESS in 1h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/d63a8c7fea904fde8677cf622470b9bb : SUCCESS in 1h 29m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3a21a267451c4bf78d4f6229d05d63d2 : SUCCESS in 1h 00m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5352b4ce199c63a9f51070c105984d5d8a7eefbe","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-10 21:33:23.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74318630832dca6ca2a681cbb86f62d7eefa772b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-11 13:46:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b7b51fa0398328277ba80de44ce04221311ddf2e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-06-11 15:11:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d2928256cec3fb3d11a207f91b6283841b9a004d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-06-30 06:57:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"87579adf173396688b0426a45d9dd327b7bc1a47","revisions":{"c8ed75759685619ecc0740ffb51a80d5e3d6f5d0":{"kind":"REWORK","_number":1,"created":"2026-03-24 12:36:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/981880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/981880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/981880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/981880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/981880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/981880/1"}}},"commit":{"parents":[{"commit":"79cf1f787fa9c03adba8575dc94324b97963f84e","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79cf1f787fa9c03adba8575dc94324b97963f84e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-03-24 09:36:10.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-03-24 12:34:52.000000000","tz":60},"subject":"Bump testing dependencies","message":"Bump testing dependencies\n\nUpdated testing dependencies to prepare for python 3.14\n\nRequires:\n- hacking\u003e\u003d8.0.0, it fixes pep8 job with python3.14\n\nChange-Id: Idaf84dcd6946418ed8baf55d6596b2de6064e9f7\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8ed75759685619ecc0740ffb51a80d5e3d6f5d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8ed75759685619ecc0740ffb51a80d5e3d6f5d0"}]},"branch":"refs/heads/master"},"87579adf173396688b0426a45d9dd327b7bc1a47":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-10 18:42:28.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/981880/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/981880/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/981880/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/981880/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/981880/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/981880/2"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-03-24 09:36:10.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-06-10 18:42:28.000000000","tz":0},"subject":"Bump testing dependencies","message":"Bump testing dependencies\n\nUpdated testing dependencies to prepare for python 3.14\n\nRequires:\n- hacking\u003e\u003d8.0.0, it fixes pep8 job with python3.14\n\nChange-Id: Idaf84dcd6946418ed8baf55d6596b2de6064e9f7\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/87579adf173396688b0426a45d9dd327b7bc1a47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/87579adf173396688b0426a45d9dd327b7bc1a47"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"79cf1f787fa9c03adba8575dc94324b97963f84e","ours":"c8ed75759685619ecc0740ffb51a80d5e3d6f5d0","theirs":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
