)]}'
{"id":"openstack%2Foctavia~984944","triplet_id":"openstack%2Foctavia~stable%2F2024.2~Id5a80a915fd66bcb252dc8f7636453d4cf318968","project":"openstack/octavia","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-04-20 09:00:15.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Id5a80a915fd66bcb252dc8f7636453d4cf318968","subject":"Make [task_flow].persistence_connection secret","status":"MERGED","created":"2026-04-16 13:50:22.000000000","updated":"2026-04-27 15:37:06.000000000","submitted":"2026-04-27 15:35:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984944","meta_rev_id":"f453e43ec35b183397b84bce14d09017f21d2986","_number":984944,"virtual_id_number":984944,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-27 15:35:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-04-16 13:58:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":2,"date":"2026-04-27 13:58:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":1,"date":"2026-04-27 13:58:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-16 13:58:08.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-04-16 16:55:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0556603e4d8f95936ffa369c819c4239d227f23d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-04-16 13:50:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9af62f10ff2730b3e81ae556434ac4572d43e324","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-04-16 13:58:08.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7aa77325ce8d098d14bd7b82b1dbb540c807330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 16:55:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/512d0942c1e84872822453ede6cc33ff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51c543af283c485bac8e8854dfb9e2f5 : SUCCESS in 16m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80cb2b3fbff348fb977168e8657a9215 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ff6264ea69d4398887e0fd90b281d02 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6033048344bf4a8383dd5b4c540e2690 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72b82bdda4df4671be1f3528d00ce252 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/017e57aef0a54ade8ccdc8d7f9263915 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c95ee177be54f35ab6018cf48e909fc : SUCCESS in 5m 19s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/71d1d397788b469e8072a21a3455d3c3 : SUCCESS in 4m 42s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/7d1162e0eebb428f96b28fcd1b11e710 : SUCCESS in 7m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a7dcb1426c464914a0c9dfb4627d16ca : SUCCESS in 2m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8784452db62d43fd98fa2fb17fb4b641 : SUCCESS in 6m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/332606d24a354a6e9ac6f2290e9be38f : SUCCESS in 6m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/53646b36727f49b0adebd36f81549abc : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/eedc916da6604576b1db2781399eb759 : SUCCESS in 1h 02m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/0cba2ddd58d1496bb3e78e1e2dd96698 : FAILURE in 1h 19m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f7927624bc1c465a9f2f201f0dd07fbb : SUCCESS in 59m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/1b761ce1df204bc6b004e55dc7654320 : FAILURE in 1h 29m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/bde114c4f370430a8b5813d5ea2c22de : SUCCESS in 59m 08s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/1bdbaa4ad601486aa1b32bcca4a98e43 : SUCCESS in 33m 24s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4199c044bff74911bc7c97b6a98eeb9e : SUCCESS in 39m 56s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-2 https://zuul.opendev.org/t/openstack/build/87bc92e6a49045e998c033d4be58240f : FAILURE in 19m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4b225893c0114395af953575cb147653 : SUCCESS in 31m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-2 https://zuul.opendev.org/t/openstack/build/55d2d39669674c7ca7ab08cd4c2d4847 : SUCCESS in 29m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-2 https://zuul.opendev.org/t/openstack/build/e5f62ebc880a43a585a72e2fac4a85dc : TIMED_OUT in 3h 03m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-2 https://zuul.opendev.org/t/openstack/build/35f5cdd1440d4eeab669a40292877c45 : SUCCESS in 1h 27m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-2 https://zuul.opendev.org/t/openstack/build/5d9f1ac00349489bad354a4b0310b1b1 : FAILURE in 16m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db42a2fbf9c06a408683951ba81aa54e2e2ed56a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-04-20 09:00:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c869771f3a7b836d1642dca1f53d9ec483657a78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 10:31:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46ce3dd626424cb097dda8b345668c95\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32b6c9d826b3461d80cb4fa65960c99e : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d7caa2e749b49d1942aa272c7248f62 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40ca6e51c8e64883b3ab83b23bf6f2e6 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1d77f6aeb854d05a0debc6b6388667c : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e4686bbae914455b1f58dca4c469776 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3ccb619da5049f8a0202cb6425fdc4b : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04269beecf694f6587cabd5a3963312e : SUCCESS in 5m 42s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bb40072e4a8e4950a966d13ad5058157 : SUCCESS in 4m 09s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/1473310ce79e4ed7b338ee5f7bf5ca39 : SUCCESS in 8m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/04b77b50cf0645c08f398f82d1f2b179 : SUCCESS in 4m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a0b94cfced4f498ca4aaa7fbcf54e91a : SUCCESS in 3m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6a961ca26166414fa778eb9f450cc7a4 : SUCCESS in 9m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/984c89ee361441268975171821130ed2 : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b791d76c19004bacadebfa8e8c54a93a : SUCCESS in 1h 00m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/395a40b272ef4deea6ca571d6475d9e9 : SUCCESS in 1h 24m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/66f1263ad4d7469bb132958efef5e473 : SUCCESS in 1h 04m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/d6cc7533bc224a77bb191b11c4e8a688 : SUCCESS in 1h 29m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/d906fdefce0f4d9aa4a0ec4d80a4af7d : SUCCESS in 56m 44s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b27cf04cc875450ba03a58c74a932e0c : SUCCESS in 30m 02s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4dc8178657894434a77646f61d4821d4 : SUCCESS in 37m 38s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-2 https://zuul.opendev.org/t/openstack/build/1484f6ec4a274e9da21f173d30e44042 : FAILURE in 17m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/bee5a44e896e434fb3499bc3945465b3 : SUCCESS in 29m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-2 https://zuul.opendev.org/t/openstack/build/8eefc5c622af4d85a19687e757765b35 : SUCCESS in 29m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f43149b784644bf3baf64824c6353cdd : FAILURE in 41m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-2 https://zuul.opendev.org/t/openstack/build/9a64f4e3cc5e45b3a9b2e26653b1c311 : SUCCESS in 1h 23m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-2 https://zuul.opendev.org/t/openstack/build/cb88812388c2419e8b5b062910b6206a : FAILURE in 15m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"745f98fb0f18b380e0a08a4598d44697b8e7ccf0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-04-27 13:58:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"785efcb5d17bfcc27d189c451f6affe01f914ffc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 13:58:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d0d7f94e1f28bb8678b4d1faf8f8208087c235d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 15:35:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fbd5b1f865b4ea9b6da5a24c2bc7c67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57589f2b224e4dda8455a103294227b0 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bc4f350ae224dd8b5497aeebe0213bd : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c4749fcae5b4403590436d0f1af8d751 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2b20a303db84b1c9c7dc5ae1378bab2 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f310e5ba8325450b83bd2e2c41da995d : SUCCESS in 5m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a7ef46a3c1254014a2122d90c4c8ba9e : SUCCESS in 5m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cea7e67eb57b4ec196d5fe53da88f8f2 : SUCCESS in 9m 32s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/3bdd35c6665743f29de52fdd7966514b : SUCCESS in 1h 07m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/cefa4dafecfc442f90b3dcd8fc3dc590 : SUCCESS in 1h 34m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/06eec7626f744290af5b5b9415b39373 : SUCCESS in 1h 12m 54s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/ba000280ba92411baed0ab4e6bc67bac : SUCCESS in 39m 36s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/76e5eb6aaf6b43e2ae886e4befbe1867 : SUCCESS in 46m 29s","accounts_in_message":[],"_revision_number":1},{"id":"22164fe36b10922c1361e63697102552008ceeee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 15:35:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f453e43ec35b183397b84bce14d09017f21d2986","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 15:37:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1929e5147ecb4a69a5b5fe70b4161c23\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4a639ed20244c3cbfc36873e36f9d8a : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31b4fccbb739453080d494ce698e037c : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e5a1d860c4a52634952e03dc56839bbd7feba72","revisions":{"3e5a1d860c4a52634952e03dc56839bbd7feba72":{"kind":"REWORK","_number":1,"created":"2026-04-16 13:50:22.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/44/984944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/984944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/984944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/984944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/984944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/984944/1"}}},"commit":{"parents":[{"commit":"621edb197f7d45df6ee2c5136c0a14e229e5aab2","subject":"Remove failing drivers from enabled_provider_drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/621edb197f7d45df6ee2c5136c0a14e229e5aab2"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-07-24 09:03:29.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-04-16 13:50:17.000000000","tz":120},"subject":"Make [task_flow].persistence_connection secret","message":"Make [task_flow].persistence_connection secret\n\nChange-Id: Id5a80a915fd66bcb252dc8f7636453d4cf318968\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit bad5457e97ce28a3d87841f5061bd61d03876241)\n(cherry picked from commit 1a04c5c4010caf16698d9ffe6fd7272259358fd1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3e5a1d860c4a52634952e03dc56839bbd7feba72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3e5a1d860c4a52634952e03dc56839bbd7feba72"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
