)]}'
{"id":"openstack%2Foctavia~989914","triplet_id":"openstack%2Foctavia~master~I0d225f01d808d6e48d554978206ab12440e84418","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-09-10 13:08:23.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-09-10 11:31:15.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-10 13:08:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0d225f01d808d6e48d554978206ab12440e84418","subject":"Remove legacy Upstart configuration","status":"MERGED","created":"2026-05-24 22:46:14.000000000","updated":"2026-09-10 13:10:10.000000000","submitted":"2026-09-10 13:08:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989914","meta_rev_id":"daa5eab89c4dd9453c4deb298c6d0caa201f5eb8","_number":989914,"virtual_id_number":989914,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 13:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-08-31 10:31:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-31 08:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-10 11:31:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-08-31 10:31:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-28 13:35:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-08-28 13:35:20.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-08-28 16:09:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2a43527f960984f61ccb8692d6f1515e8a37796e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-24 22:46:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0be602d602e33e6707af3e6c754deca8bac2863","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-24 22:47:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"8604463d077a58e79e3107e9c4a0e777cd386527","tag":"autogenerated:gerrit:restore","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-28 13:32:50.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"8855bfcfd4652504703fa88a7ef6b76c51a5fcbe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-28 13:33:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"39ee2710f7a4d5141e27fa4907a59bf34ef739f2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-28 13:33:09.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e23bc04959ac04c6776c96ff5d2f4bd8542eddee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-28 13:34:38.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5b246735056cc63541a21d5c362e0149f7c981a7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-28 13:35:08.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"7cecbf77a90a3817164b96d854ca8bb2acf70787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 16:09:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/387458a3c0424eb7adf6742cc46cbd35\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef8e5702a98844f7a88dfd17dfb4af4d : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/058ccb00b2b8420f84893f2e55398a15 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/018f7a1ae10445fdbb24699bf8078f2d : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dcf8e01558ea4745b3664c8b17bd4f31 : SUCCESS in 4m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d602142592ac450c83e4fb0057dbb24e : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a78761e3050a4f869c4d372106d47e02 : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c49dd9640c174a6bb42b42ea66d4d983 : SUCCESS in 9m 39s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/43221da8abcc427584d8d408b7155705 : SUCCESS in 4m 46s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/9b11bec8dcd24b32baff1f8493761be2 : SUCCESS in 5m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/133c814c048c4476a0471b3dff0aae4d : SUCCESS in 3m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2339c931096645eea1b51acb1eaa90f1 : SUCCESS in 3m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/06d5090d708e4da0877786bff9e525a0 : SUCCESS in 1h 11m 16s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6593662204b746b3b06abc9908705b0c : FAILURE in 22m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/67153df7db6744b38b92e9aff49b156c : SUCCESS in 59m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6b05043137574c02a119e974cdcaba07 : SUCCESS in 1h 30m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e8181f93a18d473daee3c28b931f4ee5 : FAILURE in 21m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b4d6b2eb2ff9461fb6288d9ce969d172 : SUCCESS in 39m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2d081a500fd04442bb2746b6da37ee52 : SUCCESS in 46m 06s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8e38edb9730a46cfa7be9a4db752e694 : FAILURE in 21m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ea030ff62374956b4ca91c6991552eb : FAILURE in 22m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/10347d70172a4f978b67b358b2e6c9cd : FAILURE in 23m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ae133e7d3693460db5f4deee7e39916a : SUCCESS in 2h 33m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c29731c015c54c2e9111cfdabc416faf : SUCCESS in 1h 25m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/1ca5ecdeb0e849a08b61735c73da3736 : FAILURE in 7m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9ae4c8959066452794bb1f9fdcacfa09 : FAILURE in 7m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a6844f5084b82e17c1766a1ef17887c1bc941ee1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-31 06:51:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7f684427fc406c914d0e919ceadab58686333cd6","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-31 07:52:30.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"aa331870826ad4ef873d01a75cad9fbd56ca8b6f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-31 07:55:42.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"d7206ab7d800e256d684f091418441150f8d9822","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-31 07:55:44.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"331c83c6e85e077b40dcff0372b677fcf6a41c68","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-31 07:56:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e7edfe2b853e386523836294628a5e7b05c6ef15","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-31 08:25:45.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"113a904fe36c39e1e284ce07ff4f4cad984a86d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 10:27:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b560aa63c61841e89fc60dcd4ff2fcc9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee334c0787ab4c70879e74f4b83b4c35 : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/235b591d0cea4eedb6bebe41582f298c : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b5751907f7e4d2d812d320af56932a4 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d052601a8e4465b82d09e20ac6939b3 : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0bd551d7a65b468ea4dc101cf34fbff5 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff82642fc3824650befd70fc96e65593 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/251d8381fef745dab23ef3b9dd3225bf : SUCCESS in 4m 10s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/17514ca292a84b64bb83c35e39694d7c : SUCCESS in 4m 50s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/ce4191da21664cdcb77c4be9dcf358e0 : SUCCESS in 4m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e34bfa98a4a942ee9884cbfaef4ed49a : SUCCESS in 3m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bf627f6727634dae81d7c4d2176a1a06 : SUCCESS in 6m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e189ea724a3b46b7920fc7bd4910047f : SUCCESS in 36m 58s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fb539fcbea644dd3bb0a62f05e491aba : SUCCESS in 1h 25m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8d20b441031b4ce9960d45842ca56290 : SUCCESS in 1h 06m 26s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b46b7956fc8b4346bb7b84f588e8f468 : SUCCESS in 1h 42m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7d8071a4041746b694311818f3297012 : SUCCESS in 1h 30m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/471463e41aa641d195504c1a00c24bc1 : SUCCESS in 55m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0f86d94478c047e290cf1a22f6e25f7c : SUCCESS in 43m 59s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b450c1fc08aa4662b4ce8bee34a59ef6 : SUCCESS in 1h 08m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13cb16dd51f648e0baf4605a4ef9bf7f : SUCCESS in 41m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5c3b02836a914c5d902b586c65c3d79f : SUCCESS in 55m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/84d9d79331ac4d1db450cb6c571f7749 : SUCCESS in 2h 25m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f59a521933d24a6da883320ff4e63c53 : SUCCESS in 1h 57m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/404f49b9971742e78ccc883e03e8dcbc : SUCCESS in 1h 31m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3a9e270c51e04dcb981034fca444abcd : SUCCESS in 1h 00m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"68890523b6dd603c69218ae6bdc90d05386c1132","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-31 10:31:35.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f3440268b48ff3555379447cefbac09a4a12fb4a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-09-10 11:31:15.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5c7d2467ff31f5fef5a71947b515afbe18b19e47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 11:32:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"639cd33f752b6a7109f324f5467a8738dbd352ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:08:23.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d0a4152985343268d06695551f45a76\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ab8cb43a6f146d0ae78a2d37145e86f : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8c0d5b1b5e746159f4279533089f5f4 : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe3a38eacd9a4872b588363b5c89b88c : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55e825dfda57423a98ce650848dd74ea : SUCCESS in 12m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2894d3b3559047409cba6e6b71c34f3d : SUCCESS in 9m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/78510aecf20044b4bbe5f85eebea855d : SUCCESS in 3m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/09762e5f9d9046628005990d7e7a78e8 : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d96b49b97fcf470ea56693d585a38218 : SUCCESS in 1h 12m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b8f16e6bbfae4e3f9e5de656481f9fae : SUCCESS in 1h 34m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d78570e3cd884c31a89885bb321d287a : SUCCESS in 1h 00m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/057ec6dd66b0431fa7828de8ef74aa06 : SUCCESS in 55m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1183efd300ae4910be0a5c3fa98e8006 : SUCCESS in 57m 02s","accounts_in_message":[],"_revision_number":5},{"id":"023ed017711e1457fa6efd9c580e5733bd461d27","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:08:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"daa5eab89c4dd9453c4deb298c6d0caa201f5eb8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:10:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/835211d2581b402bb59bd00a398c55b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6b23430464a41f88b7d3ee131134feb : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2d01e3f57ede649af659b668bc874275e231d104","revisions":{"dac8147e25a86e6a3c1cbe40a404caa4ccdfe3e6":{"kind":"REWORK","_number":1,"created":"2026-05-24 22:46:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/14/989914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/989914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/989914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/989914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/989914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/989914/1"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-24 22:46:14.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-24 22:46:14.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I0d225f01d808d6e48d554978206ab12440e84418\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dac8147e25a86e6a3c1cbe40a404caa4ccdfe3e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dac8147e25a86e6a3c1cbe40a404caa4ccdfe3e6"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"32677156167a8a450575df00860f9288aabb58bc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-28 13:33:03.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/14/989914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/989914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/989914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/989914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/989914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/989914/2"}}},"commit":{"parents":[{"commit":"fbc4af60a9dfa922343e0f86f0541b4971c1b908","subject":"Merge \"Fix VRRP split-brain on amphora failover in ACTIVE_STANDBY\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fbc4af60a9dfa922343e0f86f0541b4971c1b908"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-24 22:46:14.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-28 13:33:03.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I0d225f01d808d6e48d554978206ab12440e84418\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32677156167a8a450575df00860f9288aabb58bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32677156167a8a450575df00860f9288aabb58bc"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","ours":"dac8147e25a86e6a3c1cbe40a404caa4ccdfe3e6","theirs":"fbc4af60a9dfa922343e0f86f0541b4971c1b908","merge_strategy":"recursive","contains_conflicts":false}},"059315a6d5129d031ddc39f4938ea9f3996c7c08":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-08-28 13:33:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/14/989914/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/989914/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/989914/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/989914/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/989914/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/989914/3"}}},"commit":{"parents":[{"commit":"fbc4af60a9dfa922343e0f86f0541b4971c1b908","subject":"Merge \"Fix VRRP split-brain on amphora failover in ACTIVE_STANDBY\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fbc4af60a9dfa922343e0f86f0541b4971c1b908"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-24 22:46:14.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-28 13:33:09.000000000","tz":0},"subject":"Replace outdated usage of egrep by grep","message":"Replace outdated usage of egrep by grep\n\nChange-Id: I0d225f01d808d6e48d554978206ab12440e84418\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/059315a6d5129d031ddc39f4938ea9f3996c7c08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/059315a6d5129d031ddc39f4938ea9f3996c7c08"}]},"branch":"refs/heads/master","description":"Edit commit message","conflicts":{"base":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","ours":"dac8147e25a86e6a3c1cbe40a404caa4ccdfe3e6","theirs":"fbc4af60a9dfa922343e0f86f0541b4971c1b908","merge_strategy":"recursive","contains_conflicts":false}},"8133b422861cdbf9e93091498e40e3a4a2fbcebe":{"kind":"REWORK","_number":4,"created":"2026-08-28 13:34:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/14/989914/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/989914/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/989914/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/989914/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/989914/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/989914/4"}}},"commit":{"parents":[{"commit":"fbc4af60a9dfa922343e0f86f0541b4971c1b908","subject":"Merge \"Fix VRRP split-brain on amphora failover in ACTIVE_STANDBY\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fbc4af60a9dfa922343e0f86f0541b4971c1b908"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-24 22:46:14.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-28 13:34:38.000000000","tz":0},"subject":"Replace outdated usage of egrep by grep","message":"Replace outdated usage of egrep by grep\n\nChange-Id: I0d225f01d808d6e48d554978206ab12440e84418\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8133b422861cdbf9e93091498e40e3a4a2fbcebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8133b422861cdbf9e93091498e40e3a4a2fbcebe"}]},"branch":"refs/heads/master"},"2d01e3f57ede649af659b668bc874275e231d104":{"kind":"REWORK","_number":5,"created":"2026-08-31 07:55:42.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/14/989914/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/989914/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/989914/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/989914/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/989914/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/989914/5"}}},"commit":{"parents":[{"commit":"fbc4af60a9dfa922343e0f86f0541b4971c1b908","subject":"Merge \"Fix VRRP split-brain on amphora failover in ACTIVE_STANDBY\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fbc4af60a9dfa922343e0f86f0541b4971c1b908"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-24 22:46:14.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-31 07:55:42.000000000","tz":0},"subject":"Remove legacy Upstart configuration","message":"Remove legacy Upstart configuration\n\nChange-Id: I0d225f01d808d6e48d554978206ab12440e84418\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d01e3f57ede649af659b668bc874275e231d104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d01e3f57ede649af659b668bc874275e231d104"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
