)]}'
{"id":"openstack%2Foctavia~993871","triplet_id":"openstack%2Foctavia~master~Ib576661dcc859f3bd07cb8db1e6a1449dde53a3e","project":"openstack/octavia","branch":"master","attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-06-18 09:21:14.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"34451":{"account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"last_update":"2026-06-19 02:52:54.000000000","reason":"\u003cGERRIT_ACCOUNT_3031\u003e replied on the change","reason_account":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"}}},"removed_from_attention_set":{"3031":{"account":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"last_update":"2026-06-19 02:52:54.000000000","reason":"\u003cGERRIT_ACCOUNT_3031\u003e replied on the change","reason_account":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"}}},"hashtags":[],"change_id":"Ib576661dcc859f3bd07cb8db1e6a1449dde53a3e","subject":"Change response codes for unsupported provider options.","status":"NEW","created":"2026-06-18 04:39:26.000000000","updated":"2026-06-19 02:52:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"700449e99fc0c7617ed4e3e233f0689648de8ca6","_number":993871,"virtual_id_number":993871,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-18 11:59:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":-1,"date":"2026-06-18 13:56:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-18 07:14:07.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2026-06-18 07:52:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-18 08:17:31.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-06-18 13:56:17.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"}],"messages":[{"id":"48462401d08c1ee45616672ceb96431af67280cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2026-06-18 04:39:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc260155c191cc21669e7fe2ee20727c50192a1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2026-06-18 04:40:39.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d0f4dafedde82590d5844922c40e0a51985de171","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2026-06-18 04:41:45.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b67995f14e918e5d424e3dbe0b08fc8da03e5247","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2026-06-18 05:20:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"67ee73d3a95a12cb7a4a518e3620078432e34612","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2026-06-18 05:21:51.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"dd1394bbd72d9cfc6968458e06a72626501f044e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2026-06-18 05:23:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c3a98f5a51b07824aa40d85899e723f351361ac2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-06-18 07:14:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"409bce81497be452ffd5419e6c9dc140662ba12d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 07:52:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9b6464a53ac47d09e913832063bb721\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/774725adbb364d4083b2e9bc9365d0aa : FAILURE in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9e310357f7e4dd8af39dedd3d8b9d58 : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7498cd57f81c4601877ceedd269a4f6f : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4bdde2dc3ef7488a87ccdb97b69b8411 : SUCCESS in 5m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e6609571efa94f35aa2f4a3fe3aa7643 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fbe4be7648e4a34aeb9d424ffa5bd72 : SUCCESS in 14m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc7ee867805e4c01ab22bce64e320310 : SUCCESS in 8m 32s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/9dec14e496974c5d85006e0277a28d89 : SUCCESS in 6m 25s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/72c0f6112ad54e6581d97210a5ead2fc : FAILURE in 5m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8602336a0a0240eca51d8486d1f766b3 : SUCCESS in 4m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ef881986f3b94f1c95395a4cff8e1e3f : FAILURE in 5m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/913f1eeb4ffc45db97db198116512e1c : FAILURE in 1h 09m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/75622524f00f48d7a3a81a38830176bc : FAILURE in 1h 31m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/21e445f83ad5444c8c24f692a62ac3ea : FAILURE in 55m 04s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/18ad89cbf1b14f5881c3a17e94b0d97c : FAILURE in 1h 35m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/15551f63c6044b1dbf4e1213ba7ddce0 : FAILURE in 1h 06m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2fa7566fca9b4d0b921f0cf602b22668 : SUCCESS in 35m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3fb7d4dc31d84eb68a4d1b72094d5ff4 : SUCCESS in 48m 02s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/013af5e859ef411fac77bb85b7fc6d61 : SUCCESS in 41m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb273911ac9c429487907d6810181535 : SUCCESS in 37m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f6ac1461015343aca07aade2ca49f5da : SUCCESS in 34m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1abdaffa939a487c8a36f66ad35f204c : FAILURE in 2h 26m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/796b67ff46c1482bb16b7f9d5996e1bf : FAILURE in 1h 28m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/cfd7a0a45c904e8b8c8c51a2d58bdb68 : FAILURE in 1h 25m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a88008b15c584f6b891db465c46f371d : FAILURE in 56m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b8115bc898073ad81d049c03ac5018bab271d300","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-06-18 08:17:31.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f1e0969b5b98ae421be59c73b628742502a71848","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2026-06-18 09:21:14.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"6788dffe1fcb3a857489ed74ff51a5c003d79af3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 11:59:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27ad5837bf43449fb7ad82491a8b8320\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84219b194d6742a793d48c7195cd5f81 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f907efe569c4056a2ccf0a76bf14286 : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3902d6243daa4b48a8ab25d9351816a8 : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8036c286d9a44592be7be4d286f8cc9a : SUCCESS in 3m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e2c8d8b5b4bd480b8868b703330807e4 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f3d42721fed4b75a2ab0200635b467f : SUCCESS in 13m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f1d169c4e754114955e021c932e8983 : SUCCESS in 8m 19s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/9dd14c56ecb84e22adcf0077730158e5 : SUCCESS in 3m 08s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/f1ab7800070b4dbb932e3b17c3b54bfe : SUCCESS in 4m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/36690ad8c9474c5788cdfe0f332b5bde : SUCCESS in 3m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/177b00bd6f40418cb546484266d68f91 : SUCCESS in 6m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/689f7c51abc248a78006dde962138041 : FAILURE in 31m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/12bb5390daad476abd3424d11039afd2 : FAILURE in 1h 38m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e6ceb8d367274302a80813a8e7a4e94c : FAILURE in 59m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4c6263a92a5f484ca2e74c3e9f1bea55 : FAILURE in 1h 34m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/95bee09b66b04719863a186312a76b74 : FAILURE in 1h 09m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3c7ddacac31d4bdaa815074bc6440cd5 : SUCCESS in 38m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e73ada0ca69c450fa9fac2dceeb90fbf : SUCCESS in 48m 46s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a8a4e88a651d45fe8c3a5ae3d7494ff6 : SUCCESS in 43m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b518e69423534753adaa470b6d3ba9f3 : SUCCESS in 39m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/218018bbb0a147ca99eb288030861a04 : SUCCESS in 40m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cdb28b8a8d724995845535a3d2173b22 : FAILURE in 2h 37m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a560ea93cff349169b46991af770c447 : FAILURE in 1h 32m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/a1266e0d55054255b54fa007884ee421 : FAILURE in 1h 22m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d4e1fa13162a42b2b95eb69efca36b2b : FAILURE in 1h 07m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"911a773a9ac3ccb783b0057bbaf5d0947bd04d9b","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2026-06-18 13:56:17.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"700449e99fc0c7617ed4e3e233f0689648de8ca6","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2026-06-19 02:52:54.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0fc79ece9815495f7b105ccad9faadf4d3c8d874","revisions":{"0705967a5e107305f7b0ee62f1ed16dd57c43e04":{"kind":"REWORK","_number":1,"created":"2026-06-18 04:39:26.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/71/993871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/993871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/993871/1"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2026-06-18 04:36:04.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2026-06-18 04:39:00.000000000","tz":600},"subject":"Change response codes for unsupported provider options.","message":"Change response codes for unsupported provider options.\n\nA user can cause 501 errors just by trying to get\na provider to do something it can\u0027t. This should be\na 400 BadRequest response.\n\nThe issue with a 5XX error is that it indicates an\nissue with the system and something operations need\nto look into. A 400 error indicates to the user that\nthe request they made was bad.\n\nFixes: #2157080\n\nChange-Id: Ib576661dcc859f3bd07cb8db1e6a1449dde53a3e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0705967a5e107305f7b0ee62f1ed16dd57c43e04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0705967a5e107305f7b0ee62f1ed16dd57c43e04"}]},"branch":"refs/heads/master"},"64bf1453fbe500029eaef0d530d1b7eff312b470":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-06-18 04:40:39.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/71/993871/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/993871/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/993871/2"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2026-06-18 04:36:04.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2026-06-18 04:40:14.000000000","tz":600},"subject":"Change response codes for unsupported provider options.","message":"Change response codes for unsupported provider options.\n\nA user can cause 501 errors just by trying to get\na provider to do something it can\u0027t. This should be\na 400 BadRequest response.\n\nThe issue with a 5XX error is that it indicates an\nissue with the system and something operations need\nto look into. A 400 error indicates to the user that\nthe request they made was bad.\n\nCloses-Bug: #2157080\n\nChange-Id: Ib576661dcc859f3bd07cb8db1e6a1449dde53a3e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64bf1453fbe500029eaef0d530d1b7eff312b470"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64bf1453fbe500029eaef0d530d1b7eff312b470"}]},"branch":"refs/heads/master"},"2bf38e92efdcd9643a3e8b4592ef51ab8d79fd3f":{"kind":"REWORK","_number":3,"created":"2026-06-18 04:41:45.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/71/993871/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/993871/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/993871/3"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2026-06-18 04:36:04.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2026-06-18 04:41:45.000000000","tz":0},"subject":"Change response codes for unsupported provider options.","message":"Change response codes for unsupported provider options.\n\nA user can cause 501 errors just by trying to get\na provider to do something it can\u0027t. This should be\na 400 BadRequest response.\n\nThe issue with a 5XX error is that it indicates an\nissue with the system and something operations need\nto look into. A 400 error indicates to the user that\nthe request they made was bad.\n\nCloses-Bug: #2157080\n\nChange-Id: Ib576661dcc859f3bd07cb8db1e6a1449dde53a3e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bf38e92efdcd9643a3e8b4592ef51ab8d79fd3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2bf38e92efdcd9643a3e8b4592ef51ab8d79fd3f"}]},"branch":"refs/heads/master"},"23d1faa6e58af939ae9e73fe32184e6ab5b191fc":{"kind":"REWORK","_number":4,"created":"2026-06-18 05:20:34.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/71/993871/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/993871/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/993871/4"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2026-06-18 04:36:04.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2026-06-18 05:20:22.000000000","tz":600},"subject":"Change response codes for unsupported provider options.","message":"Change response codes for unsupported provider options.\n\nA user can cause 501 errors just by trying to get\na provider to do something it can\u0027t. This should be\na 400 BadRequest response.\n\nThe issue with a 5XX error is that it indicates an\nissue with the system and something operations need\nto look into. A 400 error indicates to the user that\nthe request they made was bad.\n\nCloses-Bug: #2157080\n\nChange-Id: Ib576661dcc859f3bd07cb8db1e6a1449dde53a3e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23d1faa6e58af939ae9e73fe32184e6ab5b191fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23d1faa6e58af939ae9e73fe32184e6ab5b191fc"}]},"branch":"refs/heads/master"},"3eb3c4bb6aa211f18a2830112bc99e997ff5c790":{"kind":"REWORK","_number":5,"created":"2026-06-18 05:21:51.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/71/993871/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/993871/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/993871/5"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2026-06-18 04:36:04.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2026-06-18 05:21:51.000000000","tz":0},"subject":"Change response codes for unsupported provider options.","message":"Change response codes for unsupported provider options.\n\nA user can cause 501 errors just by trying to get\na provider to do something it can\u0027t. This should be\na 400 BadRequest response.\n\nThe issue with a 5XX error is that it indicates an\nissue with the system and something operations need\nto look into. A 400 error indicates to the user that\nthe request they made was bad.\n\nCloses-Bug: #2157080\n\nChange-Id: Ib576661dcc859f3bd07cb8db1e6a1449dde53a3e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3eb3c4bb6aa211f18a2830112bc99e997ff5c790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3eb3c4bb6aa211f18a2830112bc99e997ff5c790"}]},"branch":"refs/heads/master"},"3711fc91a92280b8c58affaf5b1478222f085f3c":{"kind":"REWORK","_number":6,"created":"2026-06-18 05:23:10.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/71/993871/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/993871/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/993871/6"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2026-06-18 04:36:04.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2026-06-18 05:22:52.000000000","tz":600},"subject":"Change response codes for unsupported provider options.","message":"Change response codes for unsupported provider options.\n\nA user can cause 501 errors just by trying to get\na provider to do something it can\u0027t. This should be\na 400 BadRequest response.\n\nThe issue with a 5XX error is that it indicates an\nissue with the system and something operations need\nto look into. A 400 error indicates to the user that\nthe request they made was bad.\n\nCloses-Bug: #2157080\n\nChange-Id: Ib576661dcc859f3bd07cb8db1e6a1449dde53a3e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3711fc91a92280b8c58affaf5b1478222f085f3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3711fc91a92280b8c58affaf5b1478222f085f3c"}]},"branch":"refs/heads/master"},"0fc79ece9815495f7b105ccad9faadf4d3c8d874":{"kind":"REWORK","_number":7,"created":"2026-06-18 09:21:14.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/71/993871/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/993871/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/993871/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/993871/7"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2026-06-18 04:36:04.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2026-06-18 09:21:00.000000000","tz":600},"subject":"Change response codes for unsupported provider options.","message":"Change response codes for unsupported provider options.\n\nA user can cause 501 errors just by trying to get\na provider to do something it can\u0027t. This should be\na 400 BadRequest response.\n\nThe issue with a 5XX error is that it indicates an\nissue with the system and something operations need\nto look into. A 400 error indicates to the user that\nthe request they made was bad.\n\nCloses-Bug: #2157080\n\nChange-Id: Ib576661dcc859f3bd07cb8db1e6a1449dde53a3e\nSigned-off-by: Sam Morrison \u003csorrison@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0fc79ece9815495f7b105ccad9faadf4d3c8d874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0fc79ece9815495f7b105ccad9faadf4d3c8d874"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
