)]}'
{"id":"openstack%2Foctavia~994193","triplet_id":"openstack%2Foctavia~master~I057984f1ff06baebc7ab9b140911e9e97fbe83b3","project":"openstack/octavia","branch":"master","topic":"oslotest-autospec","attention_set":{"8213":{"account":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"last_update":"2026-06-20 20:25:45.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I057984f1ff06baebc7ab9b140911e9e97fbe83b3","subject":"tests: Enforce autospec\u003dTrue on mock patches","status":"NEW","created":"2026-06-20 20:09:59.000000000","updated":"2026-06-28 16:17:49.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3a43b3b83f4d190e0d4a41d121388428b2866473","_number":994193,"virtual_id_number":994193,"owner":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-22 10:07:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-20 20:25:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5006b01eba07f15e82676503f85a3b17f8de2ffd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-20 20:09:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c833d67c8c7d0e29131e02772b52a95c196bb815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-20 20:25:45.000000000","message":"Patch Set 1: Verified-1\n\n(13 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4417f32f25946878765f13af1d048cf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b69b6a6e25174d978dbad0184edcf287 : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3babddc026b4aaaa2d9ca0435908cf4 : FAILURE in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bcb80237feed4658b77697cfa23bfd81 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75b5c381b7f54ed5a3c7f6e73ccba998 : SUCCESS in 3m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/da40320d7a8c4307ae96f44f65054733 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a509fcdccb34447f8328528ac0381247 : FAILURE in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63d173e926f440af8c1df8fe0c760feb : SUCCESS in 9m 10s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/1bcce1d950444db39742b4639a71c80f : SUCCESS in 7m 28s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/0f1e22f845584d6a807ba116c8b2c30d : SUCCESS in 3m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6852718fe50e4fa8a59b7a9a6e2a6323 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"fc432139c8618d7a983a508c76178c75bf5d98c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-22 08:48:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1f2e04a924968447e49089c69625ae5e18444fa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 09:07:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b867dbbc628c4f77a2bf4528bd999e1a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1586c16fe69d44ac907ad806a430cd8c : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15884e96f92941e2b0de29a77cccd111 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a47b0cfcaf7473788cebe3a2a4ec1b3 : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89dca3c3c68b4721aaaa39b3af0c75c9 : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2a49e5a6f24b4425b3d268520e747a5e : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d2e615c14844f7fbdf9c567c488b9ee : FAILURE in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70b980a20e704a2984b2295b42345d69 : SUCCESS in 10m 28s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/41e7c9e0f6e74831a2895820bf08833a : SUCCESS in 6m 39s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/373cac68b365417b8349d5b5343f2686 : SUCCESS in 5m 50s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2701ef7a9b9d401699b24536e21a55bf : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"83cbce314b446da1017c4f94ca491ef5ef032a59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-22 09:52:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"05f1ab42db404be22121833989afc1585ef3b79c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 10:07:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c9b2847a9324699baf6cc3c16d5296f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f69eeab7cf864daa9196cc4978bf28b8 : SUCCESS in 3m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/15cdee3064774b6bb10c04000944e9bb : SUCCESS in 1m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a630e5410234c759af1d1d46cd257e3 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3206a117b374c7b8bd8d8410ce48a1f : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc66679b17d949fc9acef40236a0ef95 : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a836222b10d48868b5c2d4c1d1a8989 : SUCCESS in 4m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5e5770d4693244c89b062618918b54bc : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8edaa824b01a4ec9b35b626a06ec46d6 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d92cae59ef124a229bbedf4ed8921d8d : SUCCESS in 9m 27s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/c5958f51a6294751bf347762c6becfa7 : SUCCESS in 3m 46s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/f43535f205e74bb5b0621f98bfc26f61 : SUCCESS in 5m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6b1ea08630264c8ebf7f5bba56d26fd4 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":3},{"id":"3a43b3b83f4d190e0d4a41d121388428b2866473","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-28 16:17:49.000000000","message":"Topic set to oslotest-autospec","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6ada43e07398195871d56e8d04501e38431ece55","revisions":{"6e9d2836216c3346bfe25728f3d8e30abf57f248":{"kind":"REWORK","_number":1,"created":"2026-06-20 20:09:59.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/93/994193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/994193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/994193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/994193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/994193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/994193/1"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-13 15:26:42.000000000","tz":0},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-20 20:09:57.000000000","tz":0},"subject":"tests: enforce autospec\u003dTrue via oslotest mock_fixture","message":"tests: enforce autospec\u003dTrue via oslotest mock_fixture\n\nUsing mock_fixture.patch_mock_module() will enforce autospec\u003dTrue on all\nthe mock.patch and similar calls. This will enforce signature checks on\nthe mocked methods and functions, catching potential bugs missed by\notherwise false positive unit tests.\n\nSets autospec\u003dFalse for:\n- octavia.api.common.pagination.request (pecan ObjectProxy that resolves\n  to None outside a request context; create_autospec fails on it)\n- driver_listener.memoryview (Python builtin; create_autospec fails on it)\n\nChange-Id: I057984f1ff06baebc7ab9b140911e9e97fbe83b3\nSigned-off-by: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e9d2836216c3346bfe25728f3d8e30abf57f248"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e9d2836216c3346bfe25728f3d8e30abf57f248"}]},"branch":"refs/heads/master"},"c504a8f0ec2cb73fcfec5ec5d4e045714406da6c":{"kind":"REWORK","_number":2,"created":"2026-06-22 08:48:12.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/93/994193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/994193/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/994193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/994193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/994193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/994193/2"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-13 15:26:42.000000000","tz":0},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-22 08:48:03.000000000","tz":0},"subject":"tests: Enforce autospec\u003dTrue on mock patches","message":"tests: Enforce autospec\u003dTrue on mock patches\n\nUsing mock_fixture.patch_mock_module() will enforce autospec\u003dTrue on all\nthe mock.patch and similar calls. This will enforce signature checks on\nthe mocked methods and functions, catching potential bugs missed by\notherwise false positive unit tests.\n\nSets autospec\u003dFalse for:\n- octavia.api.common.pagination.request (pecan ObjectProxy that resolves\n  to None outside a request context; create_autospec fails on it)\n- driver_listener.memoryview (Python builtin; create_autospec fails on it)\n\nChange-Id: I057984f1ff06baebc7ab9b140911e9e97fbe83b3\nSigned-off-by: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c504a8f0ec2cb73fcfec5ec5d4e045714406da6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c504a8f0ec2cb73fcfec5ec5d4e045714406da6c"}]},"branch":"refs/heads/master"},"6ada43e07398195871d56e8d04501e38431ece55":{"kind":"REWORK","_number":3,"created":"2026-06-22 09:52:50.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/93/994193/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/994193/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/994193/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/994193/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/994193/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/994193/3"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-13 15:26:42.000000000","tz":0},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-22 09:52:42.000000000","tz":0},"subject":"tests: Enforce autospec\u003dTrue on mock patches","message":"tests: Enforce autospec\u003dTrue on mock patches\n\nUsing mock_fixture.patch_mock_module() will enforce autospec\u003dTrue on all\nthe mock.patch and similar calls. This will enforce signature checks on\nthe mocked methods and functions, catching potential bugs missed by\notherwise false positive unit tests.\n\nSets autospec\u003dFalse for:\n- octavia.api.common.pagination.request (pecan ObjectProxy that resolves\n  to None outside a request context; create_autospec fails on it)\n- driver_listener.memoryview (Python builtin; create_autospec fails on it)\n\nChange-Id: I057984f1ff06baebc7ab9b140911e9e97fbe83b3\nSigned-off-by: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ada43e07398195871d56e8d04501e38431ece55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ada43e07398195871d56e8d04501e38431ece55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
