)]}'
{"id":"openstack%2Foctavia~995465","triplet_id":"openstack%2Foctavia~stable%2F2026.1~Iabd410b06d1f73fda5601f99328cf0501d1dbb38","project":"openstack/octavia","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-07-30 16:25:02.000000000","reason":"Change was submitted"},"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-07-30 16:25:02.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-07-01 07:24:13.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-07-29 11:50:21.000000000","reason":"removed on reply"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-07-30 16:25:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iabd410b06d1f73fda5601f99328cf0501d1dbb38","subject":"Fix octavia-worker multiprocessing for Python 3.14","status":"MERGED","created":"2026-06-30 11:29:05.000000000","updated":"2026-07-30 16:26:11.000000000","submitted":"2026-07-30 16:25:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995465","cherry_pick_of_change":994057,"cherry_pick_of_patch_set":3,"meta_rev_id":"01b65bab2f9e61c5df70431cfc4cb1fa649cfc0a","_number":995465,"virtual_id_number":995465,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-30 16:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-07-01 07:24:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2026-07-29 11:50:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-30 11:30:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2026-07-30 14:46:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}],"reviewers":{"CC":[{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-30 11:29:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":39044,"name":"Alicja Filipek","display_name":"alaf01","email":"alicja.filipek@cleura.com","username":"alaf01"},"state":"CC"},{"updated":"2026-06-30 11:29:42.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-06-30 11:29:42.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-06-30 11:29:42.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-06-30 14:22:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-08 18:10:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"39544537f98aa04ddb87c77dcbb9e59e4f4e9f60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-30 11:29:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"da94f68f91a36ec4c95acb1b1e5a07921e793b33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-30 11:29:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fc73ded79488069f9aca8882dc6e1aa6d1a20c7a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-30 11:30:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8796f6a0a2104200985ca4a7b54d8e63163974b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 14:22:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29fa5cbc525046ef99eca1bc59cc6973\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2734b544a63040ffba0f57fbd024893a : SUCCESS in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70eb4d6380fc49029b860edd1b7aae57 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4a32c6b3a044a39b73e0cdf7487e455 : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60d493510b1e438393a15f1ceab97080 : SUCCESS in 4m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aa55da028f224fde939c2f2c9e38c16e : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4756ae19eed45b0a57d6b1675ced71c : SUCCESS in 13m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fdd2afc9d1646e4b91e161341ba2f3e : SUCCESS in 4m 16s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/c319903c91f740558753e94ecd75c1e9 : SUCCESS in 5m 17s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/a256189e059a48a7bac6604bb7eeb77d : SUCCESS in 5m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/95bb0f35d684400d847a8b52b9d9b958 : SUCCESS in 3m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/11bdd9f96bb54f408a554938abcb77d9 : SUCCESS in 7m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1f857ead739341b09fe02ac8cebb5bc3 : SUCCESS in 8m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8f270425d8db4b5cb259c40c3cb4a845 : SUCCESS in 1h 13m 01s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3713badcba984a8fa4c99dddd83f5d8c : SUCCESS in 1h 36m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/081fcf0e99cc45f584ecc5f316dae748 : SUCCESS in 1h 16m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/38420ac279c54a1892d9c1567b8aa74f : SUCCESS in 1h 37m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/76e2bba913044ba2b2247d4d6ae6d9fd : SUCCESS in 1h 29m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b45e14838beb4a91ac190d999cedff52 : SUCCESS in 38m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c1d9a9b2f133404987b4b2769e07f86a : SUCCESS in 41m 41s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/6544055f95d64e919c7b79764974e0dd : SUCCESS in 43m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/60834f682bdd49e48c98d836f3edad64 : FAILURE in 48m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8bec6fe25dc44bc7886e95b96d266603 : FAILURE in 50m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b8ab07dd4b874a2ba33fc4b759d1403b : SUCCESS in 2h 51m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4efa1f06bc64485abea2c476fad633a2 : SUCCESS in 1h 35m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/7e97cd98aeaa480a80653a3317f0f6e6 : FAILURE in 7m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8fcc7585156e4ec3a9b32448dfea3f9e : FAILURE in 8m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad3c185ed8b7658b9a61f13a32962866113457a5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-01 07:24:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"26192dac47a679015ba102ca7bcd465f2c9b3e22","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-29 11:50:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"deb522c2357ecb6ffc5432a6212e964c271e0fb4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-30 14:46:16.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"288c20f165d4526c2a3831eb9ab3915a086596d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 14:47:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2909787ed8131d075c7fcb6934243ecf8266f21c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 16:25:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c0e06640fa84658b0153ba6ad5f136e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eceb0da820fe408e832fbbc3b0fb6793 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65715df050764e2c9d0b4b823051f5e5 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d72192b34aa84a10ab3b5e5d149877df : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4670b95c6b240be9d8074343a4a8066 : SUCCESS in 13m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d78f37307e443e69986a4ac97a90442 : SUCCESS in 9m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/19e8e7693c0c4a14b0191bafe77ff4a5 : SUCCESS in 3m 36s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/08f7ff583fb348db9f419dac3231a8d7 : SUCCESS in 5m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f94f6584474a4e2b8eb3b677b70bbb97 : SUCCESS in 37m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7b04910677724deeb7ccce611ee6f34a : SUCCESS in 1h 37m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e4cc4a67eae944e28976907d0bfbba8d : SUCCESS in 1h 02m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eb162095b14145058d9348068a702272 : SUCCESS in 39m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3ebe62d1119b4fd3bd5d8f8e9f68a713 : SUCCESS in 44m 59s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/0b27770029744e239140ede12d290394 : SUCCESS in 43m 00s","accounts_in_message":[],"_revision_number":2},{"id":"601df3a339cac8b101e827b7c3c18076101451ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 16:25:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"01b65bab2f9e61c5df70431cfc4cb1fa649cfc0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 16:26:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13bb9d7642ac49f4936b3c3966c39594\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7746f854d8cb43169f497b390027d1e0 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fba029667a734c8c8b1d142deb23b078 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"50316329bea5ca55b3db9841bf2f804b97247872","revisions":{"b78e43d1595a4809985708a497bfcd29190f7399":{"kind":"REWORK","_number":1,"created":"2026-06-30 11:29:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/995465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/995465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/995465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/995465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/995465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/995465/1"}}},"commit":{"parents":[{"commit":"2d14acc0d767de25cc840f375a7d44da192061be","subject":"Add SKI and AKI extensions to amphora certificates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2d14acc0d767de25cc840f375a7d44da192061be"}]}],"author":{"name":"Alicja Filipek","email":"alicja.filipek@cleura.com","date":"2026-06-19 08:20:13.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-30 11:29:05.000000000","tz":0},"subject":"Fix octavia-worker multiprocessing for Python 3.14","message":"Fix octavia-worker multiprocessing for Python 3.14\n\nPython 3.14 changed the default multiprocessing start method from\n\u0027fork\u0027 to \u0027forkserver\u0027 on Linux. The forkserver method requires\npickling of all arguments passed to child processes, which fails\nfor oslo_config.ConfigOpts due to an unpicklable lambda in\n_ConfigFileOpt.__init__.\nFix by explicitly requesting \u0027fork\u0027 context when creating\ncotyledon.ServiceManager instances.\nCloses-Bug: #2156391\n\nSigned-off-by: Alicja Filipek \u003calicja.filipek@cleura.com\u003e\nChange-Id: Iabd410b06d1f73fda5601f99328cf0501d1dbb38\n(cherry picked from commit 05d3b16dac1b93afa5d689b790c740d034d61c7b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b78e43d1595a4809985708a497bfcd29190f7399"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b78e43d1595a4809985708a497bfcd29190f7399"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"2a9a9530e210da04bc327653aa4cb6e8e960ef63","ours":"2d14acc0d767de25cc840f375a7d44da192061be","theirs":"05d3b16dac1b93afa5d689b790c740d034d61c7b","merge_strategy":"recursive","contains_conflicts":false}},"50316329bea5ca55b3db9841bf2f804b97247872":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-30 11:29:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/995465/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/995465/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/995465/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/995465/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/995465/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/995465/2"}}},"commit":{"parents":[{"commit":"c02f11aeb057d6a347faf3e826552f4742d93688","subject":"Bump hacking and pyupgrade for Python 3.14","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c02f11aeb057d6a347faf3e826552f4742d93688"}]}],"author":{"name":"Alicja Filipek","email":"alicja.filipek@cleura.com","date":"2026-06-19 08:20:13.000000000","tz":120},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-30 11:29:48.000000000","tz":0},"subject":"Fix octavia-worker multiprocessing for Python 3.14","message":"Fix octavia-worker multiprocessing for Python 3.14\n\nPython 3.14 changed the default multiprocessing start method from\n\u0027fork\u0027 to \u0027forkserver\u0027 on Linux. The forkserver method requires\npickling of all arguments passed to child processes, which fails\nfor oslo_config.ConfigOpts due to an unpicklable lambda in\n_ConfigFileOpt.__init__.\nFix by explicitly requesting \u0027fork\u0027 context when creating\ncotyledon.ServiceManager instances.\nCloses-Bug: #2156391\n\nSigned-off-by: Alicja Filipek \u003calicja.filipek@cleura.com\u003e\nChange-Id: Iabd410b06d1f73fda5601f99328cf0501d1dbb38\n(cherry picked from commit 05d3b16dac1b93afa5d689b790c740d034d61c7b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50316329bea5ca55b3db9841bf2f804b97247872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50316329bea5ca55b3db9841bf2f804b97247872"}]},"branch":"refs/heads/stable/2026.1","description":"Rebase","conflicts":{"base":"2d14acc0d767de25cc840f375a7d44da192061be","ours":"b78e43d1595a4809985708a497bfcd29190f7399","theirs":"c02f11aeb057d6a347faf3e826552f4742d93688","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
