)]}'
{"id":"openstack%2Foctavia~996546","triplet_id":"openstack%2Foctavia~master~I076c1656969877aa35c6866e7ff86343963e7d1d","project":"openstack/octavia","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-07-09 17:09:51.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-07-09 06:15:58.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":["remove-enforce-scope-flag"],"change_id":"I076c1656969877aa35c6866e7ff86343963e7d1d","subject":"Remove setting of oslo_policy[enforce_scope] flag","status":"MERGED","created":"2026-07-08 21:04:59.000000000","updated":"2026-07-09 17:11:44.000000000","submitted":"2026-07-09 17:09:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996546-secure-rbac","meta_rev_id":"caccbeb2985cbce9de01acff46b1026feb77592c","_number":996546,"virtual_id_number":996546,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-09 17:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"all":[{"value":1,"date":"2026-07-09 07:20:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":2,"date":"2026-07-09 06:15:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-09 01:00:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":1,"date":"2026-07-09 15:34:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-08 21:29:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-09 01:00:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-07-09 01:00:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-07-09 07:19:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-07-09 07:19:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"CC"},{"updated":"2026-07-09 07:20:41.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"}],"messages":[{"id":"01e5235bfb594648f5b3b1e000b788d2b4cf69e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-08 21:04:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a71fa456f6dc146778cb2e100e707f443f6a4152","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-08 21:17:11.000000000","message":"Hashtag added: remove-enforce-scope-flag","accounts_in_message":[],"_revision_number":1},{"id":"6200f3d10e945f9343fdd3c776c9ce312b378254","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 21:29:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ff4f34d2a73483da2e0596d7f1c6b08\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bbb1ccf4185475fa6901177be62a71f : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45c4b361b74541a0949b6d2dd7a4ad20 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc759786bc6c4395bf961f2a6cc3cf71 : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d47e8c7ff9a441c80c8b990e46736fe : SUCCESS in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c358f751f7284bdd80893e6ed789c2b4 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c255441879b14f0abaa65725762768f5 : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9ccc8bccda0459ea4bbd4356acd51f5 : SUCCESS in 9m 28s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/e7ba555ab1284908806efd0a53d0bf16 : SUCCESS in 5m 08s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/eb5e1447d7254830a78471924df8cb5b : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8cb0ebf5877b224bb99a7d4de8f772aeb8b2545f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-09 01:00:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ed91021c9c76c074c64a234cf01bae8545b7b430","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-09 06:15:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"837f5eb2e2a4ee1801f1095ab25a22e51e6a14f1","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2026-07-09 07:20:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8558498cb1bf18dc0344d3609aea2a3bfcaefc1d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-09 15:20:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5890f96fe8b6c5f2a941cf6d8816e50224eafe01","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-09 15:34:06.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2b7d1cd3bbbef9281da8ac27d13041f8cd2ab52f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-09 15:34:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ecb18ae4776c0ce7ddaa96264549e55f55c137d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 15:35:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8cecc6a141a3da0d3a20d242da55bedc0be81cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 17:09:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80b74bfc11984e1d88019e7b3c6186cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb793e28c546465cbedc867911cf2ba6 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63efa0242b06450b91dfed8fde510d2c : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f1d510e0abdb4a738cc5ecb1d4bf03ac : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0cab98f34394772819023c3b027c35a : SUCCESS in 13m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90e2b9bb1d53439ab27acf2728b9d02d : SUCCESS in 9m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/47ae5e4564d64d0aa8e65d3784219ec7 : SUCCESS in 1h 22m 57s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2492450f2ce841c48f57a850085e95be : SUCCESS in 1h 33m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a1a8455e7a704d37910ce2330e306764 : SUCCESS in 1h 04m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b234f2b8c83a446d8a0034913077b225 : SUCCESS in 35m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3b157c3827054fc5bf7ae5d39b84241b : SUCCESS in 48m 06s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c54e3e4cea5547f2849382fe59f9b038 : SUCCESS in 43m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d0eb3c16936acc42256053ec759f4e91f28339c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 17:09:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"caccbeb2985cbce9de01acff46b1026feb77592c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 17:11:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c44886cdf525437da2616344bca4adc3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f1a8b4925b64a3a886bdd8b0909ef52 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78d38afc126c37e2ec65859e8cfc770eb5457075","revisions":{"78d38afc126c37e2ec65859e8cfc770eb5457075":{"kind":"REWORK","_number":1,"created":"2026-07-08 21:04:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/996546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/46/996546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/996546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/996546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/996546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/46/996546/1"}}},"commit":{"parents":[{"commit":"9686d904a02adf167213b9a8969f36fd68559892","subject":"Merge \"Fix allocation of interfaces to netns on reboot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9686d904a02adf167213b9a8969f36fd68559892"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-07-08 21:04:35.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-07-08 21:04:57.000000000","tz":0},"subject":"Remove setting of oslo_policy[enforce_scope] flag","message":"Remove setting of oslo_policy[enforce_scope] flag\n\noslo_policy[enforce_scope] config option has been removed\nfrom oslo.policy 6.0.0\n- https://review.opendev.org/c/openstack/oslo.policy/+/986475\n- https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/t\u003e\n\nRequirement change (Needed-By) is updating the upper-constraints\nfor oslo.policy which need projects to stop using the enforce_scope\nflag\n\nNeeded-By: https://review.opendev.org/c/openstack/requirements/+/996511\n\nChange-Id: I076c1656969877aa35c6866e7ff86343963e7d1d\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
