)]}'
{"id":"openstack%2Foctavia~996896","triplet_id":"openstack%2Foctavia~master~Id449668fb2af20ed8d625e6f74292536601dabae","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Id449668fb2af20ed8d625e6f74292536601dabae","subject":"tox: Drop redundant override of install_command","status":"NEW","created":"2026-07-11 17:15:00.000000000","updated":"2026-07-11 19:53:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"86d9a7f37bf350792bc4b7977548061762a59523","_number":996896,"virtual_id_number":996896,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-11 19:53:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-11 19:53:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df2d73c38d671c41f6fb8284eb5a588776531aee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-11 17:15:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86d9a7f37bf350792bc4b7977548061762a59523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 19:53:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14af55377c2641eca4cba6a10a552f83\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/de210bff57b54f9cbe7bd03fe96b667a : SUCCESS in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70f158733ca74d2f98d94d3e54cecd47 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58ffdafa69e340dd8d6a164b929816ac : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/654715a197c3493b95c79b971c38de44 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3736a6a8f7b741ccac1dadbc104f031b : SUCCESS in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6c2380d046d84f449c8af59efa0b7191 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f808b485bddf4d80bd7d06ae72e3fac2 : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bcf41331afa41af9d41acf2f6f900eb : SUCCESS in 9m 59s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/aaa68f2acbbe40c29ed943ea0ba3d9d5 : SUCCESS in 6m 29s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/6282687f844f4d128bee1cc4d0d8c301 : SUCCESS in 5m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3ce5e40a1eda49d0b8493276a217de20 : SUCCESS in 3m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4552918494a442bd81547a648f2e64ee : SUCCESS in 6m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/50ee201fbdcc43199c84a51d898297ed : SUCCESS in 1h 03m 23s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/dfbba50630304a2d9032c357a3309bb3 : SUCCESS in 1h 22m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/13186db7ef804722924440441004695d : SUCCESS in 1h 05m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/030a86a6091b421ca203ccc97eba705b : SUCCESS in 1h 26m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c50895476dc84f78b7c41e1b0304dbe6 : SUCCESS in 1h 06m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5dadf354950b4cf5bafbebe13592625e : SUCCESS in 35m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bd7497797ce0494f93f6ee805b3d6aa5 : SUCCESS in 31m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e6951cf49f234ae0a32694185def6f81 : SUCCESS in 32m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d7dcf56b630846899ec3540257ecaf8f : SUCCESS in 2h 12m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/93431ee0e78b4ea5b4eae23ba26db9ad : SUCCESS in 1h 21m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/e8a3af4cce184bbc91d03829fcff4af8 : TIMED_OUT in 2h 36m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/72346e87246d442e8a47192903ea6ab0 : SUCCESS in 2h 28m 08s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99d1b9532bb7643c0baf61d30d1905e4c6cebd53","revisions":{"99d1b9532bb7643c0baf61d30d1905e4c6cebd53":{"kind":"REWORK","_number":1,"created":"2026-07-11 17:15:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/996896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/996896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/996896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/996896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/996896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/996896/1"}}},"commit":{"parents":[{"commit":"920a62fe806f3a172c1bed18e5e72ef55a90ae6f","subject":"Merge \"Feature: allow denylist of network_ids\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/920a62fe806f3a172c1bed18e5e72ef55a90ae6f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:14:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 17:14:49.000000000","tz":540},"subject":"tox: Drop redundant override of install_command","message":"tox: Drop redundant override of install_command\n\nIt is effectively equivalent to the default value:\n\n  python -I -m pip install {opts} {packages}\n\nChange-Id: Id449668fb2af20ed8d625e6f74292536601dabae\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99d1b9532bb7643c0baf61d30d1905e4c6cebd53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99d1b9532bb7643c0baf61d30d1905e4c6cebd53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
