)]}'
{"id":"openstack%2Foctavia~997100","triplet_id":"openstack%2Foctavia~master~Ic8a363598016c0a286c56e33ec09cd69fe670ccb","project":"openstack/octavia","branch":"master","attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-07-15 08:36:58.000000000","reason":"\u003cGERRIT_ACCOUNT_6577\u003e replied on the change","reason_account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-07-15 08:36:58.000000000","reason":"\u003cGERRIT_ACCOUNT_6577\u003e replied on the change","reason_account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"}}},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2026-07-15 08:36:58.000000000","reason":"\u003cGERRIT_ACCOUNT_6577\u003e replied on the change","reason_account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"}}},"hashtags":[],"change_id":"Ic8a363598016c0a286c56e33ec09cd69fe670ccb","subject":"Allow to use value with leading dash and hash in l7rule","status":"NEW","created":"2026-07-13 17:15:55.000000000","updated":"2026-09-20 13:51:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"036e9ea46472e899d2e7ddf8944cef57021c74ed","_number":997100,"virtual_id_number":997100,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-20 13:51:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-13 19:55:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-15 08:36:58.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-07-15 08:36:58.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"3755a149fb3cf0f0f2becae77bdd1d2d7c1ad3de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2026-07-13 17:15:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e83110f0c3ff38921fed4f01d06b6cb17fcb1e36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 19:55:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63db0c432cb94703a047d0f505ec79f0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8637e5ac857b4c87822df0affdd93de4 : FAILURE in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f686de17d4974efbb9431b765d9880db : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5857fc63bd6476ea96103443653069c : FAILURE in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6467ae2cdadc4cb3a2b71c1b79df2fab : FAILURE in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8862333a7d4a4c3781d154be73e5b853 : FAILURE in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe69f85c22fb46e5827ac80ef798b897 : SUCCESS in 13m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f677e884ff144b7b227a18085d4eaa3 : SUCCESS in 8m 00s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/149c611f216340c9b33d9cd8fdfd7a64 : FAILURE in 3m 15s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/77b4c5d13f894c71bce94596f39abb7c : SUCCESS in 5m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/99d1783f030f498f8200c6325b3b031e : SUCCESS in 3m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c7e5e98de6fa46fb84794b1aea189e86 : SUCCESS in 4m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4a036b7502f24f9b93b2a6b0b6697ad3 : SUCCESS in 45m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/adae8d33f97949e19215a61c2c6ad071 : SUCCESS in 1h 25m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8a635502450b4360bf4d87fd34829245 : SUCCESS in 1h 05m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6983cbd9ee1d449aa230877d598f9af8 : SUCCESS in 1h 33m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b6043daef35a4d0b8e5b8e5bd116c227 : SUCCESS in 1h 04m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e11b8d934eb747a29c11e2a1606790e0 : SUCCESS in 34m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/412a2ab648b84776bfa960217d5d6572 : FAILURE in 1h 03m 47s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/fce6c571534e484b98658ef30841f734 : SUCCESS in 45m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71ac9702264746d782233b045003e2f7 : SUCCESS in 36m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/416c16f4629c4ec8ab572bc9657bf50d : SUCCESS in 32m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6327adcc2d294f6294a3d96a84fcd5c5 : SUCCESS in 2h 06m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5cd664b0cc9e446194e62f7d13f0ea6d : SUCCESS in 1h 27m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/365eae5348e3458bb315b34ea7826c8f : TIMED_OUT in 2h 37m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/391c60135cc34d559227b1ef254b84f8 : SUCCESS in 2h 28m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aae3fe530e999d25599a0444b4f2e7fe6e363a46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2026-07-14 08:02:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3854321bd6c89c08d127ef9accba3c1fd8743798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 10:41:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12ad5227ed4347709d62eac1fa0e1de0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fbfe37204c7475e87e4c562241e4ed6 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e0cc3e62c3242188ae85c22c034f8c1 : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/639d2256532840b6a1d9b6d6c78bf468 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2039da6b13f349648fbae5890ebee669 : SUCCESS in 3m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/744206c614a04bedabdb27675739513b : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b81e8c28555c49db8990f1a324615f0e : SUCCESS in 14m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86d86cb90b52472d9d5110bc6089f4df : SUCCESS in 8m 47s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/05161589557e4b1882d33c84fa24db5c : SUCCESS in 4m 56s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/a9c82898ce284eabac320d0dfe5abc16 : SUCCESS in 4m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b4289d15c40d4ed38621e8602f7de1b0 : SUCCESS in 2m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ddd374222d7046a0a3f2805ea0590992 : SUCCESS in 5m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d504d55e8662401ab9793ff9f2ad8acf : SUCCESS in 34m 16s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4b6227334cd64adcb01244be2b0e8a3d : SUCCESS in 1h 23m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/aa89551a5057480eb9de95b2a141fd84 : SUCCESS in 1h 01m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/761f98b432ff44ac88ee0a9990967e9d : SUCCESS in 1h 26m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3c95285d55d345158552fbdeea1b6b61 : SUCCESS in 1h 18m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4bd9962e8d904888964369e128d139aa : SUCCESS in 35m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fa31e55eb3374f98a5124c840d7eda51 : SUCCESS in 45m 09s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/da13fc3ab9cd463fb2395217bf43b472 : SUCCESS in 43m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3fde39f3c78d4e5dad03dc41268a0e42 : SUCCESS in 30m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1b0de3a14e3749f686761741c928dd35 : SUCCESS in 35m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6036e444c278492f89986323b24e2b3f : SUCCESS in 2h 28m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/de953e14dddd4c3392267da6276c20f8 : SUCCESS in 1h 21m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/f5f4e5a9ca654bdb91160791e9a71c31 : TIMED_OUT in 2h 36m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/271b1f085dad44f0b00bdf17dcea9b3f : SUCCESS in 2h 35m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eb2c359bc37ba175250b4dc1100a6e56d3e9e52f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2026-08-24 10:28:40.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d463f8e1d0c309b77743aa7dadd15c34c617f071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 13:14:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1580ed34c733479dab3a47a2f86a2834\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a583a171f139440e9657b44bce71add9 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/772dd04124ab4d9ba576503ead092e8a : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5febaad2f184e36be75c75bad393add : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68b920741949456bbfe2113868632dab : SUCCESS in 4m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9a75f3dfff7540dc944b3b6a5015fd59 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59e10b46d0784e1586ed6fd68c913b2a : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f834c8ac164a41fda5392f9ba4f58e6f : SUCCESS in 4m 20s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/82776e97e2e0438685c6a10c7587eff2 : SUCCESS in 4m 26s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/e558def5e51d4f638830948c65072138 : SUCCESS in 5m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/685b99d1cbe2412b9a398b8127359ee8 : SUCCESS in 3m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b1e071ba24134f41b2373447ce3eaa29 : SUCCESS in 4m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/82856765abe54f24b2bd691fe4559352 : SUCCESS in 33m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b2ca5f4eb9b74c0aaf3bc3e692032819 : SUCCESS in 1h 22m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d520b5f778a3458883b2e0eebf36804c : SUCCESS in 55m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a01d02f9a75e4a2ca698c44579a5577d : SUCCESS in 1h 27m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e4fde38fab924d82ad3b05cc5bfbedab : SUCCESS in 1h 06m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cda63d4846e247aeaf7068916b842837 : SUCCESS in 34m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d7718d4374e1493b9b9368528245677a : SUCCESS in 41m 53s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/98af48af9b684ba68b0a158fadcf37b5 : SUCCESS in 41m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c26a94100441420ea6c2b2a122019500 : FAILURE in 22m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/718fdd408ed9419bbf32d8a00acbd9a3 : SUCCESS in 34m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/309172d88d8845c486150ddb78aa71a0 : SUCCESS in 2h 43m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7535c2818b2c42b79be764399d5437fa : SUCCESS in 1h 29m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/f54251cc729b43949b7858fb3324d772 : SUCCESS in 1h 11m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e4a7b26c1fef4d38ad3b1e9bfdfc3a2e : SUCCESS in 55m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"edb9948c7186906927c2d7b9bf8b548aeea07c03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2026-09-20 11:37:45.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"036e9ea46472e899d2e7ddf8944cef57021c74ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 13:51:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aefd3627d1674f839259fa8884f9b2c0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/327c59b34f4542faabe8543af33aa7c4 : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f429d51068864d07bc711253b5099031 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b426431562e7454fb0a75feb0090aa1d : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c5ebeb2bb3f344a7837c2cf47f787025 : SUCCESS in 2m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/26c1e5186d8a4d7ea89b93bc9f8e9fb7 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/331471dfd8084d7a8de4f6a6e8f7ee1b : SUCCESS in 14m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a148c41482254744a49182276ae894ba : SUCCESS in 7m 45s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/05e194f0a4c448269954431a06835143 : SUCCESS in 4m 19s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/117f1b96d86c40c19cd1590c34f9d5bd : SUCCESS in 8m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a1b5e6da7fb94d1885cab38451c8bb3f : SUCCESS in 2m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/301dc611b75e4df4b74c5cf71acb0566 : SUCCESS in 5m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/de18397cde8a4818be5f95da585cbe9a : SUCCESS in 1h 05m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/990d0b36d9e4445db6a2e6f17d27ef9b : SUCCESS in 1h 25m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f614950b847f40609eb6a3e72cb351cc : SUCCESS in 1h 05m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b1bccccec71d4b458df1b1d5c8610af9 : SUCCESS in 1h 28m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4edcf7e4b4294093a502f5da7e272c8f : SUCCESS in 1h 10m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ee8b3bc37e5e4fe1870aed6daca9a880 : SUCCESS in 37m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7c764570173e4387ad0b2d9fb53db95b : SUCCESS in 43m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc42ca92caa34988bc6589adf8971c66 : SUCCESS in 33m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8333db49dc0a47179c7deb87817f35e4 : SUCCESS in 36m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/932297cb7f414825ae509bd4d110da60 : SUCCESS in 2h 12m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/05a002e72402449e8031cc3be55f639b : SUCCESS in 1h 22m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/ace9e1a084b543c7bf6eb099f2236c96 : FAILURE in 6m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c552dcde7dc44cac928f16b93835200d : SUCCESS in 1h 00m 08s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"51a25d8d5acdc80f7de83b8d0bb172fedc24ed3b","revisions":{"53ae32741643a7485bbc2b8c512bf34034d7c8c5":{"kind":"REWORK","_number":1,"created":"2026-07-13 17:15:55.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/00/997100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/997100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/997100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/997100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/997100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/997100/1"}}},"commit":{"parents":[{"commit":"78d38afc126c37e2ec65859e8cfc770eb5457075","subject":"Remove setting of oslo_policy[enforce_scope] flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2026-07-13 17:10:04.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2026-07-13 17:15:43.000000000","tz":240},"subject":"Allow to use value with leading dash and hash in l7rule","message":"Allow to use value with leading dash and hash in l7rule\n\nCloses-Bug: 2158969\n\nSigned-off-by: Sergey Kraynev \u003csergejyit@gmail.com\u003e\nChange-Id: Ic8a363598016c0a286c56e33ec09cd69fe670ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/53ae32741643a7485bbc2b8c512bf34034d7c8c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/53ae32741643a7485bbc2b8c512bf34034d7c8c5"}]},"branch":"refs/heads/master"},"b3db1bf52a4f195cebccc4a91916e1c10dd55612":{"kind":"REWORK","_number":2,"created":"2026-07-14 08:02:07.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/00/997100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/997100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/997100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/997100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/997100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/997100/2"}}},"commit":{"parents":[{"commit":"78d38afc126c37e2ec65859e8cfc770eb5457075","subject":"Remove setting of oslo_policy[enforce_scope] flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2026-07-13 17:10:04.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2026-07-14 08:01:56.000000000","tz":240},"subject":"Allow to use value with leading dash and hash in l7rule","message":"Allow to use value with leading dash and hash in l7rule\n\nCloses-Bug: 2158969\n\nSigned-off-by: Sergey Kraynev \u003csergejyit@gmail.com\u003e\nChange-Id: Ic8a363598016c0a286c56e33ec09cd69fe670ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3db1bf52a4f195cebccc4a91916e1c10dd55612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3db1bf52a4f195cebccc4a91916e1c10dd55612"}]},"branch":"refs/heads/master"},"738acfd96db1c8de3b149d93043baa6a8a4ad38b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-08-24 10:28:40.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/00/997100/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/997100/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/997100/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/997100/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/997100/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/997100/3"}}},"commit":{"parents":[{"commit":"9a0e0a765ce0602cb38110655cd0cb73d8133e9c","subject":"Merge \"Fix VRRP and HAProxy peer ports open to 0.0.0.0/0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a0e0a765ce0602cb38110655cd0cb73d8133e9c"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2026-07-13 17:10:04.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2026-08-24 10:28:40.000000000","tz":0},"subject":"Allow to use value with leading dash and hash in l7rule","message":"Allow to use value with leading dash and hash in l7rule\n\nCloses-Bug: 2158969\n\nSigned-off-by: Sergey Kraynev \u003csergejyit@gmail.com\u003e\nChange-Id: Ic8a363598016c0a286c56e33ec09cd69fe670ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/738acfd96db1c8de3b149d93043baa6a8a4ad38b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/738acfd96db1c8de3b149d93043baa6a8a4ad38b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"78d38afc126c37e2ec65859e8cfc770eb5457075","ours":"b3db1bf52a4f195cebccc4a91916e1c10dd55612","theirs":"9a0e0a765ce0602cb38110655cd0cb73d8133e9c","merge_strategy":"recursive","contains_conflicts":false}},"51a25d8d5acdc80f7de83b8d0bb172fedc24ed3b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-20 11:37:45.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/00/997100/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/997100/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/997100/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/997100/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/997100/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/997100/4"}}},"commit":{"parents":[{"commit":"57460f16a2864b889b4c89e79d978a45cdd84bf6","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57460f16a2864b889b4c89e79d978a45cdd84bf6"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2026-07-13 17:10:04.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2026-09-20 11:37:45.000000000","tz":0},"subject":"Allow to use value with leading dash and hash in l7rule","message":"Allow to use value with leading dash and hash in l7rule\n\nCloses-Bug: 2158969\n\nSigned-off-by: Sergey Kraynev \u003csergejyit@gmail.com\u003e\nChange-Id: Ic8a363598016c0a286c56e33ec09cd69fe670ccb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51a25d8d5acdc80f7de83b8d0bb172fedc24ed3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51a25d8d5acdc80f7de83b8d0bb172fedc24ed3b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"9a0e0a765ce0602cb38110655cd0cb73d8133e9c","ours":"738acfd96db1c8de3b149d93043baa6a8a4ad38b","theirs":"57460f16a2864b889b4c89e79d978a45cdd84bf6","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
