)]}'
{"id":"openstack%2Foctavia~997182","triplet_id":"openstack%2Foctavia~master~Ia7067c1774a435d484637f46f3a83e2aea8cea02","project":"openstack/octavia","branch":"master","topic":"bp/update-flavor-profile-compute-flavor","attention_set":{"37604":{"account":{"_account_id":37604,"name":"Richard Bachmann","display_name":"rbachman","email":"richard.bachmann@cern.ch","username":"rbachman","status":"Does cloud operations and network stuff at CERN."},"last_update":"2026-08-24 22:20:02.000000000","reason":"\u003cGERRIT_ACCOUNT_39270\u003e replied on the change","reason_account":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"}},"39115":{"account":{"_account_id":39115,"name":"Vinicius Marques Rodrigues","email":"vinicius.marques.rodrigues@cern.ch","username":"viniciusr"},"last_update":"2026-08-26 08:23:56.000000000","reason":"\u003cGERRIT_ACCOUNT_39270\u003e replied on the change","reason_account":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-08-12 14:35:13.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"39270":{"account":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"last_update":"2026-08-26 08:23:56.000000000","reason":"\u003cGERRIT_ACCOUNT_39270\u003e replied on the change","reason_account":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"}}},"hashtags":[],"change_id":"Ia7067c1774a435d484637f46f3a83e2aea8cea02","subject":"Allow updating compute_flavor in a FlavorProfile","status":"NEW","created":"2026-07-14 08:39:27.000000000","updated":"2026-08-26 08:23:56.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":11,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"5b2f8d9a1302fafa5a812b535a6968661db3056a","_number":997182,"virtual_id_number":997182,"owner":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":37604,"name":"Richard Bachmann","display_name":"rbachman","email":"richard.bachmann@cern.ch","username":"rbachman","status":"Does cloud operations and network stuff at CERN."},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-25 00:24:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37604,"name":"Richard Bachmann","display_name":"rbachman","email":"richard.bachmann@cern.ch","username":"rbachman","status":"Does cloud operations and network stuff at CERN."},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":37604,"name":"Richard Bachmann","display_name":"rbachman","email":"richard.bachmann@cern.ch","username":"rbachman","status":"Does cloud operations and network stuff at CERN."},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37604,"name":"Richard Bachmann","display_name":"rbachman","email":"richard.bachmann@cern.ch","username":"rbachman","status":"Does cloud operations and network stuff at CERN."},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":39115,"name":"Vinicius Marques Rodrigues","email":"vinicius.marques.rodrigues@cern.ch","username":"viniciusr"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37604,"name":"Richard Bachmann","display_name":"rbachman","email":"richard.bachmann@cern.ch","username":"rbachman","status":"Does cloud operations and network stuff at CERN."}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-14 08:53:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-05 13:02:13.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-08-24 13:37:26.000000000","updated_by":{"_account_id":37604,"name":"Richard Bachmann","display_name":"rbachman","email":"richard.bachmann@cern.ch","username":"rbachman","status":"Does cloud operations and network stuff at CERN."},"reviewer":{"_account_id":37604,"name":"Richard Bachmann","display_name":"rbachman","email":"richard.bachmann@cern.ch","username":"rbachman","status":"Does cloud operations and network stuff at CERN."},"state":"REVIEWER"},{"updated":"2026-08-25 14:16:49.000000000","updated_by":{"_account_id":39115,"name":"Vinicius Marques Rodrigues","email":"vinicius.marques.rodrigues@cern.ch","username":"viniciusr"},"reviewer":{"_account_id":39115,"name":"Vinicius Marques Rodrigues","email":"vinicius.marques.rodrigues@cern.ch","username":"viniciusr"},"state":"CC"}],"messages":[{"id":"799fd9ac4e5eed1e0cbbeb21f2ef18879e86cf3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"date":"2026-07-14 08:39:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e39c75644590e85ee11b715751b164043aee0b05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 08:53:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/25cdd970187f4177983a9752cbd19157\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7231cca994344f82ad3fc51cd05b29de : FAILURE in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c5db9af4292428fb18b249fe86a85a2 : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/020abc12c4d14a6bb0263f14086d46f0 : SUCCESS in 8m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5df9087914308cbde52697749b782115219b27b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"date":"2026-08-05 10:03:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e7358e29b92fe6862b16a33846fba165742b1830","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"date":"2026-08-05 10:55:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cc3486405599ff91946ad3aeb2735ddb88f264b4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-05 13:02:13.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8d6b789182e73adb0479370754697b64e4ee4150","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 14:00:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e70b64a34f0b422fb5ebd7ec79c8735b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3eca40e4c97d4bef8151604dcc70cea9 : FAILURE in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/870ad333bee741e7835d28d847dc396f : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7aea969dd3204a6ea15168247435ae44 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/16c78749d4734f8b877121f5f60b60f1 : SUCCESS in 4m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6ed764ee3a644249af36be3fed51aff6 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e937c888e2c46cc90c7630d6a996b68 : SUCCESS in 14m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8128e0b8a00451f99ac8338ef253db1 : SUCCESS in 5m 09s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/f9b6a0e063a24c639f2a3334a1a87d92 : SUCCESS in 2m 56s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/ed89e6fa47504066a8368b45a0a62ef4 : FAILURE in 7m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/46d06832090b4c94b0f8cd467fdf044d : SUCCESS in 4m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ff983086d68348148d96da33fb787c58 : FAILURE in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4c73499d5c894cfbacd229126f03381c : SUCCESS in 1h 25m 28s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5307fce3ddfa4ff7a5105566e0f47b53 : SUCCESS in 1h 25m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2c2b60a6a7d8434db95d4d4d0f658c41 : SUCCESS in 1h 00m 04s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e7a19045b30049f593347ea54bd49778 : SUCCESS in 1h 33m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3c59637500324708a98906a2cd0f3019 : FAILURE in 35m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e013f109558346739102fd7d0b0a27f3 : FAILURE in 36m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b2a61364c7fb414ba2967b077a049bd8 : SUCCESS in 1h 03m 35s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7affd3999b8143a0a247f8bfaf263f71 : SUCCESS in 58m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5484dd94ef02425993b53d79dd580569 : SUCCESS in 46m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/96d80e562e4144368643c58043d864e5 : FAILURE in 35m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cad46c7f200b49528eb0d1ebd1d17dc5 : TIMED_OUT in 3h 03m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/80f84c37bb9f4ede91f52cf6eee41f44 : SUCCESS in 1h 22m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/4f16b887362f48d0946bfb6a6951e3d4 : TIMED_OUT in 2h 36m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1ab78faeb2f94b8b863aa1e90f742523 : SUCCESS in 2h 29m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4fd26628ccab55d97bc7894ce886c8d095412a58","author":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"date":"2026-08-05 15:55:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3f9329200fd5828dec964b9324a5e4262d3b50a3","author":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"date":"2026-08-05 16:51:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c95440974c5b780caf21f905955faff9964b26f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"date":"2026-08-12 14:35:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"d1c9102e2b5b2a2faf956471b2e254b8a685c50b","author":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"date":"2026-08-12 14:44:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3a954cb56672569482585dbcf12f17cd2ede235f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 17:05:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f1d33be7cba4cc0a9e7ccb2891da994\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/711d8f5484f34ef0868e91cc72fbe005 : FAILURE in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e21d653935ee4dc88f5df9df7d9f0b6f : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d0f7d6fbba264d8597c65555cf4ab3df : FAILURE in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdb6f8f8f5ff46bdaa34ccf31a6ae986 : FAILURE in 3m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/46cd5df7b9d649b08da390e81bd8fe5c : FAILURE in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdf8e675df494dee93e64e697e512fce : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db07d62b220c45fb8002910d9c9eecb5 : SUCCESS in 7m 45s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/1fae1904e20a43c8918313b75f1848df : FAILURE in 3m 39s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/f4a406b429864d189e7463268b7ac3c5 : SUCCESS in 4m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/da7b7bc8f2914b22a2caa3e17080bdcd : SUCCESS in 2m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2e1b9cc5587548ab9edb4571d005432d : SUCCESS in 4m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6a9d4ba262464c298515a7729b853fbc : SUCCESS in 1h 04m 41s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/466b0e60abee417fb926a2c59e95cd94 : SUCCESS in 1h 30m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8490b969c0d34836a77f0033607f5682 : SUCCESS in 1h 01m 40s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9f0b4e755e6b40cf903fbf122e8458c8 : SUCCESS in 1h 28m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f586dd4d45c84c42bd6ec3f4926635f4 : SUCCESS in 1h 13m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/03e5c1a4b54a41a39a9057bc42745c8b : SUCCESS in 35m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6902491806b64e9daafbdda3a634403c : FAILURE in 20m 03s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e8cbd294aa79441782550ab189d74406 : FAILURE in 19m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb3ba7decb6a4c3ba08b46d2b4e2c4b8 : SUCCESS in 32m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/694ea8aae8be4276b38f5a3b6a556e94 : SUCCESS in 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4e78aaa3ba1648e9a03a2d947474120d : SUCCESS in 2h 28m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ea2ba3718de74fc3a4352c4c2568f55e : SUCCESS in 1h 29m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/31900c56bea944b8b570541d3b04cdb8 : SUCCESS in 1h 13m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/80da2bac69fd4ee29c68e0b74d7a9cd9 : SUCCESS in 1h 04m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"48529a3fcfbc1c30e13ac551964e848040c480e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"date":"2026-08-19 10:29:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"7362c22ae398944212600415112643c1336201b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"date":"2026-08-19 10:31:47.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"3cadd1eea54e7928275a34685629112c5aa39b62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"date":"2026-08-19 10:32:57.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"6ac3b08faaae993ad1b42f54c02c11a325a62106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 12:50:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ff5ef00c18944e882145da3ed92607f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db56c7452d04424aae7d985e21a18f75 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70e8727a92d64b4eb6e708b973e30696 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0593cf99cbdc4986b29cbdcb919f2d94 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18dfe7aaca6744e88cc2467def1cd6d4 : SUCCESS in 5m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/19393151f58840a38c4edf8c814b5c43 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b400758a8e4575b174d9851d47a46a : SUCCESS in 14m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c3720ca9e2a4f949d7003680b0b566f : SUCCESS in 9m 15s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/dcdf1ade4e4041bdad799d873b0cd364 : SUCCESS in 4m 37s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/755ca6d661b444bd9b5ed0b81cd0f50f : SUCCESS in 7m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6e5edfa09c994ca6bdbb82528ed897c7 : SUCCESS in 3m 15s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7aa6b67a261149558c6474f8afdd5917 : SUCCESS in 6m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/912c8d052c2c412d8e8a27b3f97dc488 : SUCCESS in 1h 07m 55s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2b474810942e464c9225654433ab0819 : SUCCESS in 1h 27m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/be3c98e5b60b486881561909d47f28cb : SUCCESS in 1h 02m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d5cfaabb37924448bbfa3a84cb58cd98 : SUCCESS in 1h 23m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3edc77abcec3413298a7b602d41ddc7d : SUCCESS in 57m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/892b6a4a8f0149bbbbdd36534a253474 : SUCCESS in 32m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2479e3fb24ea40db957900b2eb1e30ef : SUCCESS in 47m 14s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/70ae77adc10242ad8280d78f3cb75a5b : SUCCESS in 46m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a25987dc76249eaba8da64fabfa4667 : SUCCESS in 35m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8e6bbda0088d460ead72473006263d79 : SUCCESS in 33m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9ad4917b8b6848f9b0222b931f23cbdc : SUCCESS in 2h 16m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b809e326f53c40fdb7f825e566f64642 : SUCCESS in 1h 19m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/11eb6c8a5bf4418c85880a61cb39aa86 : SUCCESS in 1h 12m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e1e7f5900c754bbfae73a3c4b960068c : SUCCESS in 55m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6b2c56b3cf5b20e2ef9b29551c7d316d3532b467","author":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"date":"2026-08-21 14:28:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7c4ed0fb9b0cf29cf9b43a77c59a9216d4d5ffbf","author":{"_account_id":37604,"name":"Richard Bachmann","display_name":"rbachman","email":"richard.bachmann@cern.ch","username":"rbachman","status":"Does cloud operations and network stuff at CERN."},"date":"2026-08-24 15:34:29.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"72bcfdbece24408057c1898e68bff12303b18e79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"date":"2026-08-24 22:10:04.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"c2bf35b8ee9ed06fca3f17c6044692d8388337fd","author":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"date":"2026-08-24 22:20:02.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3144cd31c1df9eefd51f26f63b25d0ea9cf783cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 00:24:32.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5a54b0ffe4849f8a5885ae2b7959dd9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8dd240e919b4d018e8c052c9408557b : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/894036c8f4cf4d9e919621205cb30c93 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b18c4e6d77024e5289de597c3b609773 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac098789ab664e15b71cf465f8e5fcb0 : SUCCESS in 4m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8d85924ca1bf4d4d805e14a5281b6b8d : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab239d133af476d959c8d09a94f0b72 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c4918a75c6a4f039e9ceed2470e3de8 : SUCCESS in 8m 58s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/8958f4dba9014c0e99c9620a4a61a60b : SUCCESS in 4m 13s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/ee2edb36d363408ca31f1efba448a237 : SUCCESS in 5m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d3a0e3ccb03347ed8d3c6cd64e5abfe8 : SUCCESS in 4m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e29c6ab434df464a86429a6f5cfc30ef : SUCCESS in 4m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c7aef9159be047d88159daa2efd9922a : SUCCESS in 1h 04m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b4c605ab79db4d6fa79e151a9c79f572 : SUCCESS in 1h 18m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/73705467f2b8405cafd0b43628d6166e : SUCCESS in 1h 04m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/162d8d5b093a41dd9d70b407eb1ade60 : SUCCESS in 1h 27m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2d656f40907c40e6a593edbd0872a75b : SUCCESS in 1h 00m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7b424095253d4d7599a18d769b66c3e2 : SUCCESS in 35m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6d6ae775029e45a18125268e73e1c0d6 : SUCCESS in 42m 36s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7a545d6f8c354861ba71c02aace41bfa : SUCCESS in 44m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2acf9aaac4c042f4a790a656e0e01c50 : SUCCESS in 32m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/54177b1550f247109ffcbbd6280fd5fa : SUCCESS in 32m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/668a4b7ee183484ab1180bd28d8af070 : SUCCESS in 2h 12m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/54895eb0360547e28ebd7c3681763cc0 : SUCCESS in 1h 18m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/b7d80dab32ad4ac9afb299558416b91d : SUCCESS in 1h 10m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fe110817c0424c58a456ed21c8943b95 : SUCCESS in 54m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9197cec0e582e89c794347c75adba42fdf13a9b3","author":{"_account_id":39115,"name":"Vinicius Marques Rodrigues","email":"vinicius.marques.rodrigues@cern.ch","username":"viniciusr"},"date":"2026-08-25 14:16:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5b2f8d9a1302fafa5a812b535a6968661db3056a","author":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"date":"2026-08-26 08:23:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"848f8c49fa02b5473a4da047d65f5ce4101e7ff8","revisions":{"c7a68f0aa4e0d69e1c9a4dc2f7c87f53909f8051":{"kind":"REWORK","_number":1,"created":"2026-07-14 08:39:27.000000000","uploader":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"ref":"refs/changes/82/997182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/997182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/997182/1"}}},"commit":{"parents":[{"commit":"78d38afc126c37e2ec65859e8cfc770eb5457075","subject":"Remove setting of oslo_policy[enforce_scope] flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]}],"author":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-07-14 08:27:56.000000000","tz":540},"committer":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-07-14 08:39:13.000000000","tz":540},"subject":"WIP: Updating compute_flavor in a FlavorProfile","message":"WIP: Updating compute_flavor in a FlavorProfile\n\nRelated-Bug: #2158125\n\nChange-Id: Ia7067c1774a435d484637f46f3a83e2aea8cea02\nSigned-off-by: Shogo Yoshioka \u003csh-yoshioka@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7a68f0aa4e0d69e1c9a4dc2f7c87f53909f8051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7a68f0aa4e0d69e1c9a4dc2f7c87f53909f8051"}]},"branch":"refs/heads/master"},"08697321d8a2b351be2bebe238738a2d3149595f":{"kind":"REWORK","_number":2,"created":"2026-08-05 10:03:40.000000000","uploader":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"ref":"refs/changes/82/997182/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/997182/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/997182/2"}}},"commit":{"parents":[{"commit":"7b719284d616d24c0f85f3b751790beb56da0d6d","subject":"Merge \"Use native architecture when building Amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b719284d616d24c0f85f3b751790beb56da0d6d"}]}],"author":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-08-05 10:03:17.000000000","tz":540},"committer":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-08-05 10:03:17.000000000","tz":540},"subject":"Allow updating compute_flavor in a FlavorProfile","message":"Allow updating compute_flavor in a FlavorProfile\n\n- Adjust Octavia flavor profile validation so that even if a profile is\n  referenced by any Octavia flavor, an update is allowed when it modifies only\n  the `compute_flavor` key.\n- Continue to reject updates that modify `provider` or any other `flavor_data`\n  keys when the profile is referenced by any Octavia flavor.\n- Add the `FlavorProfileUpdateContentsError` exception to explain\n  why a `flavor_data` update failed.\n\nRelated-Bug: #2158125\n\nChange-Id: Ia7067c1774a435d484637f46f3a83e2aea8cea02\nSigned-off-by: Shogo Yoshioka \u003csh-yoshioka@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08697321d8a2b351be2bebe238738a2d3149595f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08697321d8a2b351be2bebe238738a2d3149595f"}]},"branch":"refs/heads/master"},"295609433f31191cdf14a09e0fa748f27b099425":{"kind":"REWORK","_number":3,"created":"2026-08-05 10:55:41.000000000","uploader":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"ref":"refs/changes/82/997182/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/997182/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/997182/3"}}},"commit":{"parents":[{"commit":"7b719284d616d24c0f85f3b751790beb56da0d6d","subject":"Merge \"Use native architecture when building Amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b719284d616d24c0f85f3b751790beb56da0d6d"}]}],"author":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-08-05 10:03:17.000000000","tz":540},"committer":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-08-05 10:54:58.000000000","tz":540},"subject":"Allow updating compute_flavor in a FlavorProfile","message":"Allow updating compute_flavor in a FlavorProfile\n\n- Adjust Octavia flavor profile validation so that even if a profile is\n  referenced by any Octavia flavor, an update is allowed when it modifies only\n  the `compute_flavor` key.\n- Continue to reject updates that modify `provider` or any other `flavor_data`\n  keys when the profile is referenced by any Octavia flavor.\n- Add the `FlavorProfileUpdateContentsError` exception to explain\n  why a `flavor_data` update failed.\n\nRelated-Bug: #2158125\n\nChange-Id: Ia7067c1774a435d484637f46f3a83e2aea8cea02\nSigned-off-by: Shogo Yoshioka \u003csh-yoshioka@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/295609433f31191cdf14a09e0fa748f27b099425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/295609433f31191cdf14a09e0fa748f27b099425"}]},"branch":"refs/heads/master"},"ae694bb4e64b6e7b8d1e8d8955363015ca1f9ae9":{"kind":"REWORK","_number":4,"created":"2026-08-12 14:35:13.000000000","uploader":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"ref":"refs/changes/82/997182/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/997182/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/997182/4"}}},"commit":{"parents":[{"commit":"7b719284d616d24c0f85f3b751790beb56da0d6d","subject":"Merge \"Use native architecture when building Amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b719284d616d24c0f85f3b751790beb56da0d6d"}]}],"author":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-08-05 10:03:17.000000000","tz":540},"committer":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-08-12 14:26:07.000000000","tz":540},"subject":"Allow updating compute_flavor in a FlavorProfile","message":"Allow updating compute_flavor in a FlavorProfile\n\n- For using Amphora provider only, we adjusted the validation process for\n  flavor profiles so that updates that modify only the `compute_flavor` key\n  are allowed, even if the profile is referenced by any Octavia flavor;\n  the Amphora driver validates that no other keys changed.\n- Any update that modifies other flavor_data keys still rejected,\n  even for Amphora.\n- For all non-Amphora providers, nothing changes:\n  any attempt to update flavor_data or provider_name while the profile is\n  referenced is still rejected.\n\nRelated-Bug: #2158125\n\nChange-Id: Ia7067c1774a435d484637f46f3a83e2aea8cea02\nSigned-off-by: Shogo Yoshioka \u003csh-yoshioka@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae694bb4e64b6e7b8d1e8d8955363015ca1f9ae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae694bb4e64b6e7b8d1e8d8955363015ca1f9ae9"}]},"branch":"refs/heads/master"},"43b71345cdd0f21df6cbd7a2c498d8d7966c6dad":{"kind":"REWORK","_number":5,"created":"2026-08-19 10:29:14.000000000","uploader":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"ref":"refs/changes/82/997182/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/997182/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/997182/5"}}},"commit":{"parents":[{"commit":"7b719284d616d24c0f85f3b751790beb56da0d6d","subject":"Merge \"Use native architecture when building Amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b719284d616d24c0f85f3b751790beb56da0d6d"}]}],"author":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-08-05 10:03:17.000000000","tz":540},"committer":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-08-19 10:26:39.000000000","tz":540},"subject":"Allow updating compute_flavor in a FlavorProfile","message":"Allow updating compute_flavor in a FlavorProfile\n\nCurrently, when an Octavia flavor profile is referenced by any Octavia\nflavor, Octavia does not allow updating `flavor_data`.\n\nIf an administrator needs to keep the same flavor profile name but swap\nthe compute flavor because hardware was retired, they must ensure no\nOctavia flavor references that profile; as long as any load balancer\nstill uses that flavor, Octavia blocks changes, forcing users to delete\nand recreate their load balancers, which is impractical when hardware\nrefreshes happen regularly.\n\nTherefore, administrators need to update only the `compute_flavor` entry\nin a flavor profile\u0027s `flavor_data` without touching the Octavia flavor\nitself, even if the flavor profile is currently referenced.\nSo, for using Amphora provider only, we adjusted the validation process\nfor flavor profiles so that updates that modify only the `compute_flavor`\nkey are allowed, even if the profile is referenced by any Octavia flavor;\nthe Amphora driver validates that no other keys changed.\n\nRelated-Bug: #2158125\n\nChange-Id: Ia7067c1774a435d484637f46f3a83e2aea8cea02\nSigned-off-by: Shogo Yoshioka \u003csh-yoshioka@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43b71345cdd0f21df6cbd7a2c498d8d7966c6dad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43b71345cdd0f21df6cbd7a2c498d8d7966c6dad"}]},"branch":"refs/heads/master"},"277021bede6d726626f39c9c0ca532175fd5e512":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-08-19 10:31:47.000000000","uploader":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"ref":"refs/changes/82/997182/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/997182/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/997182/6"}}},"commit":{"parents":[{"commit":"7b719284d616d24c0f85f3b751790beb56da0d6d","subject":"Merge \"Use native architecture when building Amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b719284d616d24c0f85f3b751790beb56da0d6d"}]}],"author":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-08-05 10:03:17.000000000","tz":540},"committer":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-08-19 10:30:49.000000000","tz":540},"subject":"Allow updating compute_flavor in a FlavorProfile","message":"Allow updating compute_flavor in a FlavorProfile\n\nCurrently, when an Octavia flavor profile is referenced by any Octavia\nflavor, Octavia does not allow updating `flavor_data`.\n\nIf an administrator needs to keep the same flavor profile name but swap\nthe compute flavor because hardware was retired, they must ensure no\nOctavia flavor references that profile; as long as any load balancer\nstill uses that flavor, Octavia blocks changes, forcing users to delete\nand recreate their load balancers, which is impractical when hardware\nrefreshes happen regularly.\n\nTherefore, administrators need to update only the `compute_flavor` entry\nin a flavor profile\u0027s `flavor_data` without touching the Octavia flavor\nitself, even if the flavor profile is currently referenced.\nSo, for using Amphora provider only, we adjusted the validation process\nfor flavor profiles so that updates that modify only the\n`compute_flavor` key are allowed, even if the profile is referenced by any\nOctavia flavor; the Amphora driver validates that no other keys changed.\n\nRelated-Bug: #2158125\n\nChange-Id: Ia7067c1774a435d484637f46f3a83e2aea8cea02\nSigned-off-by: Shogo Yoshioka \u003csh-yoshioka@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/277021bede6d726626f39c9c0ca532175fd5e512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/277021bede6d726626f39c9c0ca532175fd5e512"}]},"branch":"refs/heads/master"},"0691d2c5d32c131016099c029c13b171720b1f60":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-08-19 10:32:57.000000000","uploader":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"ref":"refs/changes/82/997182/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/997182/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/997182/7"}}},"commit":{"parents":[{"commit":"7b719284d616d24c0f85f3b751790beb56da0d6d","subject":"Merge \"Use native architecture when building Amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b719284d616d24c0f85f3b751790beb56da0d6d"}]}],"author":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-08-05 10:03:17.000000000","tz":540},"committer":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-08-19 10:32:02.000000000","tz":540},"subject":"Allow updating compute_flavor in a FlavorProfile","message":"Allow updating compute_flavor in a FlavorProfile\n\nCurrently, when an Octavia flavor profile is referenced by any Octavia\nflavor, Octavia does not allow updating `flavor_data`.\n\nIf an administrator needs to keep the same flavor profile name but swap\nthe compute flavor because hardware was retired, they must ensure no\nOctavia flavor references that profile; as long as any load balancer\nstill uses that flavor, Octavia blocks changes, forcing users to delete\nand recreate their load balancers, which is impractical when hardware\nrefreshes happen regularly.\n\nTherefore, administrators need to update only the `compute_flavor` entry\nin a flavor profile\u0027s `flavor_data` without touching the Octavia flavor\nitself, even if the flavor profile is currently referenced.\nSo, for using Amphora provider only, we adjusted the validation process\nfor flavor profiles so that updates that modify only the\n`compute_flavor` key are allowed, even if the profile is referenced by\nany Octavia flavor.\n\nRelated-Bug: #2158125\n\nChange-Id: Ia7067c1774a435d484637f46f3a83e2aea8cea02\nSigned-off-by: Shogo Yoshioka \u003csh-yoshioka@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0691d2c5d32c131016099c029c13b171720b1f60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0691d2c5d32c131016099c029c13b171720b1f60"}]},"branch":"refs/heads/master"},"848f8c49fa02b5473a4da047d65f5ce4101e7ff8":{"kind":"REWORK","_number":8,"created":"2026-08-24 22:10:04.000000000","uploader":{"_account_id":39270,"name":"Shogo Yoshioka","display_name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","username":"sh-yoshioka"},"ref":"refs/changes/82/997182/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/997182/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/997182/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/997182/8"}}},"commit":{"parents":[{"commit":"7b719284d616d24c0f85f3b751790beb56da0d6d","subject":"Merge \"Use native architecture when building Amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b719284d616d24c0f85f3b751790beb56da0d6d"}]}],"author":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-08-05 10:03:17.000000000","tz":540},"committer":{"name":"Shogo Yoshioka","email":"sh-yoshioka@kddi.com","date":"2026-08-24 22:09:41.000000000","tz":540},"subject":"Allow updating compute_flavor in a FlavorProfile","message":"Allow updating compute_flavor in a FlavorProfile\n\nCurrently, when an Octavia flavor profile is referenced by any Octavia\nflavor, Octavia does not allow updating `flavor_data`.\n\nIf an administrator needs to keep the same flavor profile name but swap\nthe compute flavor because hardware was retired, they must ensure no\nOctavia flavor references that profile; as long as any load balancer\nstill uses that flavor, Octavia blocks changes, forcing users to delete\nand recreate their load balancers, which is impractical when hardware\nrefreshes happen regularly.\n\nTherefore, administrators need to update only the `compute_flavor` entry\nin a flavor profile\u0027s `flavor_data` without touching the Octavia flavor\nitself, even if the flavor profile is currently referenced.\nSo, for using Amphora provider only, we adjusted the validation process\nfor flavor profiles so that updates that modify only the\n`compute_flavor` key are allowed, even if the profile is referenced by\nany Octavia flavor.\n\nRelated-Bug: #2158125\n\nChange-Id: Ia7067c1774a435d484637f46f3a83e2aea8cea02\nSigned-off-by: Shogo Yoshioka \u003csh-yoshioka@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/848f8c49fa02b5473a4da047d65f5ce4101e7ff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/848f8c49fa02b5473a4da047d65f5ce4101e7ff8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
