)]}'
{"id":"openstack%2Foctavia~998229","triplet_id":"openstack%2Foctavia~master~I435bbd6b3c1e3e7c4d2a00f19b9bde664a08da0c","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"32657":{"account":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"last_update":"2026-07-29 18:56:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I435bbd6b3c1e3e7c4d2a00f19b9bde664a08da0c","subject":"Add ARM64 as an accepted architecture for amphora image builds","status":"MERGED","created":"2026-07-21 21:02:49.000000000","updated":"2026-07-29 18:57:53.000000000","submitted":"2026-07-29 18:56:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998229","meta_rev_id":"5b626467dca1d98d513e9f9fbc23969cd5b27979","_number":998229,"virtual_id_number":998229,"owner":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-29 18:56:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":1,"date":"2026-07-29 13:10:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":1,"date":"2026-07-27 18:55:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-24 08:12:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":2,"date":"2026-07-29 11:45:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2026-07-29 13:42:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-21 23:42:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-23 08:54:55.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-07-24 08:12:29.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-07-27 18:55:01.000000000","updated_by":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"reviewer":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"state":"REVIEWER"},{"updated":"2026-07-29 11:45:19.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-07-29 13:10:59.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"8278d93a521f20e8104eff48a2ecce64fe19d30c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2026-07-21 21:02:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13c4432fb9e30ac00adca2687f362faeaaec62c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 23:42:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4e998cc11f74752b85d13b26789990f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41415f6495834138895f2a865bdc3932 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50bca727c3334e9db608451c2618de73 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/030a7037e0f24685b8e9787c64795a90 : SUCCESS in 5m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a32e2da7e1ec4a64807b8c110e7a09b0 : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2958a16789014e22983e7aca2715dbf1 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21762deaea914b208465d794aaea1fdf : SUCCESS in 12m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8f21c346ba44d1db52e7798ee3807d8 : SUCCESS in 5m 05s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/9b58886689644fefb362f8eb6ed4b577 : SUCCESS in 5m 56s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/ecec3183bbe44a8782b5eab069eec9a0 : SUCCESS in 5m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a444d462aa9a423381cf75bb3f590856 : SUCCESS in 3m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e53d9f3871614ad78c0243d533dd66ce : SUCCESS in 5m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8deef126781d4ad589ca07479880be45 : SUCCESS in 1h 02m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c8efd0bc98d64985b7971c022e70095d : SUCCESS in 1h 22m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fb5e620ec09241649c1b7e3b82d6bbac : SUCCESS in 1h 00m 11s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f4b863c715074a36b39a8d5fbed43cce : SUCCESS in 1h 28m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/36561a76f0ec4fa0b978e5bd802f11c3 : SUCCESS in 1h 02m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f2668f4fa8d14796aa178bea97f6f89b : SUCCESS in 35m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5798bde64d124bfe99838f4103bdb9f5 : SUCCESS in 46m 59s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c1d47d75b6de4942b896ab3bab172206 : SUCCESS in 41m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0aac62e6ad0647519605454190eac59e : SUCCESS in 32m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c2d2ddda082646b2a39f26428da8ce63 : FAILURE in 31m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d374c1308a8d4af88269823d8b2e0f04 : SUCCESS in 2h 25m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1c411b9072f345cbbe93a479959b9edd : SUCCESS in 1h 19m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/28a86b6376ee433b94761fa881d36a85 : TIMED_OUT in 2h 37m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3b5298961aad406e98a1ca91aad466ee : TIMED_OUT in 2h 36m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d8c53589a294da51475d0264c92910de54243d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2026-07-23 09:01:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"83fdec9a34ca9b6bb47983919ad765f952c7a3d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 11:40:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c40fac3b80b64fd4a41399fdb2a89d6d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ffa4b4c3ecf471fa4ffaab78eed953d : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/226bd8d2d1554aea8bdac923ecadf997 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d886a875acc485ba94b10226de97ba9 : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12f4fc633eeb46fd94ad81406247b49a : SUCCESS in 3m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/855d115925d34450b128258795455605 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07e2a048fb064799a07a4df1cd787b0c : SUCCESS in 14m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae6de43be3a342f9b4adbaa8e8424aea : SUCCESS in 8m 45s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/3c2c91a3bfed4869b6ff55d3b2520b0e : SUCCESS in 3m 35s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/4a22fb840cb44b198c845a81ce533aa0 : SUCCESS in 7m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/810e0bee772e46d28188a07ac6919e92 : SUCCESS in 3m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f949dddace1c4f45b4a0b60402f15ce9 : SUCCESS in 4m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/717b8650c9c342beb6283ca695c16ec4 : SUCCESS in 52m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e03b7d6855d34b79ab0f4f96b4f6728b : SUCCESS in 1h 20m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8bffe9bec2b14d78b8e823bb7953e771 : SUCCESS in 56m 17s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3d8f54b563524cc19a52560ca1ed1a45 : SUCCESS in 1h 25m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4418608e46b145fc89699f8b47c17b05 : SUCCESS in 1h 04m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/10a3fbb20fc7452c953bdb636682ffeb : SUCCESS in 34m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a9787bdb16d54e25b6afc2b83ede2408 : SUCCESS in 44m 59s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/57f24ab43e6543d6a83cab6a020e2853 : SUCCESS in 53m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8fbd6ccb3c64a85b09d4e5d777dcbae : SUCCESS in 31m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e656c2a5cf014e34be3493718536f3a0 : FAILURE in 38m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/944f16971a52488bb9dff79c14f8174d : SUCCESS in 2h 13m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/30e3b12a5afe4d3888343b14e40a3c01 : SUCCESS in 1h 25m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/12e444cda99644b4bf596c1c169ea9bb : TIMED_OUT in 2h 36m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b00d9ca89740405ab336cf1627f3c58c : TIMED_OUT in 2h 36m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f02f8d7eb22b2d5e3d7b7a212b3d17f11713a514","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-07-24 08:12:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b3dd677b3106a7e4612a467312a5b7c736ca5dcc","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-07-27 18:55:01.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84fc8d5feba4ab91b0b290fc055b7d2ad88e72d6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-29 11:45:19.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9a507f429c14f6aaab41f34c3989464d99352f5c","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2026-07-29 12:09:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd397411772ae437340adf6e5db09790a124bef0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-29 13:09:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1bdc85edc0824fabac9ce70dd167f517d4467443","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-07-29 13:10:59.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1361dbea7456a38fc0a21e3f2756245caacb5e30","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-29 13:42:33.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"da2ce155eacb4fe89256fd8d7b5f9b0bbdcdf617","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 13:43:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1051b2c221db4a96629971d13a8174588553b6d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 14:02:47.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a3f8b726afe4034b4f20467b96ef866\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9408fbc01e941d48e74cd8ec5120e7d : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e84a25cc5e6444ceab7f0f8717b5cd94 : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/204f56c860474a1590cd6afafc501d9b : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/901971382ad64887be16ebdfe6689432 : SUCCESS in 12m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dfc474c9d574623bc10833d49f217ff : SUCCESS in 9m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/085afdadbf214c589e1baf2ff86c7c06 : CANCELED\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ea726e0b40774766b32e42110919b330 : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6ce058a8ee134c7eb387030df099188f : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9379d75f107b4b2b90ccb9ecbb3fce60 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fb36e7aef7fc4677a88eb499863f43f0 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5bb0408384b249faac2a1756ccad518d : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a54e6d78bd5c4474b9167ee5c106d7fc : CANCELED\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2b45209d10b246deb5a46d5cf4f61974 : FAILURE in 18m 39s","accounts_in_message":[],"_revision_number":2},{"id":"4da8609fccead474bf0726c39821be7f4d64047d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-29 14:48:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bc546439e946c3a9f680db517e2eaf8407227774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 17:28:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a528f5ce55e841a395461389be0b5814\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9dc77d18024841d2aa444e1a47083147 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22497a4367c14c80bb284e3daaec9427 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7336a883ba50423b9db5156a611d4c8d : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c8737cbd75e24bfba135aff30e823ee7 : SUCCESS in 7m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aedc7fd9f8d9433b82650c6ec16e9465 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39f92fefeed141719d9e49d15db848fa : SUCCESS in 16m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da5df3353fbf4157bb81fef9b43a6b36 : SUCCESS in 9m 39s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/1a5368a7083043708e971a8449a6f525 : SUCCESS in 7m 15s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/b5c66874509c4b4e86ffa0d367d410d0 : SUCCESS in 3m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3471ddd8475040e48e61b857067ea892 : SUCCESS in 4m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6137e63754ef4401bc8ecec61e3d5cda : SUCCESS in 3m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4d2607466e874aed85b8d2db01a3f871 : SUCCESS in 1h 04m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/69c8df9d4af34ae5ad5e929d36fcfac8 : SUCCESS in 1h 36m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/43054a3d5aa84bb995641cd2750a3ca2 : SUCCESS in 1h 11m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/001b94bea49b4cd6a324db7166281d33 : SUCCESS in 1h 24m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3d807fb5bb5b436ab0ead51ffe12f634 : SUCCESS in 1h 13m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/63bddb89a9e34edfbd4bdd5b47b04810 : SUCCESS in 38m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9289228a69e4451ebb1516af69e94d57 : SUCCESS in 54m 45s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e2ebdc17b27040d1b9c28a3266dfbdc2 : SUCCESS in 45m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19ce83aeb50c4fdd8a3fa85e9f8b0416 : SUCCESS in 42m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e0972c561cb74ee4a5aeb45e94d45047 : FAILURE in 40m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/059157d2a5104c6ebe1d44047b271803 : SUCCESS in 2h 23m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0de4f156b52d4b5389910894c033cf32 : SUCCESS in 1h 21m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/71a8f85ee6154c418f11ec0fae823b57 : TIMED_OUT in 2h 37m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/75866542c9364bf480b844525635d97b : TIMED_OUT in 2h 37m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9b3ebf544655bd255f4d1506bb2be2bdce066ebf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 17:29:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"81fa7a0e23f6295e46115b4a11e4a577f37de58e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 18:56:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fa9deb087614c089dc264d4d98a0a87\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03e90c45a09745b3baf0a460d137151a : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47d3ca6e168b4678885f9f41801bdf48 : SUCCESS in 6m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd5cf292f0f740f482acdfe33b558868 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/324a983852bc425ab2ad19870321d534 : SUCCESS in 13m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e4ec838236b42e088bf0ed3e6f2311f : SUCCESS in 9m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/59c21c0784694f43b1d2e397e79729f9 : SUCCESS in 2m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/65981f65c8d54a9895028030bb6180ae : SUCCESS in 7m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7521621f659f4b50ae24a694c083be30 : SUCCESS in 1h 02m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/22172655c10541c0b6e3656f0229a53c : SUCCESS in 1h 26m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e10c5f40017049b8acd5428be58acc05 : SUCCESS in 59m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bc25de2cb39d4fc6bb6e64f804dfa389 : SUCCESS in 34m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1d384fe0d7cd47378e62a0247748d114 : SUCCESS in 49m 18s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/3f2be8c4c03d4a7aa8e90ce7a5519b26 : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":2},{"id":"5b2f5f7a3e3d607aac3316beda171b0c9770fb7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 18:56:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5b626467dca1d98d513e9f9fbc23969cd5b27979","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 18:57:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca9f11994f244747b0bb794c57d1f4d1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e5e835ff121459080cb75b716a0d77e : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0901a93ebe1a4c669e7e9ecff8ee7b45 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d6745a8bf913f54bd4d635fc548c152162e53820","revisions":{"5b76af97e0423e773f4063eb9b3e4d3c04c62233":{"kind":"REWORK","_number":1,"created":"2026-07-21 21:02:49.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/29/998229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/998229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/998229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/998229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/998229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/998229/1"}}},"commit":{"parents":[{"commit":"78d38afc126c37e2ec65859e8cfc770eb5457075","subject":"Remove setting of oslo_policy[enforce_scope] flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2026-07-21 20:59:47.000000000","tz":120},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2026-07-21 21:02:19.000000000","tz":120},"subject":"Add ARM64 as an accepted architecture for amphora image builds","message":"Add ARM64 as an accepted architecture for amphora image builds\n\nCloses-Bug: 2161068\n\nSigned-off-by: Piotr Parczewski \u003cpiotr@stackhpc.com\u003e\nChange-Id: I435bbd6b3c1e3e7c4d2a00f19b9bde664a08da0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b76af97e0423e773f4063eb9b3e4d3c04c62233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b76af97e0423e773f4063eb9b3e4d3c04c62233"}]},"branch":"refs/heads/master"},"d6745a8bf913f54bd4d635fc548c152162e53820":{"kind":"REWORK","_number":2,"created":"2026-07-23 09:01:10.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/29/998229/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/998229/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/998229/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/998229/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/998229/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/998229/2"}}},"commit":{"parents":[{"commit":"78d38afc126c37e2ec65859e8cfc770eb5457075","subject":"Remove setting of oslo_policy[enforce_scope] flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78d38afc126c37e2ec65859e8cfc770eb5457075"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2026-07-21 20:59:47.000000000","tz":120},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2026-07-23 09:01:01.000000000","tz":120},"subject":"Add ARM64 as an accepted architecture for amphora image builds","message":"Add ARM64 as an accepted architecture for amphora image builds\n\nCloses-Bug: 2161068\n\nSigned-off-by: Piotr Parczewski \u003cpiotr@stackhpc.com\u003e\nChange-Id: I435bbd6b3c1e3e7c4d2a00f19b9bde664a08da0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6745a8bf913f54bd4d635fc548c152162e53820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6745a8bf913f54bd4d635fc548c152162e53820"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
