)]}'
{"id":"openstack%2Foctavia~999475","triplet_id":"openstack%2Foctavia~master~Ief196a1d2a98ae34a0ae4c265d4a8704f50890ac","project":"openstack/octavia","branch":"master","attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-08-12 11:52:02.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-09-18 22:04:48.000000000","reason":"Added by \u003cGERRIT_ACCOUNT_37598\u003e using the hovercard menu","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-12 11:52:06.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ief196a1d2a98ae34a0ae4c265d4a8704f50890ac","subject":"Remove legacy [neutron] options","status":"NEW","created":"2026-07-31 23:49:49.000000000","updated":"2026-09-18 22:04:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3956085d37db0f542a5c7bc33494fcc96391ab9c","_number":999475,"virtual_id_number":999475,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-01 06:11:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-12 11:52:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-01 02:29:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-12 11:52:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-08-12 11:52:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-08-12 11:52:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"acdaaf6be89b2acf471c58795f9738f44bea7352","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-31 23:49:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b82287f3fb1b56ad0f45058abb519be76ed8bf31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-01 02:29:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d1f410bb66341e98ac568ddd4c3861c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1aea17349f7249a396d93568d1e012f9 : FAILURE in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e2302f86f2246f195c0ac307d627930 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/536d67894d4942f58ccac10f9f1cd319 : FAILURE in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c80a67ed01cb408db9bec6d2e3c08ff4 : FAILURE in 3m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9b4f3dca735a45bba5df792c12071042 : FAILURE in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6cc8f8715534157a8518478a91f1ffc : FAILURE in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c8a0bf256b5456bb11aac87337f3689 : SUCCESS in 7m 01s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/7ac0a416866b47b2b45842de743010d9 : FAILURE in 6m 09s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/c5823d6fe655404994c138e910e00bf1 : SUCCESS in 7m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/384aead008eb4bf0b020cbbade36ea83 : SUCCESS in 3m 34s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2041e3008e6e48d8984b298ff606cbf7 : SUCCESS in 5m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/17466b4f8a2b44a49b786e912f5bb839 : SUCCESS in 1h 03m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fad9cd891eeb4937a1f3adcd5f6eb1da : SUCCESS in 1h 20m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c166ec68bf494bcbb2dfcc112c9cb121 : SUCCESS in 53m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f93f7688fcc749bbb861c7a1c34134f7 : SUCCESS in 1h 24m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6913a06e83b944818c0a4772e07a3019 : SUCCESS in 58m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2c91e39aa9754ad6b6d2c3c92a139213 : SUCCESS in 34m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dc05d7eeba114653ba79a2050644758b : SUCCESS in 40m 45s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/6de7179f851f4d1f9b5a144cd7afd812 : SUCCESS in 44m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d82f459f08124998bda6e901a5cf0da0 : SUCCESS in 33m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ed9d3f74a6c0458186a38ee3305808a9 : SUCCESS in 35m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6e56f1dadeb846b7ad52c147f8a02804 : SUCCESS in 2h 16m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/df7c8754cdfc4e2fb9118aff0845f03c : SUCCESS in 1h 26m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/e595306c0eef4b40bc89008c2b15df57 : TIMED_OUT in 2h 37m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c4349f14a3ae45daa52d4a9bb396ec22 : SUCCESS in 2h 21m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e73de2763563e19938683973ebcdb94e8e276098","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-01 03:31:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7d62cf5419e6743fb0ac2da5f1bc7045188183be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-01 06:11:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48b058c78d2f42799473e70339299ed0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8b1a58d65794b5e8d7a2ef3ac51935a : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eee9084a0bb48a4a7f58d0ea7d295a0 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18d49d7e23ee448092eaecccac27e540 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7182b0227a040569be1cfc19cc70b6c : SUCCESS in 4m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae524402a3bc4333a7a1b9077db1d91e : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bb69e738b004f78a2a146fa30b30d91 : SUCCESS in 15m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71bbace07f6b4570a813e88622cb91c5 : SUCCESS in 9m 38s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/4aae2638a8e84d668690b7d4d472d20e : SUCCESS in 3m 09s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/61e4d2a0c39b44058356bd0ebb022874 : SUCCESS in 7m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/da2287409d05417587d27bc94fc7ae50 : SUCCESS in 3m 35s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1437572a1f8b40d78346ba977c88aaa8 : SUCCESS in 3m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/382fbece91d84249af454c4590bc7de9 : SUCCESS in 59m 52s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/68687a04858c40f9a36edfef38f5f002 : SUCCESS in 1h 20m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/325142ae04514ccdb7ecc17e6641bd86 : SUCCESS in 1h 03m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3953ef0b87ce4d73b4a14c3fa7234e47 : SUCCESS in 1h 24m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d4cb7f056439425e8e6f6ceba8de54db : SUCCESS in 59m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3b4575cd999c4970ae323efb0582bab9 : SUCCESS in 33m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a7c1021eb7474999be9d9fb0f409a2d7 : SUCCESS in 48m 51s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/6679e3ed57214cd3b80f2e2363829b63 : SUCCESS in 44m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d186290ff3a4584b79533f6ebad0e6a : SUCCESS in 31m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2cdb2604754c4a54bd4b62d440a06f8a : SUCCESS in 33m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e204e50f65544b5880132f18c88f7e73 : SUCCESS in 2h 39m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/40f2c956d8334b1a884b1db2e83a3618 : SUCCESS in 1h 21m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/9e201f797e024744a987238d172bce00 : TIMED_OUT in 2h 37m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/eec151600cb948968581c227c621a986 : SUCCESS in 2h 34m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"33d0369e331e7bb74aa26ee200fbc1995cd98ec2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-12 11:52:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ac0ceb4b63bee3fba2273a6e2c1dac283b69b87","revisions":{"05f2759d4cf2bde2cf37fa83de6f23991c65fd9b":{"kind":"REWORK","_number":1,"created":"2026-07-31 23:49:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/999475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/999475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/999475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/999475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/999475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/999475/1"}}},"commit":{"parents":[{"commit":"aa6643b013a6e753818e552c4e3345b28eb51477","subject":"Merge \"[Worker]: use retries from conf for tasflow persistence\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa6643b013a6e753818e552c4e3345b28eb51477"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-31 23:47:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-31 23:47:18.000000000","tz":540},"subject":"Remove legacy [neutron] options","message":"Remove legacy [neutron] options\n\nThese options have been deprecated for 2+ years.\n\nChange-Id: Ief196a1d2a98ae34a0ae4c265d4a8704f50890ac\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05f2759d4cf2bde2cf37fa83de6f23991c65fd9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05f2759d4cf2bde2cf37fa83de6f23991c65fd9b"}]},"branch":"refs/heads/master"},"8ac0ceb4b63bee3fba2273a6e2c1dac283b69b87":{"kind":"REWORK","_number":2,"created":"2026-08-01 03:31:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/999475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/999475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/999475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/999475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/999475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/999475/2"}}},"commit":{"parents":[{"commit":"aa6643b013a6e753818e552c4e3345b28eb51477","subject":"Merge \"[Worker]: use retries from conf for tasflow persistence\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa6643b013a6e753818e552c4e3345b28eb51477"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-31 23:47:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-01 03:31:01.000000000","tz":540},"subject":"Remove legacy [neutron] options","message":"Remove legacy [neutron] options\n\nThese options have been deprecated for 2+ years.\n\nChange-Id: Ief196a1d2a98ae34a0ae4c265d4a8704f50890ac\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ac0ceb4b63bee3fba2273a6e2c1dac283b69b87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ac0ceb4b63bee3fba2273a6e2c1dac283b69b87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
