)]}'
{"id":"openstack%2Foctavia~999893","triplet_id":"openstack%2Foctavia~master~If9c515056a39236a7b1266159731eba9ce20d18a","project":"openstack/octavia","branch":"master","attention_set":{"28258":{"account":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"last_update":"2026-09-11 19:22:33.000000000","reason":"A robot voted negatively on a label"},"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-09-11 17:00:31.000000000","reason":"\u003cGERRIT_ACCOUNT_28258\u003e replied on the change","reason_account":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-09-11 16:51:22.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If9c515056a39236a7b1266159731eba9ce20d18a","subject":"Support amphora failover when vip-subnet is full","status":"NEW","created":"2026-08-05 18:48:57.000000000","updated":"2026-09-11 23:54:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":18,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"e4b2b4d617c59853f684a96260dd4ef1672e11a7","_number":999893,"virtual_id_number":999893,"owner":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-11 23:54:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-05 21:29:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-27 19:50:03.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-03 12:46:18.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"93123b41bd50407ca8c8e07b69a3f4dda4b538f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"date":"2026-08-05 18:48:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa5333cbedcc83dcbb64f74fba03aa37c9ef89a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 21:29:14.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8917efc13e61405a9169108ded81aced\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f1ee862632a43818a79add64185557c : FAILURE in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05e09174c50f46ec899b1a9b7e4f33a0 : FAILURE in 2m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf584b05ca894a74bc52962899566854 : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8dfec709744d417f93fdeb43d4976f27 : SUCCESS in 2m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b9a6e7d9211a458687b4ac9df66db40d : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a969adcdff7e45a8abeb2e745ffe1a24 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e05f044510e499d856a0302bb68f4e7 : SUCCESS in 4m 51s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/afa21bcc4d2f4966b18e2c5f2b5e78bb : SUCCESS in 4m 18s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/0e3ba892cf3b43d9b8c0f32872e9db64 : FAILURE in 5m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4742f20147af477fb07b349ec6fd52c9 : SUCCESS in 4m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d19a6e82715142cfa1baaec936e27a5b : FAILURE in 5m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/755592ccdf6d4aadb1a1fc404913588c : FAILURE in 55m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a38162976225411c943e33413c33c23f : SUCCESS in 1h 19m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7ad87cfcb37d426c9298d290c47b8729 : SUCCESS in 56m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c465d96071e447c18766c2b5c6283b9b : SUCCESS in 1h 24m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/007597e5b5e04315afea50a5b2344922 : SUCCESS in 59m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/afc689d1a4154b6391bcbdff08f443de : SUCCESS in 33m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e8489d3178ab42f7a8c6f0c06055e3f7 : FAILURE in 46m 28s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f062377d9a994b3087aaf75a7821a439 : SUCCESS in 40m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f938ce0fbc294ef7a29494a3348fa272 : SUCCESS in 30m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9f9d3f95fdcb4ea3aba436fcec3aebbf : SUCCESS in 33m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/011ebb003b27459c894bf87b70ab93da : SUCCESS in 2h 33m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/adab327187cd4b1ab01fb28bd39dc644 : SUCCESS in 1h 18m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/83adeb82fcab47d188f7f18880ac0f5f : TIMED_OUT in 2h 37m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b06e31ceb14c4a9ba2a59e3139b02a34 : TIMED_OUT in 2h 36m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17cc786f1f9049e7906a61f240b79e1ee455cef5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"date":"2026-08-25 11:34:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"05aa0933beb1764b9c3dc922a523eccea10ebe10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 14:08:47.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9662d70aa7ac41d797acfa709f7c323a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d1f3706e1d44b889678da7869bfdbf6 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afc77f41e27641c981a43102b3cb485c : FAILURE in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb155bb4277b4f0fa27923cd270d7ff8 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/23304ff2b3ff4ddc9a18f7efd826c393 : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/86da0f7fd5df4d68b44a927ec895a16a : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84264b9042f34f89a9dda8ec30d42392 : SUCCESS in 14m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4491cc6ea3ba415b800e9ebb2d01712a : SUCCESS in 5m 21s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/57dde79b1c994b4ea916d0a1aefaf070 : SUCCESS in 5m 58s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/370374ae2270457083ef15d2cc8135b9 : SUCCESS in 5m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e5f500c57cd843119f1c47b4c0352a60 : SUCCESS in 3m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5ef1dd6f016a4db8a33e877016c7e3a8 : SUCCESS in 4m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/96e7e14c24964d64a53ab89a1fa1ecc7 : FAILURE in 1h 01m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f4c3e66423544a3784849240010a6e7f : SUCCESS in 1h 24m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/193fc4a52549406f900a3dea82e1f973 : SUCCESS in 1h 08m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d16e0f10b5f04c4ea08eb3dc82ea3018 : SUCCESS in 1h 32m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/baa2fcbe59354c5ba3be1697b3b1ae2d : SUCCESS in 1h 04m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bfe8e8bace74458983bbbc40fc3ebb78 : SUCCESS in 40m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a2a2beed4c5a4601bbdcdbdec763f061 : SUCCESS in 45m 20s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/345fca61da804e48915d62ee9ebdd616 : SUCCESS in 47m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ab06b13f87049fa876385d10a0f407a : SUCCESS in 32m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/91c8c06652674cedaad979ab0d08f9ae : SUCCESS in 40m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e20f8ad18f5041ababbad3ac30f613e1 : SUCCESS in 2h 27m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b601d670412d42f086621be173c3ee55 : FAILURE in 26m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/88af9f5ceb1a4d97a51f283ed549d6d9 : SUCCESS in 1h 19m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0ea0f39774e346e48b4ad125403450da : SUCCESS in 1h 02m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1113ddb866bf043f21e2e456003600f060c69f19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"date":"2026-08-27 19:43:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f35a516107318654ef5c5cce8337f095d8ac6ed4","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 19:50:03.000000000","message":"Patch Set 3: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c144d9745169429881c31b750190b0f5\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/fe1bbb2e904c4006b3ca6320f06e3b02 : SUCCESS in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2a30500517aaddf04cd4ead6ed879cb16feb3159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 21:17:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d30ae44584d74bf1b5382a524c47b16b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fbb9184b6334ce7bec49b7d3582b984 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2f2b7259f7c4c209d274b83189c72eb : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/789dcacbe51c4f17b48dbabcc04d898b : SUCCESS in 6m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79d3fb79f6f44a9f9a73a3be9d83ba78 : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ef947bc1aa104c29a4e3cf5508432322 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a08b01eed6a94bd8b73d57d412eb5b6b : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8569be9f2e614f1fbf8eddd0d1d99403 : SUCCESS in 3m 57s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/e0b3b5c835c342de9f7b6109f32de8fb : SUCCESS in 4m 54s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/960ea6fb1fa54cefaf63912b75477486 : SUCCESS in 7m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e53cbe16e742494b8098a99b1b970e03 : SUCCESS in 4m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5d8d4fbd404941a1863fe053626876b9 : SUCCESS in 3m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2c385ac8ce21407882ef0cd501fe8662 : SUCCESS in 27m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/99517142cd7d43a2a0311c1922991fd4 : SUCCESS in 1h 27m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3b02c96381614b9ba77ca6840732a6b6 : SUCCESS in 1h 08m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/eba8bbf10f3a4e929d656fc7889f3a7d : SUCCESS in 1h 32m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2f87e6ff107c4f5d97add49a9a281107 : FAILURE in 22m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2aea84f32633495fa6e4e0a07e8d7687 : FAILURE in 23m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/217c928bb7894ad58108ac7916d14e73 : SUCCESS in 46m 25s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/88bba53b2fcc47e9a9b06046d169b3bd : SUCCESS in 44m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f5f1b3a80797428a8fdfe56f32aa4a4e : FAILURE in 22m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dfc16c80f03a46ba9b348bfc9901171e : FAILURE in 20m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b55c125f5f2544248af20a3ae26cc3d0 : FAILURE in 30m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2087b655b8b14a54a4f4dd86a50e50cd : SUCCESS in 1h 27m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/28451787627e4c01a31f5fa064c673de : FAILURE in 8m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/44cc228c9a61431bbff24e9bdcced10f : FAILURE in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b5b53ab10620f9c939ecbb618fc0ec1732e47963","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-09-03 12:46:18.000000000","message":"Patch Set 3: Backport-Candidate-1 Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3e49c8f1c0d145452cb7b937e95b80d278f0c077","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"date":"2026-09-11 16:51:22.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"9bd4af5827dc06018df192c896171d49ac68c664","author":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"date":"2026-09-11 17:00:31.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fd2e166c87b6be4bd0f3434e0509dc8c1a9e99f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 19:22:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a00843571094c3cb383686c2476acb4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62c24b4771d64c528a6097fab04789da : FAILURE in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/800bd8ff1fb14fe7ab39a71fcfddb2c2 : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/364bcadde8db409d8d3ff0f150062bb0 : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a9f27dd1334429ea9bd8d912f408fab : SUCCESS in 4m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6c295cb9053f49d3af6a5e94172ffca4 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/433bf3c74a1142e1930bc9ab4b9bb39a : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41a41843acf147318db854993ac5881a : SUCCESS in 10m 31s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/698c56db3ab7491fa6fed809a872817b : SUCCESS in 6m 36s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/7cff8404a400488697c4ecf7f04eabe3 : FAILURE in 5m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/093ad0a4f3a345abbb43cf7b65d8b327 : SUCCESS in 3m 35s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c4c65886c03940bca9b7bba4a370d06f : FAILURE in 5m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/49e47a9935ca4caeb84e3f9bf97cafef : SUCCESS in 39m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0e96668e086849d681866af5bd2e2b40 : SUCCESS in 1h 22m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9c8bc47d325e4cd38195eba5ed5a6617 : SUCCESS in 1h 06m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/932dd6f06004454ab26d9a5d1cb7c509 : SUCCESS in 1h 29m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/84b06cc2734f4fd8ac217b589183fb58 : SUCCESS in 59m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a76adcefce9044d6b2de85ec27870acf : SUCCESS in 37m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/60d4c5e81f784e85a062a90f00a430de : SUCCESS in 43m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b274006de7eb40cda7a1e59bffc7d609 : SUCCESS in 31m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cf831340d48c4a4ea0e1dcd9542b1e6c : SUCCESS in 36m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/88ae9fdccb7040819a05c697eebdc262 : SUCCESS in 2h 29m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ca0aaa13f45b4927bdff9ae25630e514 : SUCCESS in 1h 22m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/3b6e0408abbd41d99e0a6af92eed0cb1 : SUCCESS in 1h 16m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/84000f67fa484721bfa811470fb14c11 : SUCCESS in 1h 02m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5a302d3d59500f4075113387f14478169b842473","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"date":"2026-09-11 21:16:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"e4b2b4d617c59853f684a96260dd4ef1672e11a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 23:54:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37c477e2939b4507a7b4665d7748b73c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be641ef91d1345888b2311c8e5964f64 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/437b04f174b34687ba68cfe8130d4bd1 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ad8ced5ff11143e293f529f540d1ddaa : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba4231ab6b374f458eb6da1a95022405 : SUCCESS in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bed5161b6d904ee0846ddf5a42c0af9a : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5ccbe0b6b4f4ee6b8836264c6a8b032 : SUCCESS in 13m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41956fa0a35440f2b2a384c528f1a638 : SUCCESS in 5m 17s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/69b68cb9ac89465cbd228c22a145a315 : SUCCESS in 4m 58s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/443d9f2202de4fd5810b7763f143caba : SUCCESS in 5m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/891dddec777e42db92fb77cebab51ae9 : SUCCESS in 3m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8f44311934144aa8bf1ead9aedd8bf6d : SUCCESS in 5m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/358399d182564a33bdeb33f26808aae5 : SUCCESS in 1h 14m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/873c808f3c4e4b67aeb963a1d9f5f46a : SUCCESS in 1h 24m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a73239ac945e483bb82358c8b4f91f1f : SUCCESS in 1h 07m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/aa601f6036bc4873b33a6ee2b24e8fbc : SUCCESS in 1h 28m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b69eb0f70b484ada8b29507c1197708d : SUCCESS in 1h 03m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6974cbcb5ff340ff84dc9dd8fd259919 : SUCCESS in 36m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/eea99874b8c9476e8acd3b803e39dbac : SUCCESS in 44m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95c0e2d01bf84d6c92a13b096b93b98a : SUCCESS in 33m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2d06a29b406643cdb83b709156b55d9f : SUCCESS in 35m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2ca62a96c05942219fa12b4768207d80 : SUCCESS in 2h 36m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c3df37c313d94c49822b8853ddcc7a2f : SUCCESS in 1h 22m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/f73b89e547a74f12973fb194f56b36a1 : SUCCESS in 1h 13m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c38b18d3eebc435d96e9fd0be91a4daa : SUCCESS in 1h 01m 31s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"041926b545d767e9637b3d04bf7f65e166650c35","revisions":{"859ab7cd2e10a405276e4f1bd26f9f6e2ce46bec":{"kind":"REWORK","_number":1,"created":"2026-08-05 18:48:57.000000000","uploader":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"ref":"refs/changes/93/999893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/999893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/999893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/999893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/999893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/999893/1"}}},"commit":{"parents":[{"commit":"7b719284d616d24c0f85f3b751790beb56da0d6d","subject":"Merge \"Use native architecture when building Amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b719284d616d24c0f85f3b751790beb56da0d6d"}]}],"author":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2026-07-29 02:25:14.000000000","tz":-240},"committer":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2026-08-05 18:46:55.000000000","tz":-240},"subject":"Enables amphora VRRP ports to be on different subnets on the same network when there is a loadbalancer failover.  Performs ip availability check on the vip-network and selects alternate subnet with free IPs if the vip-subnet is full. Installs routes on amphoras for all subnets in vip-network to ensure VRRP connnectivity when the amphoras have VRRP ports on different subnets.","message":"Enables amphora VRRP ports to be on different subnets on the same network when there is a loadbalancer failover.  Performs ip availability check on the vip-network and selects alternate subnet with free IPs if the vip-subnet is full. Installs routes on amphoras for all subnets in vip-network to ensure VRRP connnectivity when the amphoras have VRRP ports on different subnets.\n\nCloses-Bug: #2162916\nChange-Id: If9c515056a39236a7b1266159731eba9ce20d18a\nSigned-off-by: Austin Russell \u003ckenrusse@cisco.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/859ab7cd2e10a405276e4f1bd26f9f6e2ce46bec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/859ab7cd2e10a405276e4f1bd26f9f6e2ce46bec"}]},"branch":"refs/heads/master"},"1705ff33015ece15167bbb323d9dbb8ab346155c":{"kind":"REWORK","_number":2,"created":"2026-08-25 11:34:02.000000000","uploader":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"ref":"refs/changes/93/999893/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/999893/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/999893/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/999893/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/999893/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/999893/2"}}},"commit":{"parents":[{"commit":"1dfb009d3fe30db7c0fb43c73120f98a420ddf1b","subject":"Merge \"Fix devstack unstack fail when Redis/Valkey not installed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1dfb009d3fe30db7c0fb43c73120f98a420ddf1b"}]}],"author":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2026-08-25 11:26:49.000000000","tz":-240},"committer":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2026-08-25 11:33:03.000000000","tz":-240},"subject":"Enables amphora VRRP ports to be on different subnets on the same network when there is a loadbalancer failover.  Performs ip availability check on the vip-network and selects alternate subnet with free IPs if the vip-subnet is full. Installs routes on amphoras for all subnets in vip-network to ensure VRRP connnectivity when the amphoras have VRRP ports on different subnets.","message":"Enables amphora VRRP ports to be on different subnets on the same network when there is a loadbalancer failover.  Performs ip availability check on the vip-network and selects alternate subnet with free IPs if the vip-subnet is full. Installs routes on amphoras for all subnets in vip-network to ensure VRRP connnectivity when the amphoras have VRRP ports on different subnets.\n\nCloses-Bug: #2162916\nChange-Id: If9c515056a39236a7b1266159731eba9ce20d18a\nSigned-off-by: Austin Russell \u003ckenrusse@cisco.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1705ff33015ece15167bbb323d9dbb8ab346155c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1705ff33015ece15167bbb323d9dbb8ab346155c"}]},"branch":"refs/heads/master"},"d8dd0c439b35e8882ac04a043695c60c75bc9271":{"kind":"REWORK","_number":3,"created":"2026-08-27 19:43:37.000000000","uploader":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"ref":"refs/changes/93/999893/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/999893/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/999893/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/999893/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/999893/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/999893/3"}}},"commit":{"parents":[{"commit":"1dfb009d3fe30db7c0fb43c73120f98a420ddf1b","subject":"Merge \"Fix devstack unstack fail when Redis/Valkey not installed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1dfb009d3fe30db7c0fb43c73120f98a420ddf1b"}]}],"author":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2026-08-25 11:26:49.000000000","tz":-240},"committer":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2026-08-27 19:42:57.000000000","tz":-240},"subject":"Enables amphora VRRP ports to be on different subnets on the same network when there is a loadbalancer failover.  Performs ip availability check on the vip-network and selects alternate subnet with free IPs if the vip-subnet is full. Installs routes on amphoras for all subnets in vip-network to ensure VRRP connnectivity when the amphoras have VRRP ports on different subnets.","message":"Enables amphora VRRP ports to be on different subnets on the same network when there is a loadbalancer failover.  Performs ip availability check on the vip-network and selects alternate subnet with free IPs if the vip-subnet is full. Installs routes on amphoras for all subnets in vip-network to ensure VRRP connnectivity when the amphoras have VRRP ports on different subnets.\n\nCloses-Bug: #2162916\nChange-Id: If9c515056a39236a7b1266159731eba9ce20d18a\nSigned-off-by: Austin Russell \u003ckenrusse@cisco.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8dd0c439b35e8882ac04a043695c60c75bc9271"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d8dd0c439b35e8882ac04a043695c60c75bc9271"}]},"branch":"refs/heads/master"},"998c0f91929754823df1889eef98b5c718c2c083":{"kind":"REWORK","_number":4,"created":"2026-09-11 16:51:22.000000000","uploader":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"ref":"refs/changes/93/999893/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/999893/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/999893/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/999893/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/999893/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/999893/4"}}},"commit":{"parents":[{"commit":"93689cb7701da65b5aad387bb824c477c64b0f2b","subject":"Merge \"Remove legacy Upstart configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93689cb7701da65b5aad387bb824c477c64b0f2b"}]}],"author":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2026-08-25 11:26:49.000000000","tz":-240},"committer":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2026-09-11 16:51:08.000000000","tz":-240},"subject":"Support amphora failover when vip-subnet is full","message":"Support amphora failover when vip-subnet is full\n\nEnables amphora VRRP ports to be on different subnets on the same network when there is a loadbalancer failover.  Performs ip availability check on the vip-network and selects alternate subnet with free IPs if the vip-subnet is full. Installs routes on amphoras for all subnets in vip-network to ensure VRRP connnectivity when the amphoras have VRRP ports on different subnets.\n\nCloses-Bug: #2162916\nChange-Id: If9c515056a39236a7b1266159731eba9ce20d18a\nSigned-off-by: Austin Russell \u003ckenrusse@cisco.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/998c0f91929754823df1889eef98b5c718c2c083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/998c0f91929754823df1889eef98b5c718c2c083"}]},"branch":"refs/heads/master"},"041926b545d767e9637b3d04bf7f65e166650c35":{"kind":"REWORK","_number":5,"created":"2026-09-11 21:16:14.000000000","uploader":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"ref":"refs/changes/93/999893/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/999893/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/999893/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/999893/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/999893/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/999893/5"}}},"commit":{"parents":[{"commit":"93689cb7701da65b5aad387bb824c477c64b0f2b","subject":"Merge \"Remove legacy Upstart configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93689cb7701da65b5aad387bb824c477c64b0f2b"}]}],"author":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2026-08-25 11:26:49.000000000","tz":-240},"committer":{"name":"Austin Russell","email":"kenrusse@cisco.com","date":"2026-09-11 21:15:47.000000000","tz":-240},"subject":"Support amphora failover when vip-subnet is full","message":"Support amphora failover when vip-subnet is full\n\nEnables amphora VRRP ports to be on different subnets on the same network when there is a loadbalancer failover.  Performs ip availability check on the vip-network and selects alternate subnet with free IPs if the vip-subnet is full. Installs routes on amphoras for all subnets in vip-network to ensure VRRP connnectivity when the amphoras have VRRP ports on different subnets.\n\nCloses-Bug: #2162916\nChange-Id: If9c515056a39236a7b1266159731eba9ce20d18a\nSigned-off-by: Austin Russell \u003ckenrusse@cisco.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/041926b545d767e9637b3d04bf7f65e166650c35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/041926b545d767e9637b3d04bf7f65e166650c35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
