)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~657999","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~master~Ie95b20c94358ce176bac4161ad01a4c1428adf7d","project":"openstack/openstack-ansible-apt_package_pinning","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ie95b20c94358ce176bac4161ad01a4c1428adf7d","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-09 11:33:26.000000000","updated":"2019-05-09 12:27:07.000000000","submitted":"2019-05-09 12:27:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657999-1557404827490-85cee9de","meta_rev_id":"22244cbb72ca5e6dd615e17c0e2794f6adbbabee","_number":657999,"virtual_id_number":657999,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-05-09 12:27:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-05-09 12:06:30.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-05-09 12:06:30.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 12:06:30.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-05-09 12:27:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"729a9562d5d4a8b94c455c88805a40dff0de0ada","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-09 11:33:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c00567d07a81d340cae4489c71c32746806098e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 11:53:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/99/657999/1/check/openstack-ansible-linters/4d3b3ea/ : SUCCESS in 4m 23s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/99/657999/1/check/openstack-ansible-functional-debian-stable/5a45c19/ : SUCCESS in 4m 05s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/99/657999/1/check/openstack-ansible-functional-centos-7/c5b2d2e/ : SUCCESS in 6m 19s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/99/657999/1/check/openstack-ansible-functional-ubuntu-bionic/3669212/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/99/657999/1/check/openstack-tox-docs/7da343f/html/ : SUCCESS in 2m 09s\n- build-openstack-releasenotes http://logs.openstack.org/99/657999/1/check/build-openstack-releasenotes/7bdc2f1/html/ : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9114766a83d49d4c3373f45b0d87c38b2e9d2655","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-09 12:06:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"480df5e0a34c76cf020492188028effaeb88d4bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 12:06:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ce8e4d39c36d8fe6c3256748469cbda01454bdf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 12:27:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3baed34e6060b8e3af8f47409f685d0328089c65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 12:27:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/99/657999/1/gate/openstack-ansible-linters/b36e8e1/ : SUCCESS in 7m 43s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/99/657999/1/gate/openstack-ansible-functional-debian-stable/ebe18b8/ : SUCCESS in 6m 16s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/99/657999/1/gate/openstack-ansible-functional-centos-7/11b9cbc/ : SUCCESS in 9m 13s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/99/657999/1/gate/openstack-ansible-functional-ubuntu-bionic/4a247aa/ : SUCCESS in 5m 21s\n- openstack-tox-docs http://logs.openstack.org/99/657999/1/gate/openstack-tox-docs/10f8d58/html/ : SUCCESS in 2m 32s\n- build-openstack-releasenotes http://logs.openstack.org/99/657999/1/gate/build-openstack-releasenotes/3546f1b/html/ : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82d2897837991c68ec100fd260f15e0e716cd029","revisions":{"82d2897837991c68ec100fd260f15e0e716cd029":{"kind":"REWORK","_number":1,"created":"2019-05-09 11:33:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/657999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/99/657999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/99/657999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/99/657999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/99/657999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/99/657999/1"}}},"commit":{"parents":[{"commit":"f13f0ddea441125f07b7635d2ba5442e543a2c8c","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/f13f0ddea441125f07b7635d2ba5442e543a2c8c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:33:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:33:25.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ie95b20c94358ce176bac4161ad01a4c1428adf7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/82d2897837991c68ec100fd260f15e0e716cd029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/82d2897837991c68ec100fd260f15e0e716cd029"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
