)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~658572","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~stable%2Fstein~Ia437e8f435d865cc79a138c5eb35e0b045d76b24","project":"openstack/openstack-ansible-apt_package_pinning","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ia437e8f435d865cc79a138c5eb35e0b045d76b24","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-11 23:57:31.000000000","updated":"2019-05-12 02:09:07.000000000","submitted":"2019-05-12 02:09:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658572-1557626947368-72f24b91","meta_rev_id":"d4277acca93ab11500d8bda202c1de90790b8b2f","_number":658572,"virtual_id_number":658572,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-12 02:09:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-12 01:56:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-12 01:56:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 01:56:45.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-12 02:09:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee52d28e983ed9da2510738d7b7c847e05c490a4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-11 23:57:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d394e9922c4ad799559bc9a7a99458b9f4492408","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 00:13:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/72/658572/1/check/openstack-ansible-linters/94fb50a/ : SUCCESS in 4m 52s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/72/658572/1/check/openstack-ansible-functional-debian-stable/b928596/ : SUCCESS in 6m 53s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/72/658572/1/check/openstack-ansible-functional-centos-7/9e2d772/ : SUCCESS in 7m 16s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/72/658572/1/check/openstack-ansible-functional-ubuntu-bionic/dbed06f/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/72/658572/1/check/openstack-tox-docs/eef0076/html/ : SUCCESS in 3m 15s\n- build-openstack-releasenotes http://logs.openstack.org/72/658572/1/check/build-openstack-releasenotes/fd3a980/html/ : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0b121383a81635f633ff13bfd5cc2f9a0d2741fe","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-12 01:56:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"58c4c976c678289292cbd1845277faaf11cf90a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:56:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dbe3ddf120e5f427db1f1f925c3669ea43e4d5b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 02:09:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d6292b2ad47b377261a0b6b3c67f9c19823195c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 02:09:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/72/658572/1/gate/openstack-ansible-linters/e5c9870/ : SUCCESS in 5m 03s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/72/658572/1/gate/openstack-ansible-functional-debian-stable/524fb80/ : SUCCESS in 5m 30s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/72/658572/1/gate/openstack-ansible-functional-centos-7/23c7c52/ : SUCCESS in 6m 46s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/72/658572/1/gate/openstack-ansible-functional-ubuntu-bionic/eba7150/ : SUCCESS in 4m 35s\n- openstack-tox-docs http://logs.openstack.org/72/658572/1/gate/openstack-tox-docs/07cfcbb/html/ : SUCCESS in 2m 34s\n- build-openstack-releasenotes http://logs.openstack.org/72/658572/1/gate/build-openstack-releasenotes/9776e93/html/ : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14c2620785303097dbddf1b7dc58c418b8e95819","revisions":{"14c2620785303097dbddf1b7dc58c418b8e95819":{"kind":"REWORK","_number":1,"created":"2019-05-11 23:57:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/658572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/72/658572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/72/658572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/72/658572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/72/658572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/72/658572/1"}}},"commit":{"parents":[{"commit":"5ad929d333a7e3c4c33999737204b2c3899f7c7d","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/5ad929d333a7e3c4c33999737204b2c3899f7c7d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:57:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:57:30.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ia437e8f435d865cc79a138c5eb35e0b045d76b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/14c2620785303097dbddf1b7dc58c418b8e95819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/14c2620785303097dbddf1b7dc58c418b8e95819"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
