)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~660618","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~master~Iad14b2e089da2bbb74f0d475f4c019001d5c2670","project":"openstack/openstack-ansible-apt_package_pinning","branch":"master","hashtags":[],"change_id":"Iad14b2e089da2bbb74f0d475f4c019001d5c2670","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"ABANDONED","created":"2019-05-22 09:43:21.000000000","updated":"2019-05-22 10:44:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"77da71a66261cb8bfe4d39a0ced2a4022f96a3ec","_number":660618,"virtual_id_number":660618,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-05-22 09:54:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 09:54:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4efbbd483429d86f73487048f2817b4830cba1a","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2019-05-22 09:43:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0ecdc8f2ee8d6ed11a4d1dceca1b9dcf8c7c8a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 09:54:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/18/660618/1/check/openstack-ansible-linters/2d87ffe/ : FAILURE in 3m 41s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/18/660618/1/check/openstack-ansible-functional-debian-stable/3834ecf/ : FAILURE in 3m 39s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/18/660618/1/check/openstack-ansible-functional-centos-7/1ec3bb6/ : FAILURE in 4m 40s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/18/660618/1/check/openstack-ansible-functional-ubuntu-bionic/a7fe055/ : FAILURE in 3m 47s\n- openstack-tox-docs http://logs.openstack.org/18/660618/1/check/openstack-tox-docs/83a392c/html/ : SUCCESS in 2m 50s\n- build-openstack-releasenotes http://logs.openstack.org/18/660618/1/check/build-openstack-releasenotes/8a3ebc5/html/ : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1},{"id":"77da71a66261cb8bfe4d39a0ced2a4022f96a3ec","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2019-05-22 10:44:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c16eca02adfc68cd111313e8fcc5b5b6bd58395","revisions":{"5c16eca02adfc68cd111313e8fcc5b5b6bd58395":{"kind":"REWORK","_number":1,"created":"2019-05-22 09:43:21.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/18/660618/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/18/660618/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/18/660618/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/18/660618/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/18/660618/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/18/660618/1"}}},"commit":{"parents":[{"commit":"82d2897837991c68ec100fd260f15e0e716cd029","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/82d2897837991c68ec100fd260f15e0e716cd029"}]}],"author":{"name":"zhanglianfa","email":"sdwszlf@126.com","date":"2019-05-22 09:39:17.000000000","tz":480},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2019-05-22 09:42:32.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Iad14b2e089da2bbb74f0d475f4c019001d5c2670\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/5c16eca02adfc68cd111313e8fcc5b5b6bd58395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/5c16eca02adfc68cd111313e8fcc5b5b6bd58395"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
