)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~664282","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~stable%2Frocky~I86c556d85c18ecf90599b5c0963b5c429c79e78a","project":"openstack/openstack-ansible-apt_package_pinning","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I86c556d85c18ecf90599b5c0963b5c429c79e78a","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:43:08.000000000","updated":"2019-06-10 17:24:44.000000000","submitted":"2019-06-10 17:24:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664282-1560187484955-41fe81b9","meta_rev_id":"abadb538efa5c5b42aab1a0e8b0b6f4a45f0474f","_number":664282,"virtual_id_number":664282,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-10 17:24:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-06-10 17:10:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-06-10 17:10:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:10:22.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-10 17:24:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72b50aa3493e2ef569f34cefd9c65262779dcfd5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:43:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fb01d6dd04c3c9871a67fba8f4dd81a3c7574f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 16:17:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/82/664282/1/check/openstack-ansible-linters/7d5a770/ : SUCCESS in 5m 26s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/82/664282/1/check/openstack-ansible-functional-centos-7/9ff77c0/ : SUCCESS in 6m 44s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/82/664282/1/check/openstack-ansible-functional-opensuse-423/eead24a/ : SUCCESS in 6m 01s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/82/664282/1/check/openstack-ansible-functional-opensuse-150/4f806d2/ : SUCCESS in 6m 13s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/82/664282/1/check/openstack-ansible-functional-ubuntu-bionic/f3d92a1/ : SUCCESS in 5m 38s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/82/664282/1/check/openstack-ansible-functional-ubuntu-xenial/68e8b36/ : SUCCESS in 5m 38s\n- build-openstack-sphinx-docs http://logs.openstack.org/82/664282/1/check/build-openstack-sphinx-docs/7b02efd/html/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6b0f79e8d760db5586a226decaabfc906936021b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 17:10:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f3e7a5a8d093e4b20f699e02e0303da4e386cb81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:10:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"409efade09027d5fa2f05af2763c28a23a97cab0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:24:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"da5f5cebb02d1f8b07f280a33c28e228bbc7bc5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:24:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/82/664282/1/gate/openstack-ansible-linters/188e86e/ : SUCCESS in 5m 49s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/82/664282/1/gate/openstack-ansible-functional-centos-7/39affdd/ : SUCCESS in 7m 55s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/82/664282/1/gate/openstack-ansible-functional-opensuse-423/ecb9138/ : SUCCESS in 6m 00s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/82/664282/1/gate/openstack-ansible-functional-ubuntu-bionic/7b06949/ : SUCCESS in 5m 56s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/82/664282/1/gate/openstack-ansible-functional-ubuntu-xenial/de5d6ce/ : SUCCESS in 5m 35s\n- build-openstack-sphinx-docs http://logs.openstack.org/82/664282/1/gate/build-openstack-sphinx-docs/6299013/html/ : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50977e4921e5a86d0abc3e70b5db0ab2f5273246","revisions":{"50977e4921e5a86d0abc3e70b5db0ab2f5273246":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:43:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/664282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/82/664282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/82/664282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/82/664282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/82/664282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/82/664282/1"}}},"commit":{"parents":[{"commit":"0ce2636062f128f1e1a9570b1c29509630693504","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/0ce2636062f128f1e1a9570b1c29509630693504"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:43:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:43:07.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I86c556d85c18ecf90599b5c0963b5c429c79e78a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/50977e4921e5a86d0abc3e70b5db0ab2f5273246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/50977e4921e5a86d0abc3e70b5db0ab2f5273246"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
